)]}'
{"id":"openstack%2Fswift~748281","triplet_id":"openstack%2Fswift~master~Ic4b274da7dcf9b166513d9c138f4b0370b01eeb0","project":"openstack/swift","branch":"master","topic":"worker-management","hashtags":[],"change_id":"Ic4b274da7dcf9b166513d9c138f4b0370b01eeb0","subject":"swift-init: Don\u0027t expose misleading commands","status":"MERGED","created":"2020-08-26 17:27:47.000000000","updated":"2020-09-03 23:24:15.000000000","submitted":"2020-09-03 23:19:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748281-1599175199626-1545f216","meta_rev_id":"ecacb9f8183f354a528b6c7ed9294f4e1904216f","_number":748281,"virtual_id_number":748281,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 23:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-09-03 13:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 18:50:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 17:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 17:45:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-09-03 23:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a135a250098a1503e87d6420d9670728941bad85","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-26 17:27:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e3ad01d110ea23644179415a5305e450e8ec03c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-26 18:50:18.000000000","message":"Patch Set 1: Code-Review+1\n\nWill squash into parent -- good call!","accounts_in_message":[],"_revision_number":1},{"id":"5ddea82d9dd6f361a7bc37502f165ef17550be71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb5a3f7eacf4e8799c5656901e76ef0 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/95418c117e77438585d0f880fec58a20 : SUCCESS in 1h 10m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/534b3f69a2584b008ec53ba6320fe9a0 : SUCCESS in 46m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e3d9909683f4307b3132f9d05aec0f5 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d216d68a41d416e8ceddb1e0ce28f4b : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0646d511acc6462984cd1bf232c06331 : SUCCESS in 13m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c186b3be8eb4085984c191eaaed96c5 : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/897484f2aae642e496b69ef5614d65e9 : SUCCESS in 13m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f4ac63df6f54ac7a8d7bec95c39ddeb : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3adbda08f20b41f3acb5601aa060ae83 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57439bd0583d40b593eaa25f80f1d1bf : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95aa747dae4c4e6886d6ccf8adf0ce6f : SUCCESS in 18m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e490b1b833384af89d52888c838c2770 : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eee9632ce9f64e27b42f73dc7360410d : SUCCESS in 20m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/41fd879eb18d4cbf85d4c65f17c5f9bb : SUCCESS in 20m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e99a877acd3149228af5a19c920907ce : SUCCESS in 24m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/139199726e09402dab1eca5e0865de1f : SUCCESS in 1h 01m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00e39608bbe44986ac9d2b34c3e2e80b : SUCCESS in 45m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c984fc99762412b85286237093912b1 : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/557906b916234a05a3d4be5d687281df : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aed55cc173e4da89609130d87ce6ce4 : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd2a606a74fa4312822d85435e2efb4d : FAILURE in 10m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1751a0536d2842169781bdb92eb2b40b : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6a5c7fc3fe5d3bd9411c207b6891215dfde565af","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-03 13:20:00.000000000","message":"Patch Set 1: Code-Review+2\n\nthis did NOT get squashed","accounts_in_message":[],"_revision_number":1},{"id":"878d37fa6c5db693b95a1f64c87ab81e6e1eaba8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-03 17:45:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35c479d16211def9ec850231c067f7b8f5afee2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:59:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5d9f9aa4628b5174025253d2569333d26e2a050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:19:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5839ef744f155a21448fc6bc66c1482792fd882","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:19:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7298c68a9d4546b7a19d4ebfa51812 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/b80757823e7f48abbd0021aab3de6451 : SUCCESS in 1h 07m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72c49ce5b3ef41968a038968abf38425 : SUCCESS in 52m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a9af285c4bf8491893b64a2571b4c054 : SUCCESS in 12m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0a0e4f772d0948238732e8539f9fc4bf : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a715e4d6b19e4d18a45ae962ad286de3 : SUCCESS in 11m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcf15b672a3b4f7dba4e728fc3d526ac : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08644a18974a4e1992bda6c79587a483 : SUCCESS in 13m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eed8e3f4718d474181f6b605a3fb061e : SUCCESS in 20m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/732c84da712b49e7959b31c1788f70aa : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbcbd795e9894e1c88fc0e495e52ba50 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/174c16ae9fff46aab29b6fcd578b5134 : SUCCESS in 20m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/372eb78ec83746fd81ff87c6cd799f01 : SUCCESS in 18m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8d8c461d6d6a43ecbf573950b018185c : SUCCESS in 23m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e564f61eb06470c8a356f5ca5fe57ec : SUCCESS in 42m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/902afd18b7fc48d0b51937efe86840d1 : SUCCESS in 40m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/458148c6fd184741b4287c630f9c0110 : SUCCESS in 1h 05m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62c90a1a6b2641c7b75fdaf80aa64740 : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f7aca55a8449f8855d852ba7ac8c94 : SUCCESS in 6m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afa5053b55a049f5b87624173dea20bf : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f7b5afc83e25611eb4ef8cc893291946b3cce144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:24:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee05fa80a8f34c7594640884f270ae9e : SUCCESS in 3m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/faf8325188e24061af7770d9700c7998 : SUCCESS in 3m 02s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1a98e10a34eb47e88271498d54b7ddd7 : SUCCESS in 1m 16s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee693265fe01e9ca3b3151812d5f14db0109311b","revisions":{"ee693265fe01e9ca3b3151812d5f14db0109311b":{"kind":"REWORK","_number":1,"created":"2020-08-26 17:27:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/81/748281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/748281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/748281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/748281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/748281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/748281/1"}}},"commit":{"parents":[{"commit":"3f5e712be673a624bcebf006263301833a6a15b1","subject":"wsgi: Allow workers to gracefully exit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f5e712be673a624bcebf006263301833a6a15b1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-26 17:24:46.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-26 17:24:46.000000000","tz":-300},"subject":"swift-init: Don\u0027t expose misleading commands","message":"swift-init: Don\u0027t expose misleading commands\n\nThe kill_child_pids command does NOT need to be an @command to be\navailable in probe tests.  I don\u0027t think it\u0027s helpful to operators,\nsince restarting all children is best handled by existing\nreload/seamless-reload commands.  But mainly the seamless option is\nmisleading since it very much results in a NOT seamless reload (all\nworkers close sockets immediately, and nothing can accept new\nconnections until existing connections finish and the parent spawns\nreplacement workers - i.e. it\u0027s just like graceful)\n\nChange-Id: Ic4b274da7dcf9b166513d9c138f4b0370b01eeb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee693265fe01e9ca3b3151812d5f14db0109311b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee693265fe01e9ca3b3151812d5f14db0109311b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
