)]}'
{"id":"openstack%2Fswift~749382","triplet_id":"openstack%2Fswift~master~Ia8db40227343e9c4555267c62072a1c9bfc28c66","project":"openstack/swift","branch":"master","topic":"s3-quotas","hashtags":[],"change_id":"Ia8db40227343e9c4555267c62072a1c9bfc28c66","subject":"s3api: Make quota-exceeded errors more obvious","status":"MERGED","created":"2020-09-01 19:44:11.000000000","updated":"2020-09-13 19:54:16.000000000","submitted":"2020-09-13 19:53:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749382-1600026780470-c94f3813","meta_rev_id":"2451abb76406a2548c0df5fed300581cb0a76911","_number":749382,"virtual_id_number":749382,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-13 19:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-11 22:40:23.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-09-13 18:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-09-13 18:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-13 18:42:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-09-13 19:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27348f33a2bcfeac9d48028472550cd251fa2bb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-01 19:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0b067cb48bfeca8d88ff90cd7f33372b65fa02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:42:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f829c4c2149446eb1ddf5de4c0f4f5d : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/c0529a16c4284a11ab5c08e23c68e76d : SUCCESS in 1h 08m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f77d34a929c54218afaf5e8beebbb79f : SUCCESS in 45m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f4fc19cd4224d79b3d167d0f932806b : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9f29e392a71d4cb7badcdeebd0c0d763 : SUCCESS in 12m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9758a7504a8d4015a57fd3ca44ecac4d : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebc077d36910467bbabcfaa25b82b1e6 : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a2f2c52288df4f929f23dff73607ab2c : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6536ec223974924b8c9cd41a70d7c12 : SUCCESS in 18m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/333d9b10fbd04e06b02d109523874ba4 : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2979ec246a41485581a03e0768704975 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/056d02e4c3c94c0682059be28af98907 : SUCCESS in 18m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4933259fc9e141ae8560c603fadd4ad6 : SUCCESS in 22m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/458d2e6d238741a7a20bf378866f45e5 : SUCCESS in 24m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c5a2466124d4652a63eba0619e366a9 : SUCCESS in 25m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3991c537d3364b1e88fd5c4a4791a54b : SUCCESS in 25m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a6c943db99ef48ed8d9397ae66c63fbc : FAILURE in 44m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d663ede888b41958834bbb424344e66 : SUCCESS in 39m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/326699f4877b4a3baaeecb829a428f06 : SUCCESS in 42m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64947737044047748173f07369394096 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d798720ebda2442f880eea12b90f3cd6 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffeb0301c1c44bc1a5ea0d704463daeb : FAILURE in 10m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f048bbf8ac24ce98f82951fc8770346 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bbd818a335c0003ccb29b119b9f6f22d625b3185","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-01 22:46:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99ed9737931bc7aa9ea871bfdf8eca765591eae0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-01 23:41:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3df3f103a7171b9379f2d0d758fd3ec5614cc4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 00:47:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c216ee39d2b548e19a4896cf2686e65d : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/f382b10acc5c4b2095c54a55ffc620c3 : SUCCESS in 1h 00m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0bdf7fe396944238a7c4a3e9539ddaa : SUCCESS in 48m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d13d88ffd6ce4c6599ba42e4f480f41a : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b58f1cbfbbc049508294b87d3eef2f51 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78c9db0b90474ceda9a40ff958aa1aab : SUCCESS in 11m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71bbff0bc49c438b86e8de17574ffb32 : SUCCESS in 16m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9fa97b30caba410b9c936b8cea654dca : SUCCESS in 10m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/97470e7d92f44336b1114cc7dac88457 : SUCCESS in 29m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1b2748ed68549b086287b77c38971c3 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12a5d606d54f4e9da8ad17ebd71fc5b3 : SUCCESS in 36m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4130c53ecc044bc8b7d207c4d64186b9 : SUCCESS in 16m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/554397b67658410fb8dda092574b2bbe : SUCCESS in 37m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dc22d9d0cf734ec5ac3602457065db01 : FAILURE in 31m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1cd740e59f8e481b91e15794027730f9 : SUCCESS in 22m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35fd1025d702458884729bb6a80cc60a : SUCCESS in 21m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c507361bd4b0439a8540eb60310b4857 : SUCCESS in 39m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a012617093b4accbf4e312d106406e0 : SUCCESS in 58m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2815d1338c974430808132827ebfc0e0 : SUCCESS in 43m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bdd9f40d6ba476ab0f8f87d9db890ea : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ecb333aaa54c12871dd8a83a8150b8 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c880510705a94497af51dbfdb17da129 : FAILURE in 10m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1ef4bbc0524e5593f234fd3084b1dd : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"025f34fa96d0766fd4420149728f269cefe3ed2a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-02 15:27:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d2bb66ac17f87b10edf62a95c352847c0de738ae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-02 16:01:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fad637277040ef2f83ccf02c3a63c1fe1c9dcaf","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-02 16:28:11.000000000","message":"Patch Set 2: Code-Review+1\n\nseems to be moving in the right direction; I\u0027ll tentatively agree a pre-factoring is worth it given the bug is mostly ergonomic","accounts_in_message":[],"_revision_number":2},{"id":"71c6088ab0da30758885c041ab0086ff4e37a17d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-02 16:35:13.000000000","message":"Patch Set 2:\n\n(6 comments)\n\ni\u0027m sure it\u0027s going to be amazing, maybe it\u0027s already a lot better - it at least fixes that bug!","accounts_in_message":[],"_revision_number":2},{"id":"473370e699a71ca184d901e0dc6e87c8770d863c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:12:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17feec22ff74e4084d84305548220d9 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/bd242d39613944028a3d734110276424 : SUCCESS in 1h 05m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe2985d63aea4483acd51f344d4f21f9 : SUCCESS in 54m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/979c910fa0384103b59c387d626a1e88 : SUCCESS in 9m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/29a223c45dc04679b75db2d28f87440f : SUCCESS in 12m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56c67f34fb7f4321b46768baa2507b1a : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/587aa22ea275453a9e0b66817ce04d5a : SUCCESS in 15m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c737b94793654a52b23ed2fe0567240b : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d0a175a47afd44d5bde1657360defda7 : SUCCESS in 18m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8077584889a415d8d5304475d2345a2 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffaf6e0cc6b74ff384cc79df94c9099a : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c605a862a9b4f9eb3fbf54375f50570 : SUCCESS in 18m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1721e53d07e49e989dbae44096354e8 : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3f0bc683753340ea853303edf619cc66 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aeb84f468a52436ca4732f35d0d348f7 : SUCCESS in 20m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a00762e3c3b240d4a7b2303f1b9f7ae6 : SUCCESS in 19m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/306e809b77684e1294e069b908824f97 : SUCCESS in 46m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0a62f6d75df424d8716809c137eb8ae : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f6c7d18c5254fb3b8c93a5470bbc627 : SUCCESS in 43m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1a6efc0d1c244478cbcfbf1d04d1614 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eafaaf02958460c90f82ddba41afa0e : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f905dc26baf4c50a76fe3e11dac0c92 : FAILURE in 9m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84c0d3d6b07d4db0a2a75093b5c05f47 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d4e72b63fd0f3913a458a124f785e67ebf8c6199","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-02 23:03:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7491e0159e0b6470e854cea49b5c8aadddd7ea41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-03 00:24:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43873e201d7b1a4eff362928087f8471ad80b583","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-11 21:56:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nI think we should stack up the refactor on the surgical fix","accounts_in_message":[],"_revision_number":2},{"id":"4764e0b0d66dd3f4dc119b3e83bad338a758c2de","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 22:37:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b2eafbd4a40a4652fcd1129c4664cdc5f8656d2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 22:40:23.000000000","message":"Patch Set 3:\n\nSplit the refactor off as https://review.opendev.org/#/c/751555/","accounts_in_message":[],"_revision_number":3},{"id":"03daa2def0ed60dd75fb7d5230c1b4362968ed13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:44:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634c214f63cb4b3a9ca7dd560ce8114c : SUCCESS in 18m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9be9e3b881c04b139f9397318e93c7f8 : SUCCESS in 55m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35d60ed081404ca6bdb1b6cd22aa516c : SUCCESS in 49m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d85154e1edab45bf9635de863bf836ed : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b4257cd05fc44b34ad17f9dfd8bfebd8 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a527b0509e9419da84614fa11468e07 : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/221944c8d9aa41c38b4177fe83a831f7 : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54de03211b45491990cafb4dc20433a1 : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f13c53d64b2840cc9d5603c1fb62e74b : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c5630dbfdeb41d8a9d36600c4b2a4f8 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dbc5926fa644913829071e149046b0d : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b6ed735655a74913917750aa67d5bece : SUCCESS in 17m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5b53f1adab7a40fea1ecc29c7c81b9df : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/641f4cde26934efabfce5915f785b156 : SUCCESS in 20m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f48943c3da90465c879067f5c27f9c2f : SUCCESS in 16m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d264b0b92c5e4c5abae7fa43dee66a98 : SUCCESS in 19m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7304c8f22d342f49c13a1a1eb1e632d : SUCCESS in 47m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cee801cfc43144f682bb8d8af15f7803 : SUCCESS in 55m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9edfb0c6631e4b129c3c71469147b9bc : SUCCESS in 43m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c46feb5b80874db28b3385aeab29f259 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193bacf8d39e40b38166a2925853cdc0 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12d8a93fa70f4ea699bed0aed6d2644c : FAILURE in 10m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b8796b00b2d4a84b1a6efda50faa1c8 : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":3},{"id":"939e9601c3ad7178debdb3d164e2ef1c132de0dc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-13 18:42:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"34e139b1c6ee6300051355e15a2d10744c1af64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 18:42:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5202e266e1eb07b00d6024803ff800c26d9f9ba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 19:53:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6d5be08754f7ca49fc34200c51e85bf324c6d645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 19:53:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b1c34924d0d4f59be8dc8951531db0f : SUCCESS in 15m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/2195e881bd6a4c999351cadc6c6995d7 : SUCCESS in 1h 09m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b38aa76d6a24ee5be1ad6eb74eb26f4 : SUCCESS in 40m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1a18dcf74a6a4a81865ac745e47f1bff : SUCCESS in 10m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c0c14305f46148e7964d2188f8312b85 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eac4c03a494a4495b9965f7945a1dfe9 : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e9606f24c3684fcab52173e67ad01588 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e1266c4ce082491d8757ec92db29e0cd : SUCCESS in 12m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fd5c44ec7e9249028e52bc481d8bc5f8 : SUCCESS in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c13f9b72aff54905ae886b10cf63fa64 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db8a242be7364d51b6ba4418075d5c78 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/512a80f2adcc4a5384796abde5adc854 : SUCCESS in 17m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/72fe514cd35d40b1b8d85f80610b632e : SUCCESS in 15m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5b01a456dd3d44d198f78d1274e2af62 : SUCCESS in 17m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a015c43cf8545aeb9fe1789638230c0 : SUCCESS in 43m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ead61b9894284eb28b712519fdaffc93 : SUCCESS in 43m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76ac17523cd94fb99254a6d09300b1f8 : SUCCESS in 42m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b30565f0ea64ea7b87aa2d73eeb25c3 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fb9c63968d497cac3eec8002495776 : SUCCESS in 5m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/068f53b404784dceb6dcf427b61cf070 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":3},{"id":"c502598ddd9f2b1e50d5f83a156e8f51f900ff0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 19:54:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/230b631dff5642e88927ea81a3045e74 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/29c8faf6cd3d4f119e6ce3c9694492b5 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/427657c619bc4b4b96b13605d5e69d2b : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f8a9a6e11620d1abc537bcef107fa21af53fc6d2","revisions":{"c29ef7cf8b12daa7fb23d3debc6e9ffe40d99e8f":{"kind":"REWORK","_number":1,"created":"2020-09-01 19:44:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/749382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/749382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/749382/1"}}},"commit":{"parents":[{"commit":"0fa259e7beecaf0dd8e3d342ffbf1ec4a2b23864","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fa259e7beecaf0dd8e3d342ffbf1ec4a2b23864"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-01 19:43:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-01 19:43:46.000000000","tz":-420},"subject":"wip: s3api: Make quota-exceeded errors more obvious","message":"wip: s3api: Make quota-exceeded errors more obvious\n\nChange-Id: Ia8db40227343e9c4555267c62072a1c9bfc28c66\nCloses-Bug: #1893811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c29ef7cf8b12daa7fb23d3debc6e9ffe40d99e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c29ef7cf8b12daa7fb23d3debc6e9ffe40d99e8f"}]},"branch":"refs/heads/master"},"63e8846f5f9906932443566c570677af5d8a27af":{"kind":"REWORK","_number":2,"created":"2020-09-01 23:41:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/749382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/749382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/749382/2"}}},"commit":{"parents":[{"commit":"0fa259e7beecaf0dd8e3d342ffbf1ec4a2b23864","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fa259e7beecaf0dd8e3d342ffbf1ec4a2b23864"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-01 19:43:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-01 23:40:11.000000000","tz":-420},"subject":"s3api: Make quota-exceeded errors more obvious","message":"s3api: Make quota-exceeded errors more obvious\n\nChange-Id: Ia8db40227343e9c4555267c62072a1c9bfc28c66\nCloses-Bug: #1893811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e8846f5f9906932443566c570677af5d8a27af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e8846f5f9906932443566c570677af5d8a27af"}]},"branch":"refs/heads/master"},"f8a9a6e11620d1abc537bcef107fa21af53fc6d2":{"kind":"REWORK","_number":3,"created":"2020-09-11 22:37:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/749382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/749382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/749382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/749382/3"}}},"commit":{"parents":[{"commit":"4fefdf4d83bc43f86014906ff46a936168f71a69","subject":"Merge \"gate: Make rolling upgrade job work with either 60xx or 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fefdf4d83bc43f86014906ff46a936168f71a69"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-01 19:43:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 22:37:33.000000000","tz":-420},"subject":"s3api: Make quota-exceeded errors more obvious","message":"s3api: Make quota-exceeded errors more obvious\n\nChange-Id: Ia8db40227343e9c4555267c62072a1c9bfc28c66\nCloses-Bug: #1893811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a9a6e11620d1abc537bcef107fa21af53fc6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8a9a6e11620d1abc537bcef107fa21af53fc6d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
