)]}'
{"id":"openstack%2Fswift~749536","triplet_id":"openstack%2Fswift~stable%2Frocky~I00c6693c42c1a0220b64d8016d380d5985339658","project":"openstack/swift","branch":"stable/rocky","hashtags":[],"change_id":"I00c6693c42c1a0220b64d8016d380d5985339658","subject":"py3: Work with proper native string paths in crypto meta","status":"MERGED","created":"2020-09-02 15:24:33.000000000","updated":"2020-09-05 00:49:09.000000000","submitted":"2020-09-05 00:49:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749536-1599266949810-7e3b2230","meta_rev_id":"8b4b77469026e3084f933be71c3d6a0b884b735a","_number":749536,"virtual_id_number":749536,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 00:49:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-04 23:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 23:24:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-05 00:49:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fff3cde34212d4e7f034b65bd319d22df30952d8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-02 15:24:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6143083a0eaefc62f970cc601fd3768b4cb2fd71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/27c7b2e8a5a145adbd6875e73cbc2a31 : SUCCESS in 6m 48s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bcad1772cffe4baba79ebb42a1082c65 : SUCCESS in 11m 15s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/791c07ce756743578f83bb15cef2ab96 : SUCCESS in 5m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0360a3539c341d7afa9136999daa503 : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/26b97d14f8f5468794121f0c89f78607 : SUCCESS in 9m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5a01122653724ad99655d2c151522bec : SUCCESS in 9m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c8a45841ae04c4c855422e61c7d05f3 : SUCCESS in 11m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/09202120c5104542bd2800e5a680638c : SUCCESS in 19m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb536172c4a84cddbb5a1cc7d3f4e812 : SUCCESS in 36m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/823df73af65242299ed652f4d92c02c6 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a95733fc3694804ad6890c5a099a182 : SUCCESS in 5m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a255a15b1ac045ab991eb703072fe697 : SUCCESS in 34m 55s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/9686d3f9950548c5a58c8b958ec69637 : SUCCESS in 1h 24m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1a80027ba384435b492731888fc1033 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"83d3dc1edc7d3ebb482861b26ad2687664f2764e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-04 23:24:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b0a7a341c6f93ca98961c5691dccb6ef06f44ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:24:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"784ceb59fcd2dd300b1ee6cd7b351c43b34d2c25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 00:49:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8b4b77469026e3084f933be71c3d6a0b884b735a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 00:49:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b29569cba638409dbbb09439aa7bc599 : SUCCESS in 6m 44s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80ad585befa34401b07addfb8b08bf39 : SUCCESS in 10m 58s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/8438cd0542b94457a7607675e20abb3b : SUCCESS in 5m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c1101eccdd1468e84bf716d61c34ee7 : SUCCESS in 9m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/638133a5132642e79bd0866d2bd9dbc7 : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/32ab6f9774884eb6a5217b33f2954ada : SUCCESS in 9m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/46e734b0cc0148a5938a4e101d4ea74a : SUCCESS in 9m 40s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/78564384bc9d458c9747366f878eb527 : SUCCESS in 10m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c999e93e2ed4150a3a013365a560069 : SUCCESS in 47m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6682827643bf46bebefef345c6b7dffe : SUCCESS in 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81b3013a2c4499bbe85dec116aa33a7 : SUCCESS in 5m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b61192692174487cac1f358ddc5c39e4 : SUCCESS in 27m 06s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4255e4b4e6a64f63b98bb4bc4f976e7d : SUCCESS in 1h 18m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc9d66de3faa48cd901912816b83b6c1 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca66e2e96f2272870a03cebfb8c24b9b6bbacf16","revisions":{"ca66e2e96f2272870a03cebfb8c24b9b6bbacf16":{"kind":"REWORK","_number":1,"created":"2020-09-02 15:24:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/749536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/749536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/749536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/749536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/749536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/749536/1"}}},"commit":{"parents":[{"commit":"5aa4c5d88fec98cf4ca0536c511c61759e865ec9","subject":"Authors/changelog for 2.19.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa4c5d88fec98cf4ca0536c511c61759e865ec9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-20 21:18:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-02 15:23:17.000000000","tz":-420},"subject":"py3: Work with proper native string paths in crypto meta","message":"py3: Work with proper native string paths in crypto meta\n\nPreviously, we would work with these paths as WSGI strings -- this would\nwork fine when all data were read and written on the same major version\nof Python, but fail pretty badly during and after upgrading Python.\n\nIn particular, if a py3 proxy-server tried to read existing data that\nwas written down by a py2 proxy-server, it would hit an error and\nrespond 500. Worse, if an un-upgraded py2 proxy tried to read data that\nwas freshly-written by a py3 proxy, it would serve corrupt data back to\nthe client (including a corrupt/invalid ETag and Content-Type).\n\nNow, ensure that both py2 and py3 write down paths as native strings.\nMake an effort to still work with WSGI-string metadata, though it can be\nambiguous as to whether a string is a WSGI string or not. The heuristic\nused is if\n\n * the path from metadata does not match the (native-string) request\n   path and\n * the path from metadata (when interpreted as a WSGI string) can be\n   \"un-wsgi-fied\" without any encode/decode errors and\n * the native-string path from metadata *does* match the native-string\n   request path\n\nthen trust the path from the request. By contrast, we usually prefer the\npath from metadata in case there was a pipeline misconfiguration (see\nrelated bug).\n\nAdd the ability to read and write a new, unambiguous version of metadata\nthat always has the path as a native string. To support rolling\nupgrades, a new config option is added: meta_version_to_write. This\ndefaults to 2 to support rolling upgrades without configuration changes,\nbut the default may change to 3 in a future release.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWhen upgrading from Swift 2.20.0 or Swift 2.19.1 or earlier, set\n\n    meta_version_to_write \u003d 1\n\nin your keymaster\u0027s configuration. Regardless of prior Swift version, set\n\n    meta_version_to_write \u003d 3\n\nafter upgrading all proxy servers.\n\nWhen switching from Python 2 to Python 3, first upgrade Swift while on\nPython 2, then upgrade to Python 3.\n\nChange-Id: I00c6693c42c1a0220b64d8016d380d5985339658\nCloses-Bug: #1888037\nRelated-Bug: #1813725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca66e2e96f2272870a03cebfb8c24b9b6bbacf16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca66e2e96f2272870a03cebfb8c24b9b6bbacf16"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
