)]}'
{"id":"openstack%2Fswift~749614","triplet_id":"openstack%2Fswift~master~If4cafc9be571a31616186b9f67e7d1c0afeca0e7","project":"openstack/swift","branch":"master","topic":"sharder_auditor_hacks","hashtags":[],"change_id":"If4cafc9be571a31616186b9f67e7d1c0afeca0e7","subject":"Push shardrange audit hacks into a patchset","status":"NEW","created":"2020-09-03 00:42:23.000000000","updated":"2020-11-18 14:20:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6d1c82d45566ccc5de014db2439c726627aa369","_number":749614,"virtual_id_number":749614,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-18 14:20:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2020-11-18 09:49:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 14:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d23f4eeacb6920a3226ce394852eff15848d2dc6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-09-03 00:42:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a15f2075bd9f67a2fe8362c36ba5164b9296d147","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-09-03 00:43:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b012d75ec74fd78f8ece946343d75742ce05ad25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 01:50:22.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b8bc97c5c8411483307adc56e916d2 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/e657fe7edacc4b25a3191c74166327d3 : SUCCESS in 1h 06m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c376d42d1ad04199a1189ea8841aaf1c : SUCCESS in 44m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30bcf1bfb3c84a8191e003f7be6f5bbe : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b97f1a450144085bf2bf9e9b82a2b4f : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39ed5596e9e74f65a95d4aff24dc159d : SUCCESS in 11m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f01fbc686ed3464582ccac79cc32b0ba : SUCCESS in 13m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8422b3fbf52e4123ba83f4321894638a : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cb996cdc69fa4cacbe0f90635f783181 : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/758b71173f6e4fb8abef68a940dfb204 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1265c0d998b44243aeadff239a0fe255 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17a3816c6eeb4be0a680ec4752322785 : SUCCESS in 16m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1bf443815e384569bf2c698cb63243c3 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c67ea159cbc9493785ee9ee24b68b79c : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d710ee42a446465d8db0c0001300a9eb : SUCCESS in 17m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2bdd94ee56574bd18e25a6bc98d6c8af : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/808d059bd57f49cbbf544e879413d3a2 : SUCCESS in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3643049383ce43558b54c7fd2f7f0988 : SUCCESS in 42m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c407365db5b641cab2f7ef813acb8088 : SUCCESS in 39m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0493c06323384feea50ed829d4e9b544 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c0b8b4af894ba7bb69babe2fd63e47 : FAILURE in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf9d79a19cbf40a6bf19b1547737eb29 : FAILURE in 7m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7aabdd62b7004bb191bbb0ab84ba634c : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c6110b9eff775d3844cf36ef89b337c99a84bfe8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-09 02:06:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d816a1b559341d8d49176d377bf435446ee11420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 06:17:09.000000000","message":"Patch Set 2: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6255f1ed229344d8b00bc7578943bc5e : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/8e151289c7f24f68a677925338220cc2 : SUCCESS in 56m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e880955351ae4f7c9cb0740554870a22 : SUCCESS in 44m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27759a7355804e49843a164fe0a556b1 : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e0de0261ff24faeb5de2834644e5222 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b2d6b984db04ef98f5de6d60fe5c77a : FAILURE in 12m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d8d1204a7e44f7a5844017a2f86b50 : FAILURE in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d7491a28f0f4ad486538cf4fc307bc2 : FAILURE in 12m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe42c0e18ec414f8a902e76385d0937 : FAILURE in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/688ad4c4b13646cead09bd2e6ff0d7e9 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54a0737cc752403b90ac453f67c3b859 : FAILURE in 4m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e6f4d7526a84963b949cd962a0709eb : FAILURE in 3m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/18156b1012bd40909ab8ea36720ba996 : FAILURE in 17m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67def8779f814db9aaa3e3fdd381583a : SUCCESS in 24m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd1e96b1f325486b89f06a59ec70fbbb : SUCCESS in 22m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aaef00226a594874b68e8a5e4eb3ba25 : SUCCESS in 23m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eaa5ae19e6aa44c498845511ff08caf5 : FAILURE in 39m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26eb158a941340289049f2bc4c8a6894 : SUCCESS in 40m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/250da3bfe9034d60b097b00b8cf13d2c : SUCCESS in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f67d6d145f14e87ac383a48b16af6a2 : FAILURE in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af8ad10668748cbab3481a8844ed3c9 : FAILURE in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84b99ff46a984ee5905d06d313eed72d : SUCCESS in 22m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df5ce213e8d24cad9593dedb48efede0 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a0e6fc5004f0c8a1d4cf0a0bab27ac6c150fcf80","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-13 02:16:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b69a4dd68fbb10254fc65f94f1dbcb7f46b211de","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-13 04:17:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7949b9a4a6fc267bf9ba60ce93246e64ad16222c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 05:16:40.000000000","message":"Patch Set 4: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b1867c693c4cb7a236fb2dbd329d8a : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ecbfbece99a24fe0b1dddb0e422bf49b : SUCCESS in 56m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/03f846341e584299b65dd1625a5a915b : SUCCESS in 46m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36ab34677dea471e8f78d04d91382df0 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7505724df77a493caffa97b024359858 : SUCCESS in 10m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/beb212c631c1490eaf39a5cdc5f3650c : FAILURE in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0e9ff40e6d43bd8f311b607b20092b : FAILURE in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9eeb14174474e529915e2a37a9a6efa : FAILURE in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ec18b58ff0984fe8811db2e2d442df81 : FAILURE in 11m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/602ced6c8398495eb0a602ab246a763f : FAILURE in 21m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0488ff911644fa598c2264f88cf5996 : SUCCESS in 27m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12045aeff07d474b9861cf738a1d7efe : SUCCESS in 23m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fb2c87be162340979b1d072df007f845 : SUCCESS in 32m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cc9c3f3948904b9a95152f82718c74e8 : SUCCESS in 37m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f00c9d2f53c24dc0a1ff8e23b8578e0a : SUCCESS in 31m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1f3c2d9719d45cd91929d69566bf2f2 : SUCCESS in 25m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4e81f7cba0a043668033939f73778853 : SUCCESS in 44m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6231e7a7e196469abe73027990f5875f : SUCCESS in 39m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3a18fa676c1408688e74744d5b17350 : SUCCESS in 40m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd01ed3e540a48de9f68bcbf996e67a7 : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef5fcc9e3614d7aafa840a565270680 : FAILURE in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5380b33b70a843f19cea82f848c16666 : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e38cdecbeef4a8a937a4c9b85da8602 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":4},{"id":"75fe7ee58539e2c1190e84e157e8ff9a5d515938","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-16 06:06:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4e1dda87bef4e2ac6e8b7b52ed479f2a97bfaaec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 07:53:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc0cfee364f45e0abb9330fc84e6e37 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/103223dd06eb4f3093fa6d9f0d4e76bd : SUCCESS in 55m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/033758e37ed34e8db6ab61caa54b6cf3 : FAILURE in 1h 13m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d57379d2cfb447b999f0d39f1d6e805c : SUCCESS in 11m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6acfb75dc2bd443284818cf3dbac5054 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25497ed190ae4816b5cd900b71cf6987 : FAILURE in 14m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a5f9d0d9ba64b80916744e71abed74e : FAILURE in 16m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ddcc1c758634e57a4b13678afaaf126 : FAILURE in 20m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d8cae0b5d184609af4828b720eaeb5c : FAILURE in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11871119c8bc40ea96b056914377694a : FAILURE in 23m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c918e834e0f4a7eaa8716aa247bc4de : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d449961a4be348aba724250c770ec498 : SUCCESS in 20m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e0338fdcb33e419bb5e97d563f1b6e91 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fe408b416800484f8480559935e5971a : SUCCESS in 19m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8bf0963fb4504f0b9ab332e8fa0bb087 : SUCCESS in 21m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/efecc287e7d94e40a024bcc3c1dbc482 : SUCCESS in 21m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c424cf59d5de47b99bcc9cc51e525331 : FAILURE in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bd53678e8e74f2d93bd50cabda7097a : SUCCESS in 47m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65beba30ae124da3ba430cc9c95f4952 : SUCCESS in 44m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd0238f5456542c59f582a5212b0c460 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874f038455214db2b4e40d7a258211f3 : SUCCESS in 7m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/928ff4a05ab64755af11e3d7f7ea8297 : SUCCESS in 22m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/090c928296c44fe1aad577d5e76e352d : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":5},{"id":"498fdc6c9a5053fe224a1b3ec1092efba74cdf3c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-22 23:56:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fe58114ffa99019214b7225083b45276ea828402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 01:25:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97d528f5d174d85ba68c038067dad2c : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/c10db55eea5541f4983b1e22b26517ba : SUCCESS in 1h 04m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b761580b9401488bb55ea64aff29adf4 : SUCCESS in 49m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40bbd70008ec4fe4b9b6193dc648596c : SUCCESS in 10m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d47e47a0a7e24a058eca10088c94754e : SUCCESS in 10m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7182523b7e7e4996922ba5b3f3969316 : SUCCESS in 14m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16668e9b99eb4542a151866d873bfaf9 : SUCCESS in 15m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/61ef130e208d4023a1e779a8fa1f6113 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e771a693b0c4520892168c58c26d21e : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88788757faa748ac8f28dee12bd9dba3 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96f1a62c764c447b8485a0188a7467a0 : SUCCESS in 24m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10b6fcd706f241e48673cb84285986bd : SUCCESS in 29m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/57be2e68839e4525ade99b81ed365f3a : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c658573bf0514c85afc70f8598fa5c65 : SUCCESS in 29m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1e308e8ca15d489eb38ed4481cdc1bfb : SUCCESS in 25m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a46eea7ea38e431c9f90bd1257eeb3f7 : RETRY_LIMIT in 3m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ac2a5192ace4f6aa15696c0a9dbeeba : SUCCESS in 45m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d18b69f67234acd99319ae7d42e9910 : SUCCESS in 47m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/356ae8ae3fff498280756cd2beb48c6e : SUCCESS in 41m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abf99201b57246af81ecd49954d9a4c1 : SUCCESS in 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a3502cb60b4c40ab4c76a677e40903 : SUCCESS in 7m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4c2a7a5c86d474997d1a7ca887af654 : SUCCESS in 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d4d47e0716148498dd6096658827c06 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":6},{"id":"d5abda16ace37b7d6655b7889fbcf5fc813bc7a5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-11-18 09:47:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b602fb5aaa48d6b7e5abf0992ee3d79c3e1cce1c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-11-18 09:49:21.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a67dc7bd199586c5ff6f84da09324185c56c6c72","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-11-18 09:51:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5e2c3c8dedb821018f5357c8b3bd2d2c97e17fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 14:20:21.000000000","message":"Patch Set 7: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1affd3f561f643ee9d4d1f43b577520c : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8e98dcd1a659443e9461640f2a1ca04e : SUCCESS in 1h 02m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e15b1100ea0e41dc9b2f670b1b79dbee : SUCCESS in 48m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c483a6e063dc44bbbeee12c0bcbbf7aa : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/648bc9906a274796b937425c8edcf865 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/662fe8defcf34891b3c0a790834c5f18 : FAILURE in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f40980caf58444b1b9e33cb5a1a0f2ac : FAILURE in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e8fe510e74a14323877e3ab6accb1be0 : FAILURE in 13m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0478ef54294488af926e5c1edcfbf0 : FAILURE in 30m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01aae2fda8a0408fa23d4897521f8805 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d657d114bca44d08ee30c42c74a2b35 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5b4e8e5f938448cbfa82a376e7a62cc : SUCCESS in 19m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7634bcbc35074c1a82bfae1c53f9e341 : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/54de721fc418414084be5601aae2ef9a : SUCCESS in 22m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/256ea5e611944f90a0e0971d26cc931b : SUCCESS in 23m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce84de1cc309478c9a9feb1d4b8cd34e : SUCCESS in 25m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/965e0ad984c14d4e9af8f50cc38d044e : FAILURE in 47m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcf863b08afe4d5e90b5f12162744fc2 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c52e81f2df14b18bc4639cfa8adf5db : SUCCESS in 40m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c6590839f6c40eba9f358f62b6b7de1 : FAILURE in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b996df05cd94aac8f0fa6d482eec7e1 : FAILURE in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fce4d802d33e4a3c87fc9d1c2a5c1c54 : SUCCESS in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d0896b59cda4da0b63fbe786ca02807 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5edc85e66925f42fea3ade89b5290ad6c5ec2924","revisions":{"fddc2f3cc289f7a02bdda0d69606e47c7ff3248a":{"kind":"REWORK","_number":1,"created":"2020-09-03 00:42:23.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/1"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nNext I need to test and see if any of this actaully works :P\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fddc2f3cc289f7a02bdda0d69606e47c7ff3248a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fddc2f3cc289f7a02bdda0d69606e47c7ff3248a"}]},"branch":"refs/heads/master"},"a86f65a76ccedc80f6189c08e3a8c94e59d12fae":{"kind":"REWORK","_number":2,"created":"2020-10-09 02:06:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/2"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-10-09 02:06:06.000000000","tz":660},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nNext I need to test and see if any of this actaully works :P\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86f65a76ccedc80f6189c08e3a8c94e59d12fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86f65a76ccedc80f6189c08e3a8c94e59d12fae"}]},"branch":"refs/heads/master"},"669890ed448b5bb3e9dbe775406d724454c6fee9":{"kind":"REWORK","_number":3,"created":"2020-10-13 02:16:05.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/3"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-10-13 02:15:48.000000000","tz":660},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nNext I need to test and see if any of this actaully works :P\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/669890ed448b5bb3e9dbe775406d724454c6fee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/669890ed448b5bb3e9dbe775406d724454c6fee9"}]},"branch":"refs/heads/master"},"2fbc1b4249c0ea7c5c5083bf1964765436c813a5":{"kind":"REWORK","_number":4,"created":"2020-10-13 04:17:05.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/4"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-10-13 04:16:48.000000000","tz":660},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nNext I need to test and see if any of this actaully works :P\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fbc1b4249c0ea7c5c5083bf1964765436c813a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fbc1b4249c0ea7c5c5083bf1964765436c813a5"}]},"branch":"refs/heads/master"},"0879b4bcb9b4371643ccee1a38deb1bea85d4179":{"kind":"REWORK","_number":5,"created":"2020-10-16 06:06:59.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/5"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-10-16 06:06:40.000000000","tz":660},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nNext I need to test and see if any of this actaully works :P\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0879b4bcb9b4371643ccee1a38deb1bea85d4179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0879b4bcb9b4371643ccee1a38deb1bea85d4179"}]},"branch":"refs/heads/master"},"362b3cd58d3178b589d93d7d1f14225de2b0779f":{"kind":"REWORK","_number":6,"created":"2020-10-22 23:56:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/6"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-10-22 23:55:32.000000000","tz":0},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nAlso added some initial tests.\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/362b3cd58d3178b589d93d7d1f14225de2b0779f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/362b3cd58d3178b589d93d7d1f14225de2b0779f"}]},"branch":"refs/heads/master"},"5edc85e66925f42fea3ade89b5290ad6c5ec2924":{"kind":"REWORK","_number":7,"created":"2020-11-18 09:47:31.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/14/749614/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/749614/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/749614/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/749614/7"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-09-03 00:39:48.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-11-18 09:46:57.000000000","tz":660},"subject":"Push shardrange audit hacks into a patchset","message":"Push shardrange audit hacks into a patchset\n\nAlso a driveby to add the graphviz to manage_shard_ranges tool\nexpending it a bit more to use the RangeScanner weight calcs in the\ngraphviz SVG output.\n\nAlso added some initial tests.\n\nChange-Id: If4cafc9be571a31616186b9f67e7d1c0afeca0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5edc85e66925f42fea3ade89b5290ad6c5ec2924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5edc85e66925f42fea3ade89b5290ad6c5ec2924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
