)]}'
{"id":"openstack%2Fswift~750012","triplet_id":"openstack%2Fswift~master~I7b54f25f42577020b10029c74f8fc01fa6fc591e","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I7b54f25f42577020b10029c74f8fc01fa6fc591e","subject":"xlo: Drain error responses","status":"MERGED","created":"2020-09-04 18:36:16.000000000","updated":"2020-09-15 22:12:31.000000000","submitted":"2020-09-15 22:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750012-1600207881581-c94f3813","meta_rev_id":"d27cff7d147b120453463c916c1550892a1abf96","_number":750012,"virtual_id_number":750012,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-15 22:11:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-09-15 20:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-09-15 20:23:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-15 20:23:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-09-15 22:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"907f64afbcf4911b6ae1f3d598a734e406db65a8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-04 18:36:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4e413e4659d95830344227ddc6d340cd0f4935a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:45:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b05f490176424f8ccbb4c0a765bf40 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2e10a85dd7f54ef9918da81738286334 : SUCCESS in 1h 04m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d048f895f1564a00a8372243a47c361b : SUCCESS in 54m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c395337e871c421a915ecbe95ab06b77 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08ac95bc96e04d298de0f88f55a20f5f : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60a75e2afdeb40de9b5414162d3ec5f2 : FAILURE in 27m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f6249e59134ba39f3e36ced74e1519 : FAILURE in 28m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/db2e23306dc043b6b7a49d2b662f1196 : FAILURE in 28m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/99621c4b84514fd596c0ce49f91fe777 : FAILURE in 28m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61ec2f05343942ac8c90298ccb2b9909 : SUCCESS in 30m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69dde0395fc6413abe80fe4d409516b0 : SUCCESS in 31m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ae37e3d20094b2f85505e10f8f8b0f9 : SUCCESS in 32m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b1e31ac12f1d4b48ace0298549b5a5ea : SUCCESS in 31m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9f3aa219a36e4f1fbe2df831631f4de3 : SUCCESS in 32m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a785515fcdad426cb3d2d97c83816405 : SUCCESS in 27m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16f9383cd8ae44c2b6f850193dab5aef : SUCCESS in 51m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/82bff7dc079a45ea817776d76013802a : SUCCESS in 50m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ed9580d4d194a4592704778f1323808 : SUCCESS in 39m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3487430b23b421c80dff590f0d7ed52 : SUCCESS in 39m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d69055a0164b978e85c4211cc7e215 : FAILURE in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072f8e47d0654d9689569e609d552126 : FAILURE in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85fe33b995ba43dbb55e2d82a5b518fd : FAILURE in 9m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a9c7c906c74715abc090fca72e54da : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"51c375957897f5b54fd0a8b027b731d9139cd7e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-04 21:45:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c36aac533d101c7a72ba7afa6ad105d04534fb9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-04 22:40:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fb01d23e20a90a617d7e0f931beca5a6ae4618f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:50:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25067b5a0694c3698beb9c5fc742211 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/684562132b2a4153ae3c0c41a1207c79 : SUCCESS in 1h 03m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb1083a31b4c47a49f00d7949d9ed81d : SUCCESS in 44m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a5e5e1c7b214c568e44543352565a9b : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20046481f042427d91c8dd5a3495066a : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/472aad813d51460e9711c2b973cee9d6 : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6c60214362417c9296dca319ccfef9 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/948719dbd8f647929b7d9a0059625c8d : SUCCESS in 12m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4351a0b9bf4cb7b424f4c72c96b44c : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32908a5d34e042779917696461a4f185 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/53a893b35a214f8ba41eaff91a946564 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47b7432a29ce4baea75a42d80fd91fb3 : SUCCESS in 16m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bd2a09c24ff9447ba6dadf5d0b8fab32 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7bfb7189ce8941f19b2a273cfac81612 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/14c97dabbe2c4e27976738a6d7193160 : SUCCESS in 16m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/460a480c2fdf4c2d89f82d17dc0d8e1c : SUCCESS in 20m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/47e9f044a74645248a85c7094ade1579 : SUCCESS in 48m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/394ce1bb1cc44a85bf954cb579aec1dc : SUCCESS in 37m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7648419a494b4059b01f439802d300e8 : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58cbb3a39d8b4fb2895c84a1561b448e : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66005e3037d04a6183651d4572e30c31 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8e15118eb4842c5a7070a3c146cc6ed : FAILURE in 8m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/388f52626557486b8aa44894de5b2503 : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":3},{"id":"865ed02c1a9453d83cc289137e4844d9a5597e71","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-15 18:29:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nthis change seems reasonable, but get too ambitious and you end up with print statements\n\nhttps://blog.codinghorror.com/the-best-code-is-no-code-at-all/","accounts_in_message":[],"_revision_number":3},{"id":"4b7ed2b4184941c5e3f0770d5af6abafcb43d4e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-15 18:51:52.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nWill respin later today.","accounts_in_message":[],"_revision_number":3},{"id":"4d93b66d28e3b5083d6362215f27f559a01cbc05","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-15 19:51:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"008fa28d1cdf7175672926fd29d1e8b16286e132","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-15 20:23:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\ndrain_and_close is a nice touch, no more prints - tests make sense to me","accounts_in_message":[],"_revision_number":4},{"id":"29d28753b7de4a692730e2ab4c373614bc4e3690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 21:00:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448b13322b32425fbdd8c2212a6947fe : SUCCESS in 15m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1e6f7a71b5884a9eba6104fd36f820d7 : SUCCESS in 1h 02m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/161d2636220441fd8214dff399c6afda : SUCCESS in 42m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9a12688313d493ba354b4370e6d05ab : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b029301ecd7a4c3dbe15fb9f696f2f03 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d166a84e861b4f01a3c140f4b6a70e35 : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/247c82f112334c3c9fe06148fa3c1f76 : SUCCESS in 12m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13f6e4829c97404b9d5d8848c1d4dfc2 : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4656263f81844ae0acb416c4d035f82b : SUCCESS in 12m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bae4596be25b41e4b841fd217853cbb5 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/577dc1d2b3d74f0e9ca1d57fa56478e7 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ccbfb69efb04a7e8c0b66deff2c69b1 : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/94ad3a96380e468b9aa55a31babd8eab : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7490278b9b23407098a1e6095318f583 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/06aa438fa958456b88be6a3bb649f79a : SUCCESS in 17m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ba40b0698d14a2e93e601f8c3aa17cf : SUCCESS in 19m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f4c7b3cc2344f05bcabff2e003c80c8 : SUCCESS in 42m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af4304074dc048d3929492ed34dda7f2 : SUCCESS in 51m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b85e1817f99a4ad4b68fac5fd3d5daa9 : SUCCESS in 48m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee64159b60642aa988b0e0016e37fac : SUCCESS in 14m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b92dd9e37a4b4a903e6a3d8a89da1d : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01b19ed156b841fbb4430937e821540a : FAILURE in 7m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd9beb8487d44269a39f13c0e3387cce : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":4},{"id":"00b84c566810bc4b1df54c0f0173b7c25a9340e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 21:00:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7055dfe1b9487a2bb3395b6a443977ac366653d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 22:11:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa9ad25684e44a7852e246fc0a6a1dd : SUCCESS in 16m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/bcd4cb81ceb74aa691a8fe1c3199a661 : SUCCESS in 1h 03m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/587a7c2be3f647efacdb028f123fd6cb : SUCCESS in 47m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c0e87ea619fd4435892db721894d5084 : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7dadc09ce72f43e48922dd99969df060 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6142dbf7bf71470dadd83fd3251bda69 : SUCCESS in 13m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/890f6744db7244aab1924ab30136b40f : SUCCESS in 14m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1ba496f9282b4554b5ec1e33314d3942 : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc655966bfa444758e5d313aa4966a8a : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/672407f31a5c4b5e99de2b7bdb05e547 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce12108c9e184838829afe3db0b3ca34 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c03dd7a691c4da1b0e2038f32db13d3 : SUCCESS in 21m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07af2022124c45f2af99c498c3390968 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4596b206b116462680f4fb409f156335 : SUCCESS in 18m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b6186319cdcc4505a4c6b89d283ad702 : SUCCESS in 41m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c27ca3dafc6f4e26ac2a1352084b8e09 : SUCCESS in 41m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/098243741d7148d7a41d6a5abe5f9dc0 : SUCCESS in 43m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80fc5ecf6b274645816b450c0504794a : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac062d5a9c645989fc75a75ef3875a2 : SUCCESS in 5m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c9ad48cd48f49b8b86d5b015fe08a0b : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":4},{"id":"15d0acbeef54b59930990f5335f7c2cd19c1b973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 22:11:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d27cff7d147b120453463c916c1550892a1abf96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 22:12:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc23f599ee84b3983e31b989f0e1b8d : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1fb4b564f3854cf380c66d85b3ab39a1 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c3d262e22f4e485094fb5f2f08d33478 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a8d2146266431d3e9f57cb8fabae3adb42786360","revisions":{"e1c906d50cfdc459459c07dddd9b3c24c565a776":{"kind":"REWORK","_number":1,"created":"2020-09-04 18:36:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/750012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/750012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/750012/1"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 18:35:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 18:35:58.000000000","tz":-420},"subject":"slo: Drain error responses","message":"slo: Drain error responses\n\nOtherwise, they get logged as 499s.\n\nChange-Id: I7b54f25f42577020b10029c74f8fc01fa6fc591e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c906d50cfdc459459c07dddd9b3c24c565a776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c906d50cfdc459459c07dddd9b3c24c565a776"}]},"branch":"refs/heads/master"},"12ed3466350263dbe7c78ef03bd4585d36616b53":{"kind":"REWORK","_number":2,"created":"2020-09-04 21:45:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/750012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/750012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/750012/2"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 18:35:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 21:44:07.000000000","tz":-420},"subject":"xlo: Drain error responses","message":"xlo: Drain error responses\n\nOtherwise, they get logged as 499s. While we\u0027re at it, also drain DLO\nbodies (if they\u0027re small).\n\nChange-Id: I7b54f25f42577020b10029c74f8fc01fa6fc591e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12ed3466350263dbe7c78ef03bd4585d36616b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12ed3466350263dbe7c78ef03bd4585d36616b53"}]},"branch":"refs/heads/master"},"1bb124d444940f1f5bd5156b42538d58c4cc00bf":{"kind":"REWORK","_number":3,"created":"2020-09-04 22:40:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/750012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/750012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/750012/3"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 18:35:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 22:40:42.000000000","tz":-420},"subject":"xlo: Drain error responses","message":"xlo: Drain error responses\n\nOtherwise, they get logged as 499s. While we\u0027re at it, also drain DLO\nbodies (if they\u0027re small).\n\nChange-Id: I7b54f25f42577020b10029c74f8fc01fa6fc591e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bb124d444940f1f5bd5156b42538d58c4cc00bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bb124d444940f1f5bd5156b42538d58c4cc00bf"}]},"branch":"refs/heads/master"},"a8d2146266431d3e9f57cb8fabae3adb42786360":{"kind":"REWORK","_number":4,"created":"2020-09-15 19:51:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/750012/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/750012/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/750012/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/750012/4"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-04 18:35:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-15 19:51:16.000000000","tz":-420},"subject":"xlo: Drain error responses","message":"xlo: Drain error responses\n\nOtherwise, they get logged as 499s. While we\u0027re at it, also drain DLO\nbodies (if they\u0027re small).\n\nChange-Id: I7b54f25f42577020b10029c74f8fc01fa6fc591e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8d2146266431d3e9f57cb8fabae3adb42786360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8d2146266431d3e9f57cb8fabae3adb42786360"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
