)]}'
{"id":"openstack%2Fswift~750062","triplet_id":"openstack%2Fswift~master~Ifcf810f009a8036f250a09eec0d9a65a77342129","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ifcf810f009a8036f250a09eec0d9a65a77342129","subject":"py3: Fix non-ascii chars in staticweb listings","status":"MERGED","created":"2020-09-05 05:48:37.000000000","updated":"2020-10-07 00:32:03.000000000","submitted":"2020-10-07 00:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750062-1602030620652-f61217dc","meta_rev_id":"abaaaae686b487b1baece7f30ca962d0818794f8","_number":750062,"virtual_id_number":750062,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 00:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-09-08 02:59:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-06 20:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 02:59:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-10-07 00:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0567a99c2fda2a5a5a23226fdb226b59d33aa9d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-05 05:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff3076c32fbade73df9a14b2c2478a97ed94b448","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-09-05 06:08:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8635063c1b8f6fbcbf36615538a520ad2323c1ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 06:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d4cf157e684f2d8b1356bbf2b1bb4f : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/baf47d4371e44569aed8dc314989c10e : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fc98faa1785471fb274dde7eb999c2a : SUCCESS in 49m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/821b9fb269544974b51de57e5bea25a3 : SUCCESS in 7m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c596084b1624455aaa2e88e1e43f094 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71e84a1b375944ad8df89f201f446c29 : SUCCESS in 13m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7da0438d2587457792e61094f5b4e6a6 : SUCCESS in 12m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1a041a0cdf946bfa09d60304d880ced : SUCCESS in 12m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dee8cfbb7e1c48379f29cf4f9459ab86 : SUCCESS in 12m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d9d221bffa84aecaa7097aa6671cdd1 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5a8538a91e145e8a8d17aa9893847a0 : SUCCESS in 18m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/835b7b9c4e0749d482351ba0345940a9 : SUCCESS in 17m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fa06640d67844c7db207125c8a593c4e : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e58897350c5d486fb94ccefe1f0a1387 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77974861b8e743eda62514acc22cd845 : SUCCESS in 18m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a58d16b608944a08998773b8c13956ac : SUCCESS in 22m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6f1f1585190248a8a366dbcf426a0106 : FAILURE in 43m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce1a6e7e14cd40dda5ea685d5f2aaeed : SUCCESS in 45m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8885d2065504ca9842ec7f7c73f5209 : SUCCESS in 40m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8fa69c306a24b3a947b6e2c88c6bed3 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5701b8dc8f4f228bd41a02f4bd2fe6 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a27f6adb25394b0ab00edf705e5b08f9 : FAILURE in 7m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2d7cae73f0342f58dcddb884bf582a1 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e82792cd9457a2cf43360f37e1f85030599864fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-06 02:24:21.000000000","message":"Patch Set 1: Workflow-1\n\nSorry; this still needs tests. Just wanted to get something out after I ran into some weird issues last night. Fortunately, our standard Base2 func test mixin makes for some pretty robust tests. Shook out more code that needed changing; at the very least\n\ndiff --git a/swift/common/middleware/staticweb.py b/swift/common/middleware/staticweb.py\nindex 842a1e78b..7e25e6167 100644\n--- a/swift/common/middleware/staticweb.py\n+++ b/swift/common/middleware/staticweb.py\n@@ -135,7 +135,7 @@ from swift.common.utils import human_readable, split_path, config_true_value, \\\n from swift.common.wsgi import make_env, WSGIContext\n from swift.common.http import is_success, is_redirection, HTTP_NOT_FOUND\n from swift.common.swob import Response, HTTPMovedPermanently, HTTPNotFound, \\\n-    Request, wsgi_quote, wsgi_to_str\n+    Request, wsgi_quote, wsgi_to_str, str_to_wsgi\n from swift.proxy.controllers.base import get_container_info\n \n \n@@ -408,7 +408,7 @@ class _StaticWebContext(WSGIContext):\n         tmp_env[\u0027HTTP_USER_AGENT\u0027] \u003d \\\n             \u0027%s StaticWeb\u0027 % env.get(\u0027HTTP_USER_AGENT\u0027)\n         tmp_env[\u0027swift.source\u0027] \u003d \u0027SW\u0027\n-        tmp_env[\u0027PATH_INFO\u0027] +\u003d self._index\n+        tmp_env[\u0027PATH_INFO\u0027] +\u003d str_to_wsgi(self._index)\n         resp \u003d self._app_call(tmp_env)\n         status_int \u003d self._get_status_int()\n         if status_int \u003d\u003d HTTP_NOT_FOUND:\n@@ -465,7 +465,7 @@ class _StaticWebContext(WSGIContext):\n             tmp_env[\u0027swift.source\u0027] \u003d \u0027SW\u0027\n             if not tmp_env[\u0027PATH_INFO\u0027].endswith(\u0027/\u0027):\n                 tmp_env[\u0027PATH_INFO\u0027] +\u003d \u0027/\u0027\n-            tmp_env[\u0027PATH_INFO\u0027] +\u003d self._index\n+            tmp_env[\u0027PATH_INFO\u0027] +\u003d str_to_wsgi(self._index)\n             resp \u003d self._app_call(tmp_env)\n             status_int \u003d self._get_status_int()\n             if is_success(status_int) or is_redirection(status_int):\n\nI\u0027ll try to get this mergeable (with func tests) early next week.","accounts_in_message":[],"_revision_number":1},{"id":"529ee11eddeaf65a610f956639f9992ccc21c3e8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-07 05:08:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e23d61bc095047878a04d34a9ea57bdb4ff2172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 06:33:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb234b631ce41bdb41fb6d521d62ebe : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/c4abd1819c46404dbefcb75ab2ec5edc : SUCCESS in 1h 18m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5d8419ac1bc47a9a7d2a191e09cd6bd : FAILURE in 50m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17e2f0687d1b40958b5d89e0fcdd41bb : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e1c034a533a46ccb8559ca7407b3e22 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89730fcad37349788fd73a8bd6d8a796 : SUCCESS in 12m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51c5df8d549b47a698f9938e40f278e5 : SUCCESS in 24m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8b84f2a964a45e6a6687f818a4308d6 : SUCCESS in 15m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fa057a0051c4c22926fd09a3c1ac577 : SUCCESS in 14m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5afaee74bb0e4ca8bb73a765d62374b8 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec8ca77ea57a432786e7f304fc7d34dc : FAILURE in 36m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6565ce61f79849a79ac9095a09e9a608 : SUCCESS in 16m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9fc363a0b1ca465da6ccd16ce1717655 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/289777e8ec5a465d939a3a336a38c3e1 : SUCCESS in 36m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f0f6d33b2ed34aa0ab506d63a26e52be : SUCCESS in 16m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f18a8ee91d8e46d0bdcfb68d3b786672 : SUCCESS in 42m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e57bbf9a9e74a588b4b2e74a3fdd0d2 : FAILURE in 48m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5559696f9f364c26a426dcff037f4081 : SUCCESS in 37m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/470d17fe746c4792a81146a222d3a82e : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/947eca8ffa3a41909bfdb2bbcc1df189 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dffeb10c174c58abc74d345447cb27 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb3de8bd96574ed6bcdad9f57c706e78 : FAILURE in 10m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2486543d9304225b0674bb9a417a828 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":2},{"id":"be47b7f42e8fc03da709bd58bead194ee6a58587","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-09-08 02:59:27.000000000","message":"Patch Set 2: Code-Review+2\n\nwhen will this pain end","accounts_in_message":[],"_revision_number":2},{"id":"a6b085627553e0b344ef3d0440da5503a6fe9a5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-08 14:47:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d296ec197cd5916befde18f6162dd3d94af7f424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 17:39:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad0142b0d934add89bf1a98f55ae102 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4823229e20764d4c8fcf8bb571948984 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2585af6185194e499784d4a19f1847b7 : SUCCESS in 46m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d655f05cfb149d4b4360d5d89ba8d0f : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30b2f28fae8443aa99d134b9a0ef6275 : SUCCESS in 11m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15778a861cc844738ffcb81adbcc2672 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e95f612c74254175bab8d1cd6537084a : SUCCESS in 14m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2d04f4401804c53a4761c25db6ba284 : SUCCESS in 12m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e45a17c021bb41f996390087c7d5ac6d : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b564a59e23047cba91413498acb26ae : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31f688fa63034061b435fef5904c2a52 : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ce1d41b992443adb06d6ee83737f118 : SUCCESS in 19m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6bfc3818dbfb4d1ebd0036d1b7ffcb14 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/efe940ec94324060a3545475a542eeb4 : SUCCESS in 22m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ea80c25a59d4a588264dd504dff9cd1 : SUCCESS in 20m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3dc91a7d4a3e4cd6a3440632ac376d90 : SUCCESS in 23m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5762049419e4b03bda3815806d6e994 : SUCCESS in 46m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad5ba3a7787b476282fe0494f9b9e075 : SUCCESS in 41m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0637ed69b7204f8fa3588cb63342eabe : SUCCESS in 35m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33c4dcdc31ac458f8bea3275632ff80a : SUCCESS in 27m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72382b377040419eb21303a1a8f13671 : SUCCESS in 8m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ade4e3b48804694bbd1e83cdff226ad : FAILURE in 11m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9d3a62cf6004adfb41cc44411e8b7eb : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":2},{"id":"caa9cfd1e3eab480c6b481ab374c17cc82303496","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-06 20:43:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5fb2e575139194922a17c490a7c60575064628a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 20:43:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd5ca2491907a63deadfa46ba4ef11b79e073dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:30:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e079e65f7b4f78b688e6ebb58a250e : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8ccbfaa0b1f6467aad99a82d9517120a : SUCCESS in 1h 03m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd6263c428e04d898a6fbd1eaf347423 : SUCCESS in 52m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0a4cfd0a5379485cb96e410d8bb54640 : SUCCESS in 9m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e98509664ee4451ea276c194201371b6 : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45e2e7043a0b49c4b15836a5ed8e2402 : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7199ef75e19e4462b8233b5ce04ee428 : SUCCESS in 12m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4d8522277f44f719f5cbf552bb370b0 : SUCCESS in 12m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e260d1b1aa9247a58c3846a55333c3c1 : SUCCESS in 12m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24bbeb97629547a793cf5f4946059ec1 : SUCCESS in 22m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62362c9022264ebb8c3f86563b0b2163 : SUCCESS in 21m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2aafee1c2c9c48c1aaf27d4a251d4ab4 : SUCCESS in 18m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e9a696ad10c64d7283e4be310ac7e67d : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d14007f699ed4c7a9e73d2e1bf295b09 : SUCCESS in 21m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c9fd8ab71cc4fa3bee833bbea8ff719 : SUCCESS in 20m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/458d9ed2352c48a98d4e9a75d671d3f6 : SUCCESS in 45m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e30e5b59e22e41b7a33da7249bf9bffa : SUCCESS in 49m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0326d1d75fc4cc79be0cdd02f5b01c9 : SUCCESS in 43m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6886d1f47fd94357881f75562f714f3b : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6c3d4bae504d059ba308d97699ec06 : SUCCESS in 5m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7161d589ad3745efbedb0c754b5a6b73 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4fc422f38a6f4211bc4947f442e77e33b116dded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"abaaaae686b487b1baece7f30ca962d0818794f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:32:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5174d2bd4a6242fc9606ec6a6da3f932 : SUCCESS in 1m 19s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f0f9a4d69b8b4d91a718764bebbce722 : FAILURE in 1m 09s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d82a0ea054424a73bcee1516a7ef39b5 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba46eb0e3d7832924da98ca84eaad40e37c75516","revisions":{"d804b752b7ded5e97530b1fd573b21050a4e02f5":{"kind":"REWORK","_number":1,"created":"2020-09-05 05:48:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/750062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/750062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/750062/1"}}},"commit":{"parents":[{"commit":"7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c","subject":"Merge \"py3: Work with proper native string paths in crypto meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7015ac2fdc34bbb46b6be6084fe6a5bd41acc74c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-05 05:47:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-05 05:47:54.000000000","tz":-420},"subject":"py3: Fix non-ascii chars in staticweb listings","message":"py3: Fix non-ascii chars in staticweb listings\n\nChange-Id: Ifcf810f009a8036f250a09eec0d9a65a77342129\nCloses-Bug: #1894357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d804b752b7ded5e97530b1fd573b21050a4e02f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d804b752b7ded5e97530b1fd573b21050a4e02f5"}]},"branch":"refs/heads/master"},"ba46eb0e3d7832924da98ca84eaad40e37c75516":{"kind":"REWORK","_number":2,"created":"2020-09-07 05:08:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/750062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/750062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/750062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/750062/2"}}},"commit":{"parents":[{"commit":"cca5e8b1de162dfd65d73bf1971333af14a0dd17","subject":"Merge \"Make all concurrent_get options per-policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cca5e8b1de162dfd65d73bf1971333af14a0dd17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-05 05:47:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-07 05:04:06.000000000","tz":-420},"subject":"py3: Fix non-ascii chars in staticweb listings","message":"py3: Fix non-ascii chars in staticweb listings\n\nChange-Id: Ifcf810f009a8036f250a09eec0d9a65a77342129\nCloses-Bug: #1894357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba46eb0e3d7832924da98ca84eaad40e37c75516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba46eb0e3d7832924da98ca84eaad40e37c75516"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
