)]}'
{"id":"openstack%2Fswift~750535","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91","subject":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility","status":"MERGED","created":"2020-09-09 00:20:47.000000000","updated":"2020-09-25 06:28:37.000000000","submitted":"2020-09-25 06:25:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750535-1601015141499-dddb1e4e","meta_rev_id":"6840ce37aa1e73a9cd066031e4d8d3d6156202a9","_number":750535,"virtual_id_number":750535,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 06:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-24 03:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2020-09-09 00:20:47.000000000","_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-24 03:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 00:20:47.000000000","updated_by":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"reviewer":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"state":"REVIEWER"},{"updated":"2020-09-25 06:25:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04d7b41420ec9e3eeb69964bd96f3f9faab2eb34","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 00:20:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b1927c1d09e241ac989a0b5f530350ede0197377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:29:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4494d053614450fa7952f6e326d4671 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/7bee347dc93540138e49f3ace9650eb3 : SUCCESS in 1h 07m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38d0d664a3c346ae85faea8525074886 : SUCCESS in 1h 02m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/093d072dfb0a4dc691a53fa2d56e0864 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/687b48011e0147b7a5eb328d888fae9d : SUCCESS in 16m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e464fd2198904c15bbf2ae760455abe8 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/321f6b11f6d64d33ba181e3f6fa3f470 : SUCCESS in 12m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc18f68424b9403e945b193d18172717 : SUCCESS in 18m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1a77ba3d3d01490c801ee839b4101d6e : SUCCESS in 20m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2514463ea03b44f1953b95796078595e : SUCCESS in 10m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8156c8defe441d490f1dc7e26f49afe : TIMED_OUT in 41m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4f6715d67b244363bffaf2fdf5be2532 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fa3123f98df447eaa8d357d05605bb2 : TIMED_OUT in 41m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/72d9c35195f04bb8b8f82df807bc1716 : SUCCESS in 18m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a4e69b0b91149f9b8d1d78c333e389d : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/79f110e025f24b23a36b948818d2bb2e : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f561ed46531e4eb4ab8b18026b6ad84a : TIMED_OUT in 41m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fdd1f052c1ef4f66ae9e6b5b11ebcfa2 : SUCCESS in 9m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0147c89881c040a7bd8d316dc1074ceb : TIMED_OUT in 41m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c532a4f9f004ffb83936c957dec2052 : TIMED_OUT in 41m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e95fdd39d4074816b152e6049c52a1b5 : SUCCESS in 35m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4b101f040244891948f4cf18b906a40 : TIMED_OUT in 47m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ac5a7a70313e434faa3f0a339c0fba83 : TIMED_OUT in 47m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e94bd322102494fa6a7a930894ac275 : SUCCESS in 23m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b81cdc6a7e0e458994e386f2a047e21b : TIMED_OUT in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaec93cf133b4f94b48841f63f1e2f71 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/067887c2c72f46bc8f12c5ee1054deb8 : FAILURE in 14m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a8a2055be44153ab675abadc329f9e : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3262799b6e46961b4a135e44134a4a24d063d3b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 23:43:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acc3bf5d661da66b48750b34b22c99e4906a3278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:36:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13891f9c01294f968adb4917354ee9ea : FAILURE in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/66e6dc95ebdf4cdaa7b5d37fc6f20b62 : SUCCESS in 1h 06m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b4b02ac11419401ebd8cae1707d92faa : SUCCESS in 47m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b27c79bff5314b3ea34845f85c46a077 : SUCCESS in 10m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5cb2030460d48caa54c07df6583f117 : SUCCESS in 11m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43323157172b4eca8689ba6145f9b041 : SUCCESS in 13m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a42b2de91ad4b1aaceb909c4c18d257 : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf760b95317442c6bed4d15c98094433 : SUCCESS in 12m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/74b0a2023a15492aa0d00cf00f77b7db : SUCCESS in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba91f8b02f444414a7b49908752007b9 : SUCCESS in 10m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20f5d63cb3294a25b9ec46937a0b0cca : SUCCESS in 14m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/11378420baf546ef8056010e0859bb2b : SUCCESS in 10m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3f42501728e46ffa81edf9b7ff23861 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c4432bd309d547df81682bb8db4b954a : SUCCESS in 15m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0b558be80cfe4e3e825dc88c14797525 : SUCCESS in 10m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1fda529331144f379651eed9a231c1ca : SUCCESS in 14m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2d215d568fc54054bbcf955e83418d78 : SUCCESS in 11m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/84ca0c85a04e4ccb9e780dc4f5f1baf8 : SUCCESS in 11m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4bfb8b4e96b848ee8e3bd87e96e7ba61 : SUCCESS in 16m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71ecf663139646b299efc6e559138f1a : SUCCESS in 21m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/755773b888114589a725ebdf4a25641d : SUCCESS in 44m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc29307ed39c40559d116e4cf6313fd4 : SUCCESS in 36m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/397e00e503d7462e8f608ba343a17614 : SUCCESS in 35m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/febba6de2c8841ad8d0821cf0867d35b : SUCCESS in 32m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbe4223652dc4b2c8365bf1346d1c60c : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca82d4cd16d40d7a4c9f555f2f77df9 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b18b231a5472413f8ba9c688061279f5 : FAILURE in 9m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e629e078854495d85d8d522bc41962d : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"40634cc6b3b72c8c5439723d2740498479c658ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-23 20:53:01.000000000","message":"Patch Set 1:\n\nrecheck\n\nRecent stable checks seem to be passing -- maybe the docs failures were transient?","accounts_in_message":[],"_revision_number":1},{"id":"3763d2c8c9370c7edc1ca794e7ca53904100bae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 00:06:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d572600d855c41eebe157c8547d0d4be : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/1ebcfa61e3fc4816a5565c616cbf8efb : SUCCESS in 1h 11m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7d24ebed43f4c8ca5546ccbd3f84df8 : SUCCESS in 43m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c9e99c8fd1347d38aaf7d2c5bc3cfea : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bac50e96ba7452198a419da605a799c : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1051d15ad79d4a6984060aa59168fea0 : SUCCESS in 11m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c1c78708f644522bae2870555c01a2c : SUCCESS in 29m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/298b19c4811a4130aad9715d3a7d1ba2 : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0f91a79540eb4f3baf9661e3ccee6ee8 : SUCCESS in 13m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f42c8550b0347c9b9b8cf41442e3c37 : SUCCESS in 9m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ca5bcdefa6d04903a502cf570c150056 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cf6d4dfc303d48af9f7f5777472a8c09 : SUCCESS in 10m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ea100f9d3f249608dd4fbfc0e68f862 : SUCCESS in 9m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c825a218c4e54762995a2a0e018a3d5e : SUCCESS in 15m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a0092a11235b435e97a13d84208a4886 : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/78bbe6b6d5f34d00ae44d7de89ab8575 : SUCCESS in 15m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4a4a512ffd404c2dadc2d5b23e3eb86f : SUCCESS in 13m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bcc4d8c1f652494dad3fafecd05515ee : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/353529a14f87465c851e4cf790b94868 : SUCCESS in 15m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7792660742254ff49561d34ad5f55595 : SUCCESS in 20m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab65c71dbc0d43fe9393a44ff4a3ad56 : SUCCESS in 43m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cb4607538bd4661af5c9f34209ba7ba : SUCCESS in 34m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/332e583dfe6c41aaadec8069a3ae4594 : TIMED_OUT in 49m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/887cc219b7e54bdc80a93698c59515d5 : SUCCESS in 25m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bbfc988002f4d219474bc57e334fce1 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e58839995d4e25939b6c5a69afcd70 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/279c96ad43d542d791f302863b4c190d : FAILURE in 11m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/670a6ca0aaec412580c067d43028b9ae : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f5a6f9fbf1747cdb046b24539f9e9da9ced0a317","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 03:51:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d61c8a020cb483e1b765b1010c45d76bb26ba724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:54:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bee551a880e406da8f75f251d7142cc : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/05d3b621dbd643d78e2792e5b1659448 : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ac74b09dd3b4a4d83f6270d270a9959 : SUCCESS in 37m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9c625191ea64accaff96fe9a34d9b83 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90d90a1567b94f7791351db2bce9786a : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d95990761e8f47598bf648b6b7ff5b27 : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/75599f6171ad47b7b8c8d9ee3a766fdd : SUCCESS in 12m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f6245897d6346e0aaf0f0a2d2e49db4 : SUCCESS in 12m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d3ca75085824fbb98568127d82aa34f : SUCCESS in 11m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/293e48ffa2974627861e4025239aefc3 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ffe317daf754bdd8ad0057f02d97e16 : SUCCESS in 11m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e8a7660f78ba41d7815b86c3efe9b5e5 : SUCCESS in 9m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c761ca20f26045b5b7f7a9dd8d935af8 : SUCCESS in 10m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cbff0c35f7c647018f692142a2dd8362 : SUCCESS in 15m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/327c89d4faf44d58bb706fb92433959f : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0ead5a9170db4d709884377dafd4f984 : SUCCESS in 13m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bd4688a861374403a8cf54666654229e : SUCCESS in 10m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9eeb6c6ba92342858b16efa66af50cc1 : SUCCESS in 11m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/87e820559f5f478e8706cbd5180aed90 : SUCCESS in 15m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b23088eeb7584d018e3a6929fadd4591 : SUCCESS in 26m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df5ae8959e7e42acb476b283d9d6096f : FAILURE in 39m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e83f77f885a413bb6cc19136d99d791 : SUCCESS in 27m 50s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7e16f1ecc09c4805b0a6db422aaae191 : SUCCESS in 26m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f32dd7d0621746fea6e65c3844924fb3 : SUCCESS in 31m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1d883c362b4204ab5871cdc6e4b2f3 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4fd1cb795846e68ecb119f358a5ff7 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e590238ee4f24a18860d3fd7d6bf984c : FAILURE in 9m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26506205591c40a9bbdd4023e80487bb : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"46b1ef0052b54597dd5c3377deb8e4a64cff7b50","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 15:01:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a98cf1063c6b9e2d0efc8a3b1baecf5b91dcd72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d790534654d3451d9daafa37793a1d4f : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/36e655df4dde4e64ac8a4ba3ab3edbb2 : SUCCESS in 1h 02m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5f7bcad818f47da91e538bd9e5b60b1 : SUCCESS in 40m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63977c1fc23a4b2b9dd7b0e61ccff44e : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f9f63c90d28b4b62ab6d209c36b41fb3 : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3616da5281ad4090a0c6a18092043db6 : SUCCESS in 13m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/978f27b26ab844ed90892ded04df9218 : SUCCESS in 11m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eb1f77be7cf84cb6a4f2d6c1534e835c : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3970d2a650e342cbad60878bbc23d5b7 : SUCCESS in 12m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3a16cb2c4da4edaa396e425524bae32 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70db7f6e91c043cbbcaad53840f71ef6 : SUCCESS in 23m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/383a7daf5771468489bf719b34f3b68c : SUCCESS in 10m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8231a3e76d434a849f8496ac8704426d : SUCCESS in 9m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e259858edc2a4e1c937abbd79945f254 : SUCCESS in 15m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/254c8cf498f84d90afb7389dcbfce81e : SUCCESS in 11m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8299443a18ee4f4bab718d5a150bd0f2 : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3d8d7d8fa409406283d3e3ea74b68062 : SUCCESS in 9m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e1e0ec762dd4453a8bebe1bc91f729e6 : SUCCESS in 12m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d8ed8b9021c64c53b7b3a18da96d0328 : SUCCESS in 17m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/622ed56abdd643089484706065e626ab : SUCCESS in 21m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/11cdff5ba23d48e5ab7f94008d5250e9 : SUCCESS in 46m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c912f47b7a84f41b751bd95dca4abce : SUCCESS in 30m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1d18d5f6606e4b04a029b28b03fb0055 : SUCCESS in 26m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e1c339f92a214d3ea7d5d02acd7db2a9 : SUCCESS in 38m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88062dc1e8794a1c879f41158bf9e50b : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6a85e243c046d6a1bc30d404581ce7 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb309972c9ea47c8a561677f7a900aa5 : FAILURE in 10m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ba0f3a1f4214fc0bcd2a4b2778a0cbc : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5e358ec672ea8760f21bc0c834857da988870d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:09:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddad5495812d38de0b478acb755787f394e660c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 02:41:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3c9308ce4c45fd967cddcea5a7ea15 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/70af4adfd14248108faff052d91e4457 : SUCCESS in 1h 10m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad1c3c8676e045939e7c9e611aa6a9a1 : SUCCESS in 40m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9b770d69e8f843839c8c531f12f10ed0 : SUCCESS in 12m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b568ca037b6a40a988530a2631633540 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/866154f3820a4901882991f3fd1a5cb7 : SUCCESS in 12m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/605ab65e5fb349209e643a57a15925e0 : SUCCESS in 12m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71bd0bca7fe3484dbaa66060957da291 : SUCCESS in 13m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6aa00e51e1c74def91e133e3084bbfcd : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/291e008c997645a8b5e6a55a263aca68 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/747a52581d734db3be7a492bc3d7e403 : TIMED_OUT in 40m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f53a255dba804386875a86ac2ecb67a9 : SUCCESS in 10m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e59afad723824f509ffc164e3840f122 : SUCCESS in 11m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6f293534c5a74c7f9faa005ee100783b : SUCCESS in 16m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/50ed7c1391884cbf9b24f4ed9c47b0ec : SUCCESS in 15m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/02015b539522460a91effcdfbbd6f025 : SUCCESS in 10m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2cd4304975594a2991cca1b596fc72aa : SUCCESS in 10m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f44bf905df764b69b3a7324a0a10efec : SUCCESS in 15m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa53ba1616cb4c03a8d249c3956c48f5 : TIMED_OUT in 1h 04m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6e23ad147f5400f84b22a3e4a114854 : SUCCESS in 30m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c8e8dc5b879c45d78a1c45e7e14575c2 : SUCCESS in 26m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45d06bf82f5f4379a4c9ba4f7177831b : SUCCESS in 34m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/704a1b5b7a064cfb8dd11267d5555b95 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f654253f1b1742ce87914cde8f53832b : SUCCESS in 5m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413ff168cb7d48be8b4d56896ac4f41e : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9b7c84e52e93c095721fb96f8b1f60bc904a78cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-25 03:56:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a16d6cec0d08f4a061bba1062503a56141ff2f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:00:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3023669c1a5d414185b15d6d2b071924 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/aa3409d7b0fa4884956feebb0b1888e1 : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e7c8d0b7c604bdcb0bc60dce9a7b6c6 : SUCCESS in 37m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68c1d7bc7fe24329b6358d74bb07b34b : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/99fb5a27f85847a8b00b04a44d0a37ac : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7384926b95954009a743568b744cad49 : SUCCESS in 11m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f7df5deb7c24b96bc27891c2cf32491 : SUCCESS in 12m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fde9eb832f1a4a92baa4c01e33b89eac : SUCCESS in 12m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c12dd49c9cc4dceb1328f9f11570566 : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55500fc9f6984bf3b730bf14867d2fa6 : SUCCESS in 9m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/273acebbf8c74fb3b226cf81141ddf6c : SUCCESS in 13m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c8d03047797a43068eb3533d8b3017b5 : SUCCESS in 10m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2bac62d90744d45b0de40b027e76fa5 : SUCCESS in 11m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/42199e0390e846b9b2996a66e93cb264 : SUCCESS in 19m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8ae8fb73037549f1a3f33966d9aa62ab : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1dae421b7e54b0d86466fa794c0e7f1 : SUCCESS in 13m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0f5264f3ff7f4107bee63662a0584475 : SUCCESS in 10m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7336cd296e9e46e5a68efcd88cfc51f4 : SUCCESS in 11m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/32601f9e5db9415e9b12764358013d4c : SUCCESS in 16m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94047d1a8bbc4c0fa2d401a8de8e95a1 : SUCCESS in 21m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3668ea708e484857b2a0c4108aeeb754 : SUCCESS in 45m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f826181df8074d9da62d4c96a3a401a2 : SUCCESS in 28m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/93b22462e4a34f43907bbe1b45458925 : SUCCESS in 25m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b315d3f14374ab094c555a19307160a : SUCCESS in 31m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67d880b9c0e046ecaa0d280a69fdf701 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b585f993604175b2b4ec795cbb208e : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/858cb16608724649ad71710bda785cc1 : FAILURE in 9m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4864c8a83ba7427e86237a655e18aa7f : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1579bdfe05f068d03fa1e1d5fef2a62d7f7f61cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07b63745bf06d2a3fbb6174310baaffc7c37c617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:25:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1b04e76e2c44ce8727b7b978a7173c : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3b81db49b8d244b7a0c0075a52aec6e6 : SUCCESS in 1h 18m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd9f927801c540588b3b182eb38f3b7f : SUCCESS in 51m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/975191cbe4734d898252e18bab30e2ec : SUCCESS in 8m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/87b35821256347ceb36fcbbcb1a36295 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c8e5479073d54c02a07b5216c743c5dd : SUCCESS in 13m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/640cfebc9e374ab6a1564bfdd7b4a665 : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0fcb289d71bc49cdb21a4b6fb0334720 : SUCCESS in 12m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dfd5e23be4354ac7bcdefda9306f92ed : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/310adfa528444167867c480aab891927 : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c9fba715869f4ce8bd7686b720bf1098 : SUCCESS in 11m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ed2b82271e94469cbc2c60d1892f5007 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5707b0e975dc42fc819f07661f482768 : SUCCESS in 10m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/30e49d0aee6b4ae5900785866ce6fce3 : SUCCESS in 15m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bdf8ce419f5340279f3a37bcf0b9deb3 : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d0a2fbba63c5442eb3c3da55bd1188e4 : SUCCESS in 10m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/de32f7d535d24462a392d4483176a531 : SUCCESS in 10m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5c529ee99c0942bfbaf590bf0ca1a705 : SUCCESS in 14m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/737254d439134e7ab425ad61cefb7397 : SUCCESS in 42m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e45caae4b3e84e57b5f06e08dd269178 : SUCCESS in 32m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1956fe85e3c548d986c5ed019c77ecfc : SUCCESS in 29m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/966b316f51a84b94a9278420915e8a7d : SUCCESS in 28m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d648c41fce954e2396d5f9f7f82761be : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49997723da814f73802ee74bcfe74e07 : SUCCESS in 5m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9214ac562d8e4d0687c607e5e682234d : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"edf0e0263f41cb545c887c74abd60ecbb99350cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:25:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6840ce37aa1e73a9cd066031e4d8d3d6156202a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:28:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f625887754064728a3ea7dc45a82be99 : SUCCESS in 1m 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cf7c6e6ace0b4ac892aed4a4fc878aad : SUCCESS in 1m 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ffcdc423b8fa4cb9a3db1fb08c742bb0 : SUCCESS in 1m 56s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1e152e880804918d66caa71cd5e12e34e2c91f4","revisions":{"a1e152e880804918d66caa71cd5e12e34e2c91f4":{"kind":"REWORK","_number":1,"created":"2020-09-09 00:20:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/750535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/750535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/750535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/750535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/750535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/750535/1"}}},"commit":{"parents":[{"commit":"8ed74c264605e12717c93e9b07f4ff50924e619f","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ed74c264605e12717c93e9b07f4ff50924e619f"}]}],"author":{"name":"arzhna","email":"arzhna@gmail.com","date":"2020-07-22 00:49:17.000000000","tz":540},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-09-09 00:20:47.000000000","tz":0},"subject":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility","message":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility\n\nWhen I call the S3 API using the AWS .NET SDK, I get the following error.\n\nAn error occurred (AuthorizationHeaderMalformed) when calling the\nListBuckets operation: The authorization header is malformed;\nthe region \u0027regionone\u0027 is wrong; expecting \u0027RegionOne\u0027\n\nThe reason is that the AWS .NET SDK generates a signature by changing\nthe region name to lowercase. (AWS region names are all lowercase.)\nThe default region name of OpenStack is RegionOne, and custom region\nnames with capital letters can also be used.\nIf you set the location of the S3 API to a region name containing\nuppercase letters, the AWS .NET SDK cannot be used.\n\nThere are two ways to solve this problem.\n\n1. Force the location item of S3 API middleware setting to be set\n   to lower case.\n2. If the request contains credentail parameters that contain the\n   lowercase region name, the region name of string_to_sign is\n   modified to lowercase to generate a valid signature.\n\nI think the second way is to make it more compatible.\n\nCloses-Bug: #1888444\nChange-Id: Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91\n(cherry picked from commit 3a6e85d9ec9a899c62400944832194b3c3a6a737)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1e152e880804918d66caa71cd5e12e34e2c91f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1e152e880804918d66caa71cd5e12e34e2c91f4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
