)]}'
{"id":"openstack%2Fswift~750536","triplet_id":"openstack%2Fswift~stable%2Ftrain~Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91","subject":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility","status":"MERGED","created":"2020-09-09 00:21:04.000000000","updated":"2020-09-25 20:50:12.000000000","submitted":"2020-09-25 20:45:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750536-1601066732463-428ff9be","meta_rev_id":"afb20ff34e648531b7a82e26bc6a5e8507cafa86","_number":750536,"virtual_id_number":750536,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 20:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-25 19:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2020-09-09 00:21:04.000000000","_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-25 19:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 00:21:04.000000000","updated_by":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"reviewer":{"_account_id":29586,"name":"Arzhna Lee","email":"arzhna@gmail.com","username":"arzhna"},"state":"REVIEWER"},{"updated":"2020-09-25 20:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad6fb841d4590f3ea7bad8eb8a15d199fd7d2f87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 00:21:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"90438c54cc7234f791dacb6719755c2c8684c4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:36:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bad0b1b5dc948a49ebb0077318e2129 : SUCCESS in 6m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/baeaab6fddd641d8a202eb17269b1e0a : SUCCESS in 1h 13m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60864df84c38409fa7d9d9945c78a056 : SUCCESS in 56m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2fc6cd99cb14d7ea9dea5f0c103effa : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4b068fb11a046e09382d6d529e84097 : SUCCESS in 37m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/75fe313fe5f74354a11e8a98c17ae233 : SUCCESS in 36m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3b98eda2fc649f5bac2849942a3d55a : SUCCESS in 37m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bff3852cacfa431a98797c8b54a3d0aa : SUCCESS in 36m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c200d91c97164a2bafb36c69f01bee93 : FAILURE in 36m 58s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/70035d5d78ef49c1b962f669de423da5 : SUCCESS in 36m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6120be06b76b43c7a27aabc50c024466 : TIMED_OUT in 41m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/394b58cc78bb480a8872b32da0c1ccfe : TIMED_OUT in 40m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9d4bad565aff4608aa37178d3785fd33 : SUCCESS in 41m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c077d9de0f3b42b1a515ff28093b6644 : TIMED_OUT in 41m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7611ad8346f54372884ed24f007d6f9c : TIMED_OUT in 41m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3694b57846c640fb9104b9dd0ae991a6 : TIMED_OUT in 40m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/36542dc383574784a4eb295e92a3b75e : TIMED_OUT in 41m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0399fa579b924f6f9d6ccc74e882fd0f : TIMED_OUT in 31m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/976012d3a0e8469e9606a54b3179672b : SUCCESS in 56m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36c8f2c4cc0343b3aa57d55affd7e828 : SUCCESS in 26m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/48ab7c42aaab4e3d95b6b2a04c75d5ec : SUCCESS in 24m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b7cd61b4b394dfcac8e76ec5dce729d : SUCCESS in 28m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bebe29dd184345e48db20eef6eead1fb : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75a0549dea04ebc80320b69a3fdb9c1 : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2aafff11aefd4808a8a35b39646df177 : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e37c39b18d924307b5363ea86e51d8f6 : SUCCESS in 55m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c9ae4975425340e69a9e11e20b1eeb43 : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1eaf0c7bbce8162df6c9093f2089fa6ea5b82a1c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 23:43:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a95b3df16eb03a0d8016b061209a86e043078e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:35:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cae8faa64645f89bb6217e65e79bba : SUCCESS in 5m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9792508a91f6470fbab30a8078cd433d : SUCCESS in 44m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6e58879a93e4a26b40280ad6f50ec02 : SUCCESS in 34m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bf297e56b444da3a0ed74b40420ce67 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f63074ecc194d3ea104ca08455a5ac8 : SUCCESS in 9m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/38897118b54a412f8ebcd674068c663c : SUCCESS in 14m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f581c553f39f40bf8e55d7ab2de88d0e : SUCCESS in 12m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78917ac5fdf348ba90ccedbd781a0a1d : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/734eccf7962348e5b854e76f000b2e28 : SUCCESS in 10m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b1f606b6f5304b14b9eb21da7adcce45 : SUCCESS in 11m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/600e8bdd0cf04a86b319f1cf342d3a97 : SUCCESS in 10m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/60dcaad528414ab5b8c926f636037173 : SUCCESS in 12m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/71ff2b729a674bfaa8754f27d656c65a : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d7091ffed1af44208d58f62610c4563f : SUCCESS in 10m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8ac891ddb36e4deab66d2bb105ec9aed : SUCCESS in 10m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c43cd4b7f2a497ca4dbdeac5bad2c7e : SUCCESS in 11m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/204b36252cc14aa9aec826da7238bc4c : SUCCESS in 14m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f603f51976f43d2be3183190c68506e : SUCCESS in 21m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3404ecd6b8da4897a80e8acfceaf4c8b : FAILURE in 19m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1e9553e1686472097945d5baffe6e7a : SUCCESS in 30m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c43931a7cee6413095ca5ba005f66f8e : SUCCESS in 30m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/578abe0165da4c6686b131263cc5912d : SUCCESS in 32m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20e9438ca5394ada9a5a5e0dacae3362 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8092d9ea61dd48ad8e183e767dfcf36b : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd4f3ca8c55f4681aea5428e27f8565f : SUCCESS in 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2adc0b16bf1549a8b547a07507002afe : SUCCESS in 1h 01m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e87ccb18e6140ec8e17fefb63a5bbe3 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"289fb087ecf23c7b194d3a27cef81b60a6c70a51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-10 16:48:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bb20f8bac3b36c8910660a0de28ff6c9734d516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2fedee49d8c46faa474556eed050120 : SUCCESS in 6m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/a2982d402f0d4078a360b516d6ecb7a1 : SUCCESS in 1h 01m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a53b9b45fb34b1e855b3f5aad3aa8ab : SUCCESS in 48m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e39d21a9302441b69625ca05a3465827 : SUCCESS in 12m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8204adc7f0114413a0cd7815498f0795 : SUCCESS in 15m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a72570d0e72b478eb9e70501411c56c5 : SUCCESS in 10m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ddd5b0ff4be479188c71b7e5f88794a : SUCCESS in 11m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5979ce7b763e4c7481dfe9fb8f1b3d77 : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19479293dca04ec7a3b1f6528160d5b4 : SUCCESS in 17m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/93961f72a78443208a10ce9ce363874e : SUCCESS in 10m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5dc13a9ec9df40b4835235d92a80aca4 : SUCCESS in 13m 44s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8abb2c96de2d449aa54fad31b64eb767 : SUCCESS in 17m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/870a03c8f1324fe4b5a7437fce8ad6c0 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3963a7a6f20642d89fde43db3707eeee : SUCCESS in 10m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d2cbcfaea20e467d8ced43577b7a37be : SUCCESS in 19m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5f6350c518f5416daa58bb7ea52d287b : SUCCESS in 18m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/da1e18bc31914ecf810288fc6aabee52 : SUCCESS in 15m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b324282e26a648a9b992d3014320b7f8 : SUCCESS in 19m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a23f58587764d5b89a305bb2ad1d9a4 : SUCCESS in 27m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9bfb106b52f4c9cbccec06a05a33f88 : SUCCESS in 32m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/846aace219fc4217892e856ab79a944b : SUCCESS in 34m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40b698aa0d434bb2b2307366da02d6e0 : SUCCESS in 31m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1948608e16fa48e193fc231b6796994f : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea3e2f16a3c468896bb1ce279e0201a : SUCCESS in 8m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/407b64547235416498b7881af0b0ffa0 : SUCCESS in 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed833cf278f3403abf2481df0c5a5c8a : SUCCESS in 55m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc6c9ebb93974572898b5fd52abe2728 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"80b98abe441c775deb307e3817f2d9ec7c9295d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-25 19:39:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2dd6946ca047b182b42a8be9921ec23a02785b92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 19:40:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"367c8c55a9c8eaea5151283d59b60fa15729eaa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 20:45:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee54eb9b83347a0a882c47c6a283042 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/5f81ee8881494484bd6b1ba09e041eae : SUCCESS in 58m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b41f684cdd5347c59c6f6c94b7be4397 : SUCCESS in 42m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/773ef567ce1840d2b54306f831c4a048 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77078d4bc13d4af093a1d21bd03ba2ff : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9dcd9aa98eb452e8357ddb3b61797b1 : SUCCESS in 11m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fff51ee134714d1aaca0f9945aa6b4d9 : SUCCESS in 11m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82af7ddbf8624d6ba7e322e173bc5769 : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cda8d69cdc954117aa22376a9f4b110f : SUCCESS in 16m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5711cd29fc654ca396ba7a204b0e75d9 : SUCCESS in 11m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14e194d1314547c094625db2c1013731 : SUCCESS in 13m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/672d528985bb4b07b877a59fca46a68a : SUCCESS in 14m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a90a1be8d98b4032b73e44a0d0ba09bc : SUCCESS in 11m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9f5ed6840b6b4ac6bba9a91bea5afd6e : SUCCESS in 11m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a684d2ed1fc248cd92030a7475dacd57 : SUCCESS in 12m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0610808ab5ea443f863b71dfbcf5bc2c : SUCCESS in 13m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ac7256fd47494f75bf2381181d42fcf3 : SUCCESS in 32m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23305b3787434e71b271acb2ef195cd6 : SUCCESS in 34m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/896887d8d0af46ad98a3998621ecaf0d : SUCCESS in 27m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb2001e673c34706b682aee8dc7e6999 : SUCCESS in 25m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c4aa393c9de40e3bf57afec45df918d : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb8be96a438459f84f34e07dbb14b10 : SUCCESS in 6m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b597c098bbda4d30a0fda1efcbac9d89 : SUCCESS in 48m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f55a4b71a4946b88f89f6241d3c24d2 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6a914bd50adc36f9cded12cd31f8a6afaee7f875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 20:45:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"afb20ff34e648531b7a82e26bc6a5e8507cafa86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 20:50:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f617467c4c46c38ebaee35473c02cb : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d90c3b19f3ea460eb78b9ac50083fa40 : SUCCESS in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40b88bf5385bd53640d95939afaa6e0ba15bdd0d","revisions":{"40b88bf5385bd53640d95939afaa6e0ba15bdd0d":{"kind":"REWORK","_number":1,"created":"2020-09-09 00:21:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/750536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/750536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/750536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/750536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/750536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/750536/1"}}},"commit":{"parents":[{"commit":"eef87ee21229c8f8cc794b29e0f9abc159fc98e7","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef87ee21229c8f8cc794b29e0f9abc159fc98e7"}]}],"author":{"name":"arzhna","email":"arzhna@gmail.com","date":"2020-07-22 00:49:17.000000000","tz":540},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-09-09 00:21:04.000000000","tz":0},"subject":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility","message":"s3api: Allow lower-cased region name for AWS .NET SDK compatibility\n\nWhen I call the S3 API using the AWS .NET SDK, I get the following error.\n\nAn error occurred (AuthorizationHeaderMalformed) when calling the\nListBuckets operation: The authorization header is malformed;\nthe region \u0027regionone\u0027 is wrong; expecting \u0027RegionOne\u0027\n\nThe reason is that the AWS .NET SDK generates a signature by changing\nthe region name to lowercase. (AWS region names are all lowercase.)\nThe default region name of OpenStack is RegionOne, and custom region\nnames with capital letters can also be used.\nIf you set the location of the S3 API to a region name containing\nuppercase letters, the AWS .NET SDK cannot be used.\n\nThere are two ways to solve this problem.\n\n1. Force the location item of S3 API middleware setting to be set\n   to lower case.\n2. If the request contains credentail parameters that contain the\n   lowercase region name, the region name of string_to_sign is\n   modified to lowercase to generate a valid signature.\n\nI think the second way is to make it more compatible.\n\nCloses-Bug: #1888444\nChange-Id: Ifb58b854b93725ed2a1e3bbd87f447f2ab40ea91\n(cherry picked from commit 3a6e85d9ec9a899c62400944832194b3c3a6a737)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40b88bf5385bd53640d95939afaa6e0ba15bdd0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40b88bf5385bd53640d95939afaa6e0ba15bdd0d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
