)]}'
{"id":"openstack%2Fswift~750537","triplet_id":"openstack%2Fswift~master~Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007","project":"openstack/swift","branch":"master","topic":"release-2.26.0","hashtags":[],"change_id":"Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007","subject":"Authors/ChangeLog for 2.26.0","status":"MERGED","created":"2020-09-09 00:38:22.000000000","updated":"2020-09-18 04:35:59.000000000","submitted":"2020-09-18 04:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750537-1600403667391-0e5654ba","meta_rev_id":"bf130df8299265af6fd2b6eac26af4c147c58416","_number":750537,"virtual_id_number":750537,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 04:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-17 18:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-17 18:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 17:59:36.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-09-18 04:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5847953f7e9769bdfb4723adbc0d498ee8004441","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 00:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcb8ecbc566a576d2bcce46bb4c2878528ea2cb3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 00:38:51.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"dd992bd18f6996891caa8377376d0bc575635c33","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 00:39:09.000000000","message":"Patch Set 1: Workflow-1\n\n(Still needs the reno-ized notes.)","accounts_in_message":[],"_revision_number":1},{"id":"5a78831e7afc6d49c7368aedd7486e0380009ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d037dee99bf49c3bb3576a50bf35ab4 : SUCCESS in 9m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79af9a3efd3647208045a9e8a9ed2ef5 : SUCCESS in 8m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e7f76534b32a4f86b1e0d14530a2070c : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e67b99e9ed64af2ad178a11cb2a300b : SUCCESS in 26m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2702cecc889f4582932fda094d1b68c9 : SUCCESS in 11m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/95f52fe6887a4983a184ace0bcf0c5d0 : SUCCESS in 11m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e78481f912574e8c8bd57f51a2b8800b : SUCCESS in 11m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db15367bc51347e18f478053b3771367 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d29985faf0a465e84f0533a96287aca : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"658a6ef1b1162ce24c30fd11736ab50e9d9d4eb5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:46:12.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5973cb5ca3a3465f82a9a24a6b710be0 : SUCCESS in 15m 26s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/133476eb57bf438dbf475fa325ab92a2 : SUCCESS in 23m 04s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/d61e65f024a5400f94d16e8a5513315c : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b85032a11d8a46bd9f244c561b9c5a58 : SUCCESS in 22m 47s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/acd1eac24801437ba9e591bd628789ef : TIMED_OUT in 1h 01m 02s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/a28b347e547e4c9dba6c25b76ee5c110 : TIMED_OUT in 1h 00m 57s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/27c28758b30d4e95ad2d915a7a9dfe87 : TIMED_OUT in 1h 00m 57s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/8aea67c25d2a4579ab386792beece693 : TIMED_OUT in 1h 01m 11s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/c8f372bc2ead440d8c677325fcb0a931 : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":1},{"id":"207b6ed0525048a23f07b11494276f89c794f378","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-09 19:58:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nWow!  What a release!  great write up","accounts_in_message":[],"_revision_number":1},{"id":"ce117016c6e7bc882387f2f6f3c6f68589ca3b58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-10 23:10:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f88915ef62a851a0936f6c2716a7df252801b7ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-10 23:15:26.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3f937d6dbc9f4c071bc6d1ff006b0c959aca5d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:27:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95b828125a540bfb5a971a6203c3f2c : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59e250329b8543e59598dbc879b7f53b : SUCCESS in 4m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/540bc719775f421fa7ba56aaa840675e : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6f59c31a8fc40e6b7962e3274f56da9 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1c35d80eb254deaab7988611dbc4a97 : SUCCESS in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb147b860bf4911b6223dc124ad4b74 : SUCCESS in 10m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afaf8716715b49d4afabae0b9e2cb17c : SUCCESS in 9m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c74babc41b494c228e7f6fd8b4eb5167 : SUCCESS in 10m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec2effc8640a49abaf9f8eee7c89cec2 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5157dfe621947f1bf5b8c28f5e820dc : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"cce0491a399164cf9fa3c52135dd49d4da6d7755","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 05:23:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"10a3f77e4e614ecc716ace4997e963980900f75e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 17:22:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48f8b146bbad7e22293853338aad87d41e4aa6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:58:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05e2895a07b40eaa85ce2724477f413 : SUCCESS in 17m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39e8beee78f54abb8a79dd67390a9a9b : SUCCESS in 5m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf7bff4feb8c4f02aaaec5930d9903dc : SUCCESS in 11m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d76291d120cc42b8a0c81308bb35a766 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12ddccb5402a46bcbbfa18730f38f1fe : SUCCESS in 12m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aacec2973f1f430a8778730d1334dc26 : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/600485b56c84479ebe107414e5e5c8ac : SUCCESS in 12m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c1129ed619a749f483e684a2d76b6310 : SUCCESS in 13m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91799fe0fa949f38d439743df475bc2 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b508918b85474baea74c6b4e8af1af80 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6946d5f18f37f5a2555ddaaa3c174d4ea3d11877","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-11 17:59:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nwow, fabulous release notes Tim!  if I had a cluster that wasn\u0027t already running master I\u0027d be so excited when these drop...","accounts_in_message":[],"_revision_number":3},{"id":"4006192dd7d70f29c2fdb3145b8469387d4f66a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-12 05:20:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"291498c1f2230debe5bcd7095daa5dd8a4f51eea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-16 00:07:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8cd30b3c56f0995353c4e4a43272ccb58ad562b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:30:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e12b508cf54e2880fa4fbeff849d59 : SUCCESS in 15m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd835c52c7040a6befae408fee25bec : SUCCESS in 6m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fba89cd4ec949e8a9128a24aedf7554 : SUCCESS in 10m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bc8088112b348b097fb05acb87b269f : SUCCESS in 11m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be3fd151886f4ec8b99a945095bd573d : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bf24d079f02460494f7e27fd7da00cd : SUCCESS in 21m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/01705215f9bd4fa184eb9795635ee306 : SUCCESS in 15m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/456b77c1538549c9a0bf538d9eca8288 : SUCCESS in 11m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e92c0074fb3b49e981a4428470cc1abd : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f4499528e2464bb08ade2b370c161a : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"bffb05187d499fd599b95ce857a4dc19afae9026","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-17 18:56:34.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4dc317b8937576b4e5eb907c6fb9d0bee18b493d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-17 18:56:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"420732f656ad003d84daba2fa59df365bbc57d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:24:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c954d2de355a4853bd66e3f5761dad35 : SUCCESS in 18m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd4e2838afd5404daef731c36d6af4fb : SUCCESS in 5m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e8aab1c067f432dafb0ad65e73510a1 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d81c9489326c4018a3816cd1c40b729e : SUCCESS in 10m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/136e87261ae846509b9c6f4c1da77d78 : SUCCESS in 18m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5967eec16f0445fa909dfd8f2dcd4f1 : SUCCESS in 15m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5c620c3264d5487699d55ad2b819bd34 : SUCCESS in 16m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19080f6fd81d4aa1b41748eb06dc18be : SUCCESS in 18m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da6f40638ce443be88a99413962dc5e6 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cfe99b37804393b9f6bcd131476bf9 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":5},{"id":"e58d157096de6d6120d87e55ffda540805ec4d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:25:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2795180738001d655acecdf9960b0ad0245a8b9a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-17 20:25:02.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"4cef1909c97c976d679b127940d9aef0b1d606ae","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:02:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/af9cc89942cc4eab8b50e3b57f1a1e78 : SUCCESS in 15m 13s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b7fb61d89d0042eeb60a1710635b478d : SUCCESS in 24m 50s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b1748dea03fd47efb9709a52dc8b814b : SUCCESS in 36m 48s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8df6b9ffe9e04c90b9df4a2168165176 : SUCCESS in 22m 00s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/7b7a00772814429c99e5622614650ef4 : SUCCESS in 22m 42s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/a0793e27bf3e4d129fd9d963b2688128 : SUCCESS in 17m 06s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/61092fda4fea4037a1c35967efa0efde : SUCCESS in 17m 31s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/08f9c1973905472a9687fb52dc8a4593 : SUCCESS in 21m 25s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/8867f6b70a5e499a9fbd81db1d13df1a : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":5},{"id":"92c021e2f145ef682d8a4520c7295bfa4863b10a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:17:46.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41956dc0ebb24c99aa69a04cb60d1548 : SUCCESS in 16m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96099bd2e1e44c0f88f1464100f868ff : SUCCESS in 6m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a50105c8b62a4837b362a06c89c40493 : SUCCESS in 10m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/01ee5b2328d7457f972aa9edeaa6aed2 : SUCCESS in 11m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6968d24812ab49d9b65c1e89bef989ca : SUCCESS in 11m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b20f2b964048474191d841165549a78f : SUCCESS in 11m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c08217391a5a4ead99d22935f9053871 : SUCCESS in 11m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ef87155c0dc34d969e73b9d30d34c773 : SUCCESS in 41m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7508adfa67ad4bbab747aae6d40392a4 : TIMED_OUT in 1h 01m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b384bd7a96fe4599bca6132292a7a2f5 : SUCCESS in 58m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8eb184b5686941888f971e0b123daf3c : SUCCESS in 37m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0fb7c147d6554ae4a7127073fa106a29 : SUCCESS in 21m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e63ebc1e66374d7d861a80ec53dd752a : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b839bf8dc30c4fc284c9ccc1c53c3623 : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7baddf914b2417499ebf4e8dde0d334 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"d30cb1210660f53b80e31408b8788597e66247ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-17 21:20:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5e9734887730e1ba0eca01249470ea1520aeb42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:41:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae81f17cbe88458a9d1bdb5ab385fc21 : SUCCESS in 15m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/384ff17af07d44f19cf9e0862f88d966 : SUCCESS in 5m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d31262a0bba449dacee611dc28badd8 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20112281b48b470d9142f46d8ab538aa : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/221a9f3f0a704503b9c51a237a51066e : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/305ebd36e379452a8342c860217f0429 : SUCCESS in 12m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35354fc132d445c99bf8f82134442fbd : SUCCESS in 12m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d5cb9106d64d07ac9dad20ce4c4ec0 : SUCCESS in 11m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e7ec08c2cc54f738f31a00d0b978182 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d905568b749404f99c18f5b976b8187 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":5},{"id":"f4a61ffe2b2f7d35416cb92104da9b605178ca4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:42:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"16d90aad860a161400d4cdb8706787230144d9fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:21:47.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421713ffbe2c4b4b9a9f4978cdc89840 : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae31cd83f7e42308f97964b05475564 : SUCCESS in 6m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a222464ada914076964e4b02da5618ca : SUCCESS in 10m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/83374f077e014bbbb88a88c1fba70213 : SUCCESS in 11m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/beabfc8de0bc46c1b85698659c7e6fbe : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4af12459440549b9a854734ace06526d : SUCCESS in 15m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a4ed8188b9a7441ebdf9ead058116859 : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8fc8b7ab244466a9ce85e4602868077 : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47a533c984294c99bf48addee78c4a76 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b97b38a6c2df487c87fecaf893a321bc : SUCCESS in 19m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44638dc2d0334ecf95d1162bc1c0445a : SUCCESS in 21m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fba48d221c4e4a5fbebd2fe6e37e0efd : FAILURE in 21m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/21d5c8606fba46c98b0b902d9afd9582 : SUCCESS in 20m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4675ae562dbb48f98be5642e35a63bb0 : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4adaa082004c43849322f4ad06e049 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":5},{"id":"5fc4aa369dd8f2aad771a9b10dcfc233c5af6428","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 03:48:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8d861fc5b9ac499c2032611c75443547f49055ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:10:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08b42a9b3c34045a57a60e6b6e40379 : SUCCESS in 16m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cb24db712744719bee0bbb8c01b5187 : SUCCESS in 4m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25cec3703dda4715bad7c9b9d8799e99 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f83595ad69b742f8a2e8ddf7880d5be2 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b63105c5caf14c4abc89bd6f7a16c2e5 : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb3d21983af42b1a2ae0bb2ed3bb8f8 : SUCCESS in 11m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/38856944665f4c5f9cebaae0592051d1 : SUCCESS in 12m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9a69bb181bf349bfbd479e84f7a84a81 : SUCCESS in 12m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edb04196d3cd49d9a7ee79033f68fb6b : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90d9cb4767d4d55ae75352d26891c49 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":5},{"id":"23f040d882978b811cf66d008cb7881073a81ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:10:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c2147a5030bc128eddc1bfbca1b9ebc941a35191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:34:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cab987c48542f3921d2c309d213e6d : SUCCESS in 18m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc63f8c30a0b465bbde1db4de6522d69 : SUCCESS in 5m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d6c5600c81494d80bd0d8dd7d4d110aa : SUCCESS in 8m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9dc7cc6a12a04b1bbc7c2260cca8aa2f : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df1de3ae77894c0a917883a1fc24ad3a : SUCCESS in 15m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/68648b179b5c44269dcbd92ca7f47570 : SUCCESS in 14m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d8e6b4365a04079be519ea23d534718 : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/06df0cb611ec406d80cba75d75312ac3 : SUCCESS in 19m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec466184a3d04668bea3c4359334252c : SUCCESS in 19m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fb0edbc4c2c4cfdab72d6d500695dcc : SUCCESS in 15m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8032ce87d2a4e0a973273d1f2e16c37 : SUCCESS in 23m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab82f855659843c6b5a41e57698f6bb1 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/beba2c71ec8f46aa8f3db5b0b2c5e8ab : SUCCESS in 18m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dad5290288774c6c946eed46b74e16a4 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/087888910a1b4aa39d05c2e7447282c3 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"9e3e7237b1e4c3dfef2e6a2a6c8534c4e2264a5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:34:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6a6d01dabb28bfeca586b6f45dced7af51ed1e76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:35:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ccbf2d3e004d3ea4b9fe1cc1cde3ac : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb424fa23be4a70ab01adf21fc9bb30 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/63053f84262f475b85358d36895acaeb : FAILURE in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b24a9eecced5473dbec11c506897aca0 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2","revisions":{"413751f06c210f2bdb7b204ebf4fe0adc2d1485a":{"kind":"REWORK","_number":1,"created":"2020-09-09 00:38:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/750537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/750537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/750537/1"}}},"commit":{"parents":[{"commit":"06091172c218929bcee9c66c50e846acd8326a32","subject":"Merge \"ec: Close down some unused responses more quickly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06091172c218929bcee9c66c50e846acd8326a32"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.26.0","message":"Authors/ChangeLog for 2.26.0\n\nChange-Id: Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/413751f06c210f2bdb7b204ebf4fe0adc2d1485a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/413751f06c210f2bdb7b204ebf4fe0adc2d1485a"}]},"branch":"refs/heads/master"},"a9f0b827ee83d066ac715655311c18f30f8a1b19":{"kind":"REWORK","_number":2,"created":"2020-09-10 23:10:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/750537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/750537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/750537/2"}}},"commit":{"parents":[{"commit":"4fefdf4d83bc43f86014906ff46a936168f71a69","subject":"Merge \"gate: Make rolling upgrade job work with either 60xx or 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fefdf4d83bc43f86014906ff46a936168f71a69"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-10 23:09:53.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.26.0","message":"Authors/ChangeLog for 2.26.0\n\nChange-Id: Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9f0b827ee83d066ac715655311c18f30f8a1b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9f0b827ee83d066ac715655311c18f30f8a1b19"}]},"branch":"refs/heads/master"},"aa298b839327b927591cdee5299bfeb2ffdc45cc":{"kind":"REWORK","_number":3,"created":"2020-09-11 17:22:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/750537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/750537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/750537/3"}}},"commit":{"parents":[{"commit":"4fefdf4d83bc43f86014906ff46a936168f71a69","subject":"Merge \"gate: Make rolling upgrade job work with either 60xx or 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fefdf4d83bc43f86014906ff46a936168f71a69"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 17:14:52.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.26.0","message":"Authors/ChangeLog for 2.26.0\n\nChange-Id: Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa298b839327b927591cdee5299bfeb2ffdc45cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa298b839327b927591cdee5299bfeb2ffdc45cc"}]},"branch":"refs/heads/master"},"224963a5f01c1c8a7025c6429e8ea70b44b5be5c":{"kind":"REWORK","_number":4,"created":"2020-09-16 00:07:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/750537/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/750537/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/750537/4"}}},"commit":{"parents":[{"commit":"f5e03ee973fa12088306809bd00ba26ec1d0ab1b","subject":"Merge \"xlo: Drain error responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5e03ee973fa12088306809bd00ba26ec1d0ab1b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-16 00:07:14.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.26.0","message":"Authors/ChangeLog for 2.26.0\n\nChange-Id: Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/224963a5f01c1c8a7025c6429e8ea70b44b5be5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/224963a5f01c1c8a7025c6429e8ea70b44b5be5c"}]},"branch":"refs/heads/master"},"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-17 18:56:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/37/750537/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/750537/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/750537/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/750537/5"}}},"commit":{"parents":[{"commit":"4b2d8c2258882c9932ff52bd41db00f036cc3525","subject":"Merge \"proxy: Include thread_locals when spawning _fragment_GET_request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b2d8c2258882c9932ff52bd41db00f036cc3525"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 00:37:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-17 18:56:22.000000000","tz":-420},"subject":"Authors/ChangeLog for 2.26.0","message":"Authors/ChangeLog for 2.26.0\n\nChange-Id: Ia8e31ed0d5aefe67f2f926dc92d9acd6c0c98007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
