)]}'
{"id":"openstack%2Fswift~750679","triplet_id":"openstack%2Fswift~master~If14f53993e6a0554345b327415812423b6b57434","project":"openstack/swift","branch":"master","topic":"rolling-upgrade-ports","hashtags":[],"change_id":"If14f53993e6a0554345b327415812423b6b57434","subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","status":"MERGED","created":"2020-09-09 14:58:00.000000000","updated":"2020-09-10 18:57:02.000000000","submitted":"2020-09-10 18:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750679-1599763969659-f829c1a7","meta_rev_id":"6a2dc9094cbc21caf5e80060fbd7c49043518b8e","_number":750679,"virtual_id_number":750679,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 18:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-09-09 23:15:54.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-10 16:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 23:15:54.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-09-10 18:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3af464ae6cea38b394a375dfe831aec3daace688","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 14:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bfd17d3b2c383d506be1fcf45fca6be9cf82922","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 15:11:34.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"7a5f353646ff9fdcd476a65eb96362b553179640","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 15:37:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f4cc823653f7ba36564d5813c8c52ee79983b0d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 15:39:40.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"fff71d745f7e77e5eda2b80deaeb2e80b3772cb4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 16:05:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d4658e6c96f99c09154ab5327e05077ec857e56c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 16:05:39.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"0ff4fb6d51dfc9ea13250998b5af1c594c1269d0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:27:17.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/bc6b736fa1a14d86872ccf7131f42fda : SUCCESS in 16m 04s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c6fd60e304c347a59661012aa12b19ed : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/67087597fe52433481ac8a9b5c94240e : SUCCESS in 22m 26s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/b5c09f2798a945eeb0da1d5f0eae6499 : SUCCESS in 19m 09s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/eddd2427443f4dc1a2c31e7fa9f163fa : TIMED_OUT in 1h 01m 00s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/20b350362ba64261aec080364d1849ee : TIMED_OUT in 1h 01m 11s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/01df41a5207e45c08637edb0409b700a : TIMED_OUT in 1h 01m 17s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/dd523658d9e94b419e175490fa3642cc : TIMED_OUT in 1h 01m 22s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/e92412339b3a49288a17d3c53c6c04b4 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0fa2cd5f4908db6d5e05d20c350fac6629d328d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:09:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578c1cf78a99460d87fbd6f15fc6fa10 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/94fbe3e18a164a3385d2f48056bc17fe : SUCCESS in 1h 00m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c033ac8d8f1e4c7392ccfce67f5dceb9 : SUCCESS in 43m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72ab7ab504154d08974415e8e293814b : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a59aaf17edfe4904bb133b502a88c919 : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e07af4eebc394ee081139ff4c952d8a2 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2227cafcf6234312869607aa0074d77f : SUCCESS in 14m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4eadb66012a648b0980d68b9e5f1ac0a : SUCCESS in 17m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/71b6c1523464473b8e5e5ccfa16162ed : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0114802cf624ac7af4cbc03f889875f : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fab8e631ca54da883c9ed6864d28c28 : SUCCESS in 21m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c54ed72026c466d9093f8a445c68cce : SUCCESS in 19m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cbdfd507ddf7487cab01242d4b0d65b6 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fd62ea53383747c6bf1b31adc3bd8d2d : SUCCESS in 17m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28cdaca8ef4b4392b4b422b463260b8f : SUCCESS in 16m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae1b4cb41c53406bb3dd2fea9ef630ba : SUCCESS in 20m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d47a81c20a994aa78425a07cca148e93 : SUCCESS in 46m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/374098d57ee64e45abe4f12a489cd6c0 : SUCCESS in 42m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffc0eb798cac40ef89f5b8894beea4bd : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02fcfb1c3ccc4b36a34ceea0be22f0c3 : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e79bb31716a4236b9d6597221180c3b : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84487c7339a94d21bd9b912070e5a4d5 : FAILURE in 9m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da5ebe9dd9d14a378a48e0b8261ebfdf : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":3},{"id":"bf7a7f8438f87af77744ff0f66de0db38e7ad6d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 20:38:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba71e30778e2ea5422d9818bd1bc42592d23fbdd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 20:38:49.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"1da5a00dc0a9d0d7f10aa4313cf95d6a4df9fb73","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 22:47:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8ccf2b2ec51dacc7f806b5c752a37e939d5b3e43","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 22:47:35.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"8ee20733256c21f876408c0d773cd85d555e29fe","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-09-09 23:00:05.000000000","message":"Patch Set 5:\n\nThere\u0027s something I do not understand about this. WHy is it that the old port cannot be extracted from somewhere like object-server.conf?","accounts_in_message":[],"_revision_number":5},{"id":"e29e4f6b84e52f4880adeb5ba7f059f2219932be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-09 23:05:46.000000000","message":"Patch Set 5:\n\n*shrug* Yeah, we could also go update the bind_port via crudini like we do for bind_ip... I\u0027m open to suggestions. In part, I did it this way just to demonstrate how a rolling upgrade would likely handle it (i.e., leave existing config alone and make sure your rings reflect reality).","accounts_in_message":[],"_revision_number":5},{"id":"39702d35c9f83349b263f90e0a6c6b51c3dc6958","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-09-09 23:15:54.000000000","message":"Patch Set 5: Code-Review+2\n\nOkay, if this works for you, I don\u0027t mind.","accounts_in_message":[],"_revision_number":5},{"id":"d84dffe2d7d55778d1f221a82ea99ffa21987558","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:37:02.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/147378599d554bc5bf68e56375a49b7c : SUCCESS in 24m 23s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8d61c7df18c54158afe2c002202dedba : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/57569b5099c5434fa14e017350b95db8 : SUCCESS in 35m 08s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/af8db6ad159140d9a1b5e266aa96e9e9 : SUCCESS in 23m 07s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/515b34f9b03244668786a9fbf5982f0f : SUCCESS in 16m 03s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/0aa91ce0dd2b4d48b4651005b5176a9a : FAILURE in 14m 22s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/ff31fa1fd7b842dea9eec138e403be7a : SUCCESS in 18m 09s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/8fb231b200fa4d63baa4038e19cf6078 : SUCCESS in 23m 47s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a1ec94d367104b24a2bb838d490a7b2c : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":5},{"id":"53ff0d69fc9bec4633bb6ee6ba8b548f0db587be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:14:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d097694b607346dfb846354dae9c7b6d : SUCCESS in 18m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/f40cfa44fadc43f89516851908b25453 : SUCCESS in 51m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f64d31c7dd2419d9141cd5c75eeb26a : SUCCESS in 42m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd83eaf92b2741dfae279232f3065b3d : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97694ff604f94033bd2fe9d7b438d3c5 : SUCCESS in 12m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f212cb0661294a0c882b2c9348ba3569 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77d0fb6f81114582a7823d27a85639e4 : SUCCESS in 19m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f093bfd76e84e2c84d5b5920b4f4273 : SUCCESS in 23m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a968df0317154127857cef628261b582 : SUCCESS in 18m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dcf525c4943345fdb3410a511485c001 : SUCCESS in 23m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/23b8f0c5456445458a95fcea5f48cf3b : SUCCESS in 20m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0b53ef6da1b4ca6bc8f76f0f17a24c2 : SUCCESS in 21m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fa25802faf514b238d87cbb4593be4f9 : SUCCESS in 23m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d435fe1b46514ede98cbe249b459dfb1 : SUCCESS in 20m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7d967c3c2ab04cc8a3c173cbc1de8e77 : SUCCESS in 23m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4bcaf544052474babf9302d25fcfc5a : SUCCESS in 35m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8355085ae774930a994cb38d47e5136 : SUCCESS in 47m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a307375485344b6aa33e14dbd2a843b : SUCCESS in 48m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45fa40eb163048ae961277885271e086 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fdbe78b66744ea9b63f45c1c3590f47 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71b65f0017a4bb8b25447b34212fb63 : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4493f242ed054a28a8c130773e48d605 : FAILURE in 9m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e7ba8f1dff04d9da86fe47ce4528308 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":5},{"id":"43d4cfa95e1f42f34371dc4826eafab353cb7179","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-10 16:50:34.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8eabdc8a52d3712845772174d10b932c2a70490e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:53:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7ec80e607736fcd0d4639e4b1687492c2862823e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:52:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803e057e8e18499da2bf6b1675a89cad : SUCCESS in 16m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/586db6e9a1e14ee78a8cedcbf66c73fd : SUCCESS in 1h 10m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a1e8c05fa6547faae8ec567346ea70b : SUCCESS in 49m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1f0155cd1449416eadf163c06efa37f2 : SUCCESS in 11m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d9df96c63c9b40abbc0f10d0ba34bf44 : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e66c5917073948b78cd6d6dc1ad2b462 : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9ba26cf55ea434d850bf7e0dc79b83c : SUCCESS in 17m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c8a5c6f745fa498788a0fc2fa3366836 : SUCCESS in 15m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e13466c70a0b418cafb2b7847a8b9c27 : SUCCESS in 12m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53916133015c4f36bb351592eed62843 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/727d1bbc4d0c41f6af4526ff115498d7 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ebca25144ee467a950cbef21fed2489 : SUCCESS in 18m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ce59ab9e9e08475a971be404d156d8d7 : SUCCESS in 20m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d7810b802b747cbab4a17a73cab55b6 : SUCCESS in 19m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6cb9b50be2af436b9d1f84da30e78bd7 : SUCCESS in 40m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c5a244239d342e8a101cc6fd0bf24ad : SUCCESS in 41m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16983d08936d49ea9845d59f3622a544 : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89ea5e0708534b65a1233244135d5688 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03801074f2154449a47e526a4aaf248b : SUCCESS in 6m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cc5d801768c40a28af6e61e40db6337 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":5},{"id":"a1cd2f1593b5d60e306b95fa078dfce36310f990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:52:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"661c3626c16f8af69d0b7faf92b74bbf3dbfa239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:57:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36bde3f354034e78a55ebd404b7b1e97 : SUCCESS in 1m 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c178f5c26975498da919c6320e0d91d4 : SUCCESS in 1m 08s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/95a1de44646b4a77bdbc2c70c3eaf290 : SUCCESS in 1m 03s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ae8a89f042f691528f7522c2979b0bc4b45ff9b3","revisions":{"5860d2ecdb9e6f63a4a93be8056c4498bf8274df":{"kind":"REWORK","_number":1,"created":"2020-09-09 14:58:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/750679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/750679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/750679/1"}}},"commit":{"parents":[{"commit":"754defc39c0ffd7d68c9913d4da1e38c503bf914","subject":"Client should retry when there\u0027s just one 404 and a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754defc39c0ffd7d68c9913d4da1e38c503bf914"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:56:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:57:41.000000000","tz":-420},"subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","message":"gate: Make rolling upgrade job work with either 60xx or 62xx ports\n\nChange-Id: If14f53993e6a0554345b327415812423b6b57434\nRelated-Change: If5dd79f926fa51a58b3a732b212b484a7e9f00db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5860d2ecdb9e6f63a4a93be8056c4498bf8274df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5860d2ecdb9e6f63a4a93be8056c4498bf8274df"}]},"branch":"refs/heads/master"},"cb48f8315f76b4141128ac57e74912297c0d5f3f":{"kind":"REWORK","_number":2,"created":"2020-09-09 15:37:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/750679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/750679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/750679/2"}}},"commit":{"parents":[{"commit":"754defc39c0ffd7d68c9913d4da1e38c503bf914","subject":"Client should retry when there\u0027s just one 404 and a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754defc39c0ffd7d68c9913d4da1e38c503bf914"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:56:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 15:36:11.000000000","tz":-420},"subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","message":"gate: Make rolling upgrade job work with either 60xx or 62xx ports\n\nChange-Id: If14f53993e6a0554345b327415812423b6b57434\nRelated-Change: If5dd79f926fa51a58b3a732b212b484a7e9f00db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb48f8315f76b4141128ac57e74912297c0d5f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb48f8315f76b4141128ac57e74912297c0d5f3f"}]},"branch":"refs/heads/master"},"fdc8be28a6700487e504d49ac1fa6dfcd0635f19":{"kind":"REWORK","_number":3,"created":"2020-09-09 16:05:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/750679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/750679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/750679/3"}}},"commit":{"parents":[{"commit":"754defc39c0ffd7d68c9913d4da1e38c503bf914","subject":"Client should retry when there\u0027s just one 404 and a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754defc39c0ffd7d68c9913d4da1e38c503bf914"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:56:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 16:04:59.000000000","tz":-420},"subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","message":"gate: Make rolling upgrade job work with either 60xx or 62xx ports\n\nChange-Id: If14f53993e6a0554345b327415812423b6b57434\nRelated-Change: If5dd79f926fa51a58b3a732b212b484a7e9f00db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc8be28a6700487e504d49ac1fa6dfcd0635f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdc8be28a6700487e504d49ac1fa6dfcd0635f19"}]},"branch":"refs/heads/master"},"47f66d469777d9e53f045be7d91700fd15f3dc0c":{"kind":"REWORK","_number":4,"created":"2020-09-09 20:38:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/750679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/750679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/750679/4"}}},"commit":{"parents":[{"commit":"754defc39c0ffd7d68c9913d4da1e38c503bf914","subject":"Client should retry when there\u0027s just one 404 and a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754defc39c0ffd7d68c9913d4da1e38c503bf914"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:56:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 20:37:34.000000000","tz":-420},"subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","message":"gate: Make rolling upgrade job work with either 60xx or 62xx ports\n\nChange-Id: If14f53993e6a0554345b327415812423b6b57434\nRelated-Change: If5dd79f926fa51a58b3a732b212b484a7e9f00db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47f66d469777d9e53f045be7d91700fd15f3dc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47f66d469777d9e53f045be7d91700fd15f3dc0c"}]},"branch":"refs/heads/master"},"ae8a89f042f691528f7522c2979b0bc4b45ff9b3":{"kind":"REWORK","_number":5,"created":"2020-09-09 22:47:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/750679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/750679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/750679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/750679/5"}}},"commit":{"parents":[{"commit":"754defc39c0ffd7d68c9913d4da1e38c503bf914","subject":"Client should retry when there\u0027s just one 404 and a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/754defc39c0ffd7d68c9913d4da1e38c503bf914"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 14:56:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-09 22:47:00.000000000","tz":-420},"subject":"gate: Make rolling upgrade job work with either 60xx or 62xx ports","message":"gate: Make rolling upgrade job work with either 60xx or 62xx ports\n\nChange-Id: If14f53993e6a0554345b327415812423b6b57434\nRelated-Change: If5dd79f926fa51a58b3a732b212b484a7e9f00db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae8a89f042f691528f7522c2979b0bc4b45ff9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae8a89f042f691528f7522c2979b0bc4b45ff9b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
