)]}'
{"id":"openstack%2Fswift~751384","triplet_id":"openstack%2Fswift~master~I2d22e087a88c9e8003621feb26460ab6e5ce2a57","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I2d22e087a88c9e8003621feb26460ab6e5ce2a57","subject":"py3: Fix swift-dispersion-populate","status":"MERGED","created":"2020-09-11 18:46:23.000000000","updated":"2020-09-28 20:42:50.000000000","submitted":"2020-09-19 01:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751384-1600479750592-309c905f","meta_rev_id":"7976d2a17ea9220839191cac610f667c9373f8d1","_number":751384,"virtual_id_number":751384,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-19 01:42:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-16 15:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-16 15:44:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-19 01:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fcdc4efb9fed12c5699b4e6f51444d653affaa4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 18:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24d726a112af294904f299133a6e40e9fb1e2529","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-11 19:35:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b49ab650849089994acccbb8eec29ff38a3c102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:06:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fb5e8d72994373897b409fab5e2276 : SUCCESS in 16m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/028ba7909c834571a4f391306c4aa10a : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fc8a48c33344769935f07ddb85c3279 : SUCCESS in 47m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9cad94793e244b958cebfea0a06754f7 : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d52492b38aee48a586febb4976266206 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7860b3bc98b24c63b3c0d342b6394649 : SUCCESS in 15m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25e8a28e1f754898b66fd61de4233f41 : SUCCESS in 14m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5016ce2ff924883abac743203259c6c : SUCCESS in 15m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/722b986a40704a7f8481ac8ac9fb4f3c : SUCCESS in 20m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15c46b48412a4bc1a4f7ecd3b3d37120 : SUCCESS in 25m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c67a08a03ac24b38b8eab75245e57c9e : SUCCESS in 21m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce0ea1c4993348c1b8c2191a036ef9f5 : SUCCESS in 19m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d1e66bf5de14f85a7a5488bf2fc78ed : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bbd4b92ffdb14e09b326aec897d333e6 : SUCCESS in 22m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a93b04f8203948a3a81e3b6d91503210 : SUCCESS in 28m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/222a7cfa2b9b442884a63b90ed733bd9 : SUCCESS in 33m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4dad21b9013b4ca4904532da3583516e : SUCCESS in 55m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd992d3bc008468dbdc5abea86d5f410 : SUCCESS in 40m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3ff8902212c4ac79332adc25b6f27fa : SUCCESS in 51m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1283b645aaa433bb3f2e266392ed191 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab8d88bf05346dc9fc0e9d6ec736bce : SUCCESS in 7m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dead9453611a4b43a8737af69ddae9ae : FAILURE in 13m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5309cd5791084df580c3c36e405b8c93 : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"04a185bdd3bf2518329aca179451af3eefdb01f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-16 15:44:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMirrors the change in swift-dispersion-report, and my implementation matched exactly David\u0027s suggestion in the bug -- I\u0027m calling it good.","accounts_in_message":[],"_revision_number":2},{"id":"ab85c3738213689c8dcb2d61a9d255ca0edac7dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:44:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af186b5a1d38204c8f232061187df52bc93d64d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:55:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec1caac6cdf4d38b290ed3868e4310e : SUCCESS in 17m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d858429978e14a64928363459c69cb88 : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d725978f58914138b7004e06334ccb7f : SUCCESS in 50m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3361edff2ceb413b9144fc38c4789f50 : SUCCESS in 10m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/91333b6500194a56b0aca2d066dab45e : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69b19ef8c9f24c94ae461e92a60e3159 : SUCCESS in 14m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4943fabe85a476a84902ab0c5fb7e20 : SUCCESS in 14m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a0c4b783a7f4baab56cfdd65bc94538 : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff5ac32bc81a41d69851a6f933f4e922 : SUCCESS in 12m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9da4a58daacf4a238e8475a36bb876be : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e4392d7721e04910892337d407c34068 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/391d0ef292ce4e1a83f5b0ed8128da1c : SUCCESS in 21m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1f5b800e24354b47886db93e5458075a : SUCCESS in 19m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/89786cbc5fcd4c4fa23eef03b5b5fa0d : SUCCESS in 20m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/277194f52b684c2d8472214f570977b4 : FAILURE in 50m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bda15b6315bc42e49681e3e058bd95ec : SUCCESS in 46m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8f507ebef8a409eaff61ae041d2e1c1 : SUCCESS in 51m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39824988e9be4641b5c96d513133db51 : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3801ed6fbe474d9780e4811b7170dc21 : SUCCESS in 7m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15a1dcbaf13940be96663592f8c4ca52 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"28848738f979b7d71ec351c3d92c588477078bcd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-16 21:25:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"493033e8da3a8485358cf092d6dce5e35e8e9e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:40:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d849c22c8794bab831edf384fed4e25 : SUCCESS in 19m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/6dc91d4d641248e9aed3e398c3772afd : SUCCESS in 1h 07m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d44c49b333543ea9fc4209f2f1ab1ca : SUCCESS in 34m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/151601b6385340129ac7f752c9dac2b4 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0a1d6aa5e11453993525afe67d81b94 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4579762949c8464c9dbb02daef4478e1 : SUCCESS in 13m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ffe7839651f24c8096f761a83f7270f0 : SUCCESS in 51m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/95967192b2814096853306b88d298f88 : SUCCESS in 17m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be6d0a49d386444c96ce449cc960093a : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b13106a12de344b48c3b4e44d66fa762 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2f1e70e76194d94a4d5a980dbfad273 : SUCCESS in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17e72fbc6dcf4ae6918b3b9e5a767718 : SUCCESS in 21m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/40eff74a75b642eb91b7a6a2f2ee90f4 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00bb515ccff547be8ace663d44d74e3f : SUCCESS in 33m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28fba57f500646c0b68321622daccf6c : SUCCESS in 19m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11c887bb1cf74d09bcbd7e4084aeb2b4 : SUCCESS in 21m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52bb4bc634d94d6fb6f3ae115c51e68a : SUCCESS in 41m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9fae1586ac7d46588a276a46cfc0846f : SUCCESS in 40m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45d9e02eadba4d6482dc1dda3d9198f8 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f48b4ed93cfd4126b8d2be5828f1557a : SUCCESS in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60151b1fe4a4c0ab22de913d8087e19 : SUCCESS in 11m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df4be43331a34d9ea6ade4cbe3e05ba7 : FAILURE in 11m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8495924f4f8b465f8162787bc92b0b62 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e420dd9556f1733dcf094462fed2a0d7b31c20aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:41:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37aabd8ce5cded2d34e34f33bfd999288a0d232e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 02:45:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f70224a8e847899e845668de891c3d : SUCCESS in 15m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3d8ae04eda0a4bcfbed7f9a47f81c86a : SUCCESS in 1h 09m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f343d37dd8844e3c8773cb248b8c0f57 : SUCCESS in 44m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ba960f7fb25f4c17bd4e3def24e2f848 : SUCCESS in 8m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7195cbf5e4494bb8aac195fc064facf0 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/024e9462d47841f6afa25571136bea7f : SUCCESS in 13m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c959742ba4fc49d5be0cecf0b0256c63 : SUCCESS in 18m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/beefb1435f114a429e234bc3c76b0978 : SUCCESS in 12m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8a04e20c69d405fb172a5656f4b2f25 : SUCCESS in 25m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a506098968b48929eb6bd473713ff30 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e17d344f35b42e7a8beade983fb4a16 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a8e82bb08754a11b395e9a945bf6f08 : SUCCESS in 24m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d1b9cbe4372475a9fbcd1f565603695 : SUCCESS in 31m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5e0ca9ec23f44151a221da4ad33e0ae3 : SUCCESS in 16m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/42651bba080c4ba2accb5b33579898cd : FAILURE in 1h 05m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aab29233b9a844ad810dd7558456f0e0 : TIMED_OUT in 1h 03m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff3e97de09254e1da8744d422191c2f8 : SUCCESS in 38m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21387053f3ff491d9c4337372ec748ef : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b58f8a6262439d934ff89ec82d0d36 : SUCCESS in 5m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3ecf8ec84464a88898f2fbeffdaf39a : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ed71ad8a541c47486744770517f212019182c258","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-17 04:23:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4c69570f76181583c4cb7f9c55d3230f1bc73db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 05:28:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db76066e9184d14bd28c721a8373140 : SUCCESS in 17m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/5cf918d3f6b2422d858912ddad5594c7 : SUCCESS in 1h 04m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bff4c65fbb5947c2aa5059a3cdc9b537 : SUCCESS in 46m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51cdcbae19104a6fbf388a15d51635cf : SUCCESS in 7m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c7e1ed779f94f028eafa055dd978803 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb740f9fbf144c7cb5bcfcafff7117af : SUCCESS in 13m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47443fcaf5604608bc52da7b8c72e68f : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8ab4dc92b434c7eb83a406de3e302d5 : SUCCESS in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afb2298f00154c1a827f6437d4962df8 : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4260c9b6b1e24c86ae266f92bd6e1838 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/421e20b3930945a58f9637624fe0f8c9 : SUCCESS in 26m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d266e293f54143fd91ae81bf94bc175e : SUCCESS in 17m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1fefead6609845e494ec35fb8b2061ce : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2825db87d5974f5a9acbe6f0ef5942ca : SUCCESS in 18m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/43b9f80a1399401e9c9176479d5420ab : SUCCESS in 17m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a727eab60f4d4fce98ca583856fe6fec : SUCCESS in 33m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2af9e7c8fd6049f3bf93395958cdd54b : SUCCESS in 47m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2495059239b54985be6b05b14beedb4c : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44dc828305144c9abc5d5809f40d9dea : SUCCESS in 40m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5964e1123c854b1eb75262e3fdf213d9 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d523647b53f143869a5dee9009167dd6 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71ef3b5f90e24fdcbe85c3fc4a64bb87 : FAILURE in 7m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78e43321457f448a9f606c14339617f4 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":2},{"id":"36f25bee761c66ea38c3da482f803fce22e7e903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 05:28:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39341156f491fb0573e7577c8ec9ffad386d9631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:12:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e5d3cc52904c77a6c822889d7e5f6c : SUCCESS in 16m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/54538e84ce474a74864166faa70e3fa1 : SUCCESS in 1h 42m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e376d182570c4d34ae15d2380f0fcc3e : SUCCESS in 1h 23m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/22fdd855505e45f182a6701fee6fdf6d : SUCCESS in 13m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/75ae97a9b8b84e78a6728d54c3dcb7fb : SUCCESS in 14m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b29538c78d84f5bbc30fc73915abfe0 : SUCCESS in 51m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a94cb3253e74af69eb7e88db606a0a8 : SUCCESS in 50m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8c264e381e3b48dbace08a1fe532ad84 : SUCCESS in 13m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57342590082e4227b7446517d391d93d : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c0f5a6781df45db9eacaa7b26901124 : TIMED_OUT in 1h 01m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c28c6b5de41146af8bcb2c439624d69c : TIMED_OUT in 1h 01m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb23a1ad8eac41efa338ff490a9b549a : SUCCESS in 19m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8307b1e925b24a169fcbffac6d9f9fa5 : SUCCESS in 31m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ebd705a363f242b2819daf136ffa7e54 : TIMED_OUT in 1h 01m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/413eb7e7fd194c36aae29bd6449608b7 : FAILURE in 40m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/618ec4d3367e4dc1a05eca8d9ea61ac9 : TIMED_OUT in 1h 04m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7f273b56b9241f3ba9d8cd6b8050e14 : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1decca597c9b43838deacc56542cab83 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388f2301469e451e963b930c71b2f0c7 : SUCCESS in 11m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7825bd9867e844b8aded0fc9d89942ec : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2ec27199e3f11ee33deb0fd6c4f048d6d592b3c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 15:33:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a900e67c2e73b47d663acbc9f04dec37194a272e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:49:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5effd7f9f994ac589d282f8b0321caa : SUCCESS in 16m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e9bd331b75c24441bbd6b298cb731e0b : SUCCESS in 1h 08m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d886c9c9f0c2456288c5405940eccce6 : SUCCESS in 36m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4352725bdf3f40ec842131ac5809bb5c : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a6ffb98617e74331a80de4bac6a45f78 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f61dea77f41a4424b6de02560bb358c6 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfeb5f237a0847b394d9309a45c9dbd5 : SUCCESS in 11m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/838edac218ae4a99a27aafc3fdae6bf4 : SUCCESS in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2f19d327224e72b42ccc647d8cca8f : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/584d2da7c98446b0999e1a88fadeabe0 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d853f0143a824a8ba73f1603d273beba : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8467916260140d9b569d47795e02bb6 : SUCCESS in 15m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/34bd3903be2645c7ad41b226effa6539 : SUCCESS in 23m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b9cbe23838314e39af0121d83f33ec8d : SUCCESS in 19m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8d6ca6a2b90c454d977bfb681623ab64 : SUCCESS in 22m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2595f12b3dbd44fbaeab50a36f7b5e08 : SUCCESS in 24m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4f7e5901933f4844b487765f1f2da02e : SUCCESS in 44m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddc04504676c45e3b222f8de723ff638 : SUCCESS in 35m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8c5b7a441b449ba955957af04104db9 : TIMED_OUT in 1h 06m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9265ac62c808480ba9806ac846030094 : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab0e5020f59452599f5df72f0bd2b79 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4edbc27cf83d436f8642d50a62176577 : FAILURE in 14m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ef54754ce3a43b5972a12279d823784 : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":2},{"id":"70ac2ced77fb9c669f07a2d05d21edd886572678","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 19:19:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68d8b1b82b57270310d7992518858b4fa0c25f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:38:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e243413694346f28e0f3cb87e545e72 : SUCCESS in 16m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/1e953fcc70bf4fcfa174f60d9513e82f : SUCCESS in 1h 04m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4cc6173c0898419bbbb21d4ee5e25c23 : SUCCESS in 46m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3713e84cb4a4b8ab3037229801d752f : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b77828ac6bc462fa47c28fd69abe95b : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e6ad4368c77413a86da746f1664c127 : SUCCESS in 12m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/913ce2d779534dbb940cdcfdd5767a55 : SUCCESS in 35m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/53e83417e2264ef4b8b7aba2da8b8221 : SUCCESS in 20m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd0c02400804ad7ab3215f0b14f0489 : SUCCESS in 13m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d385772f8f0b4699a58c5d8d40315809 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ccfad5be5c5424a9afee9ee47da72b5 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f933ad74b58c490a8377f3ed3a275074 : SUCCESS in 20m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e6e7601d013499da9f490878bc3b093 : SUCCESS in 21m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/59793a1071254f04905b531c6d5817a0 : SUCCESS in 21m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b0ce1009ec9a4a7383c0b873768e582f : SUCCESS in 26m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3d4463abd9d4b99992ef96e09a79556 : SUCCESS in 21m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e82f7beaa6b4a20bed956e67166c5a5 : FAILURE in 50m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e6ac14b75134461a3f68e5228218c39 : SUCCESS in 39m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9c10d64a3d448779cfbddb27304253d : SUCCESS in 42m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6425ec71e4a44e2a8a029c3c2f88004e : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10c4dc2891343509564851371c87e20 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9291b6404586450dadff259af80dac2d : SUCCESS in 22m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00e285ccf73a42ecad0cd3d3a75bf69d : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9c7f3e680dfe54753cdf0999af744288458a8cd0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 22:46:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c7357c44a632b74731574cb65db05321861555a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 00:05:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635ca95697554c10b4d20ed0705a2956 : SUCCESS in 16m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/66e4f47c2c3b47a4a11c8797593c9430 : SUCCESS in 56m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/619ade3a9291467ba5b09357625f00eb : SUCCESS in 49m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38b392198bb0449e8d781d2e73041fdc : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ddc1963da0b842b8a6fb55a3175954f5 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0be8a19d5a86486c98f3dd86c16b036c : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b44795cda286469ab10605ef89737c6a : SUCCESS in 17m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc4a8ec3319843a995be0cad3a10add5 : SUCCESS in 12m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f627c44fe634041beea36e1ebf81afc : SUCCESS in 19m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a361bfbdbd3f44e1aa8925ee6fd8fadd : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7786e779f43046458eae7db011efd42c : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7fa47ddff1b49a0b2d483b52d43e74f : SUCCESS in 18m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/14cce083cbdb44929e9e09c345e5373d : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7d108d6499ef417882be5f6e1b0d6bfa : SUCCESS in 17m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5d9be6f16d544b03855861930d9e17c3 : SUCCESS in 17m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3caa433a959d410d9533530084d28e7c : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5477b45db03a4f75a3b5249859d5b77f : SUCCESS in 43m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b8704179fc64fa0aa90e19f0e18c439 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dea9e382cdf04c3f85ce45b6361a005a : SUCCESS in 42m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6dafaa2d2aa4d9a9a8a4ef3db6c00d1 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67568cf1f75b4542b2a9238e3de3144a : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b88c88b95c524ce5beeeac06a81d3612 : SUCCESS in 22m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42b82cda15774fac8885d612ecd8f5e1 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0e0204d14cc9bc358c3e4eb3eca279956da7233c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 00:06:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e91eae855b5aab121f509eaf987b0cd3b6ab2fa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 01:42:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a015867f5bbd49ca86b3e7bc4e44774123dcc448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 01:42:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf0348c77d04479baff362ee3d64ed6 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/171dc96974684c9c81d748e3b864a91a : SUCCESS in 59m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9042f5cdb2a40058694472f95b2781e : SUCCESS in 38m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/47dc9d5d421642e285175b61c0a343b5 : SUCCESS in 8m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3770c91ef239470b982afa0e6426fb56 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae4af4b80d98470db3f2b276f7deb4cb : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3625b56f18c64c33bf1e6d05abfd2171 : SUCCESS in 14m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d331ee5a00441e087112a616d5cb13b : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd8f83278aca4ddda7555187f22416a2 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/daf0fa521fc4472b95bd099f69706f9e : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e50b2b2e9a0748a783eccf19c99339fe : SUCCESS in 21m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a620aa893ca48cfa9f015eb26af96b6 : SUCCESS in 22m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7a59882137ec473b9691c189ac36f4db : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5166a66cd82d4002a811566a2e526223 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a5eaa3e528e54b7881d485cdfa3a1330 : SUCCESS in 17m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c1a63927aa84f069fd433c5a345e3b6 : SUCCESS in 50m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82d97592a5234674ac74d9f503615085 : SUCCESS in 37m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/605337a2dc894d15aa3086e4295d1012 : SUCCESS in 41m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5479ec75dcad400d8bb4a76b7e493cec : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609c311e9708419cb4c97cdaf1faedff : SUCCESS in 6m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32a6f594fcde4ad28eaf290c862c5daa : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f1baa689d1de6ca98e0053901d1511b66b7995ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 01:43:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1843c38378644db888e1f077d9aa4f4d : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/514867df40d54c57b4717260e9bb1a0c : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bc25e6988a3a48f0898b3367153b062f : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f87ae817c0d23e11bafb24af8fb5702504ba94e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:42:27.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 508c444697e9d0d721adc0bcc0b863b43ab3768d","accounts_in_message":[],"_revision_number":2},{"id":"cf3531cb8574f7c755dcb3dce873c252414814be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:42:38.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b00df2969698dca789bbc42adc4331308b2db753","accounts_in_message":[],"_revision_number":2},{"id":"7976d2a17ea9220839191cac610f667c9373f8d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:42:50.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 91b2cd0acba786f29fc40c2261cecc83e685b492","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b8073c7703c42a0b2b07536cd84e16ebf811b04","revisions":{"b24c5877598b41f843dcbce13b9702da275be25d":{"kind":"REWORK","_number":1,"created":"2020-09-11 18:46:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/751384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/751384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/751384/1"}}},"commit":{"parents":[{"commit":"4fefdf4d83bc43f86014906ff46a936168f71a69","subject":"Merge \"gate: Make rolling upgrade job work with either 60xx or 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fefdf4d83bc43f86014906ff46a936168f71a69"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 18:46:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 18:46:08.000000000","tz":-420},"subject":"py3: Fix swift-dispersion-populate","message":"py3: Fix swift-dispersion-populate\n\nWe fixed swift-dispersion-report already; -populate needed the same fix\nor else it\u0027d hit a \"maximum recursion depth exceeded\" error.\n\nChange-Id: I2d22e087a88c9e8003621feb26460ab6e5ce2a57\nRelated-Change: I24f4bcc3d62dc37fd9559032bfd25f5b15f98745\nCloses-Bug: #1863680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b24c5877598b41f843dcbce13b9702da275be25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b24c5877598b41f843dcbce13b9702da275be25d"}]},"branch":"refs/heads/master"},"5b8073c7703c42a0b2b07536cd84e16ebf811b04":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-11 19:35:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/751384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/751384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/751384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/751384/2"}}},"commit":{"parents":[{"commit":"4fefdf4d83bc43f86014906ff46a936168f71a69","subject":"Merge \"gate: Make rolling upgrade job work with either 60xx or 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fefdf4d83bc43f86014906ff46a936168f71a69"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 18:46:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-09-11 19:35:32.000000000","tz":0},"subject":"py3: Fix swift-dispersion-populate","message":"py3: Fix swift-dispersion-populate\n\nWe fixed swift-dispersion-report already; -populate needed the same fix\nor else it\u0027d hit a \"maximum recursion depth exceeded\" error.\n\nChange-Id: I2d22e087a88c9e8003621feb26460ab6e5ce2a57\nRelated-Change: I24f4bcc3d62dc37fd9559032bfd25f5b15f98745\nCloses-Bug: #1895346\nRelated-Bug: #1863680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b8073c7703c42a0b2b07536cd84e16ebf811b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b8073c7703c42a0b2b07536cd84e16ebf811b04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
