)]}'
{"id":"openstack%2Fswift~752327","triplet_id":"openstack%2Fswift~stable%2Fussuri~I131850d2a5c6164dfd48c9c95885d4754b5236c6","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"I131850d2a5c6164dfd48c9c95885d4754b5236c6","subject":"Switch to newer openstackdocstheme and reno versions","status":"ABANDONED","created":"2020-09-16 21:48:10.000000000","updated":"2020-09-24 03:54:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77db74a028001e72d82d2df90270154ab49c07fc","_number":752327,"virtual_id_number":752327,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-16 23:15:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-09-16 21:48:10.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 21:48:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-16 23:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f62b270b5e887a940f3ec3c096e0c9176ad32a7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-16 21:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21c9df4c7c1177dad67fa244589b56858900107c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 23:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9dee1e9c7eb461491aca8ddbf355bd7 : SUCCESS in 15m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/86402e66593c46ad99e1237f1da317f2 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e638e7c18c84301ace8e99645018a56 : SUCCESS in 5m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4d1e653245184b6eab812a09bfacbc9b : SUCCESS in 1h 11m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fd81e6af64246268e1e6b5ea836b36f : SUCCESS in 47m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a85929ecc8f4aa3903fc6d12519edc6 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/79295451b9bb4e54a6c502c50b575de0 : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59fe7ad161404f90926806ffcdd85096 : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33ff6030997c46c69f8c7a938269de8d : SUCCESS in 12m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9a1bdbfd0f14a499aa0df27c33b3634 : SUCCESS in 12m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ca43f534bd054d569e7648d43e21b619 : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d839a9066a1415bb9eb68a4f3436ac6 : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/84275362a88d45afb0b849238ab37fa6 : SUCCESS in 12m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4a3f0f8df15d4c1990a0c33e3ac79041 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3435cd1727f04cfbad532cf425e10b93 : SUCCESS in 10m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3aa398fc3c5b43fbbc5ee1c8aef8a616 : SUCCESS in 15m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d7b19400b53d490c93001354c851fc5f : SUCCESS in 9m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a192b5b654f14b32b19f78a3e7d116f7 : SUCCESS in 15m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fb33932b278f404780afd0ec152ace02 : SUCCESS in 13m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/38ca748d8f32414f8bd8679fc69aa8a9 : SUCCESS in 15m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/68758e431085463ba6fee046a64d42cf : SUCCESS in 20m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a89121256a2448b9df087e2b8aefd49 : SUCCESS in 27m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f0a5a4a7aac2465f9a636bc269009956 : SUCCESS in 46m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59eaecd4ea5d439c9746a645868eaf40 : SUCCESS in 29m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5fd14d46cbc74afd92bb66c54ef354dc : SUCCESS in 30m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e8dfb093ccf4f4084dbcb5eca25ed05 : SUCCESS in 30m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6afac1790d2c43608b070275d91a63c8 : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f3628fc2a946e5853c9a5f43a2805a : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5db266486f67457d96de8907680a5eb6 : FAILURE in 9m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8504c2d8e6d94435b1a839fdc9c2755f : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"77db74a028001e72d82d2df90270154ab49c07fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 03:54:49.000000000","message":"Abandoned\n\nI guess the docs job fixed itself...?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff9355b1c0b7f841828475aacb52d468358b58ec","revisions":{"ff9355b1c0b7f841828475aacb52d468358b58ec":{"kind":"REWORK","_number":1,"created":"2020-09-16 21:48:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/752327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/752327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/752327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/752327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/752327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/752327/1"}}},"commit":{"parents":[{"commit":"087391157426b94708c5a139b20080c933809ebf","subject":"Remove \u003cpy3.5 dependencies from requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/087391157426b94708c5a139b20080c933809ebf"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-02 16:53:59.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-16 21:47:00.000000000","tz":-420},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nDisable openstackdocs_auto_name to use \u0027project\u0027 variable as name.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I131850d2a5c6164dfd48c9c95885d4754b5236c6\n(cherry picked from commit 3d105b623d13d8fe6bb4db676196a44ac3db464f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff9355b1c0b7f841828475aacb52d468358b58ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff9355b1c0b7f841828475aacb52d468358b58ec"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
