)]}'
{"id":"openstack%2Fswift~752334","triplet_id":"openstack%2Fswift~stable%2Fussuri~Id0e8d64df0f9ea8642d63cb60e7fa7690953789d","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Id0e8d64df0f9ea8642d63cb60e7fa7690953789d","subject":"docs: Add librsvg2-bin to fix docs build","status":"ABANDONED","created":"2020-09-16 22:05:05.000000000","updated":"2020-09-24 03:54:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"191e3980a1ddda88cb5c26e74462e0de72a314ba","_number":752334,"virtual_id_number":752334,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-16 23:16:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-16 22:06:17.000000000","_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":1,"date":"2020-09-17 01:23:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 22:06:17.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"},{"updated":"2020-09-16 23:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-17 01:23:07.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"}],"messages":[{"id":"76148bc89c3abe95905dc5eab0ffc5770871cd9f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-16 22:05:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f8c1c0227d0e07e030c01e581d50b1920041881","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2020-09-16 22:06:17.000000000","message":"Patch Set 1:\n\nhere is the nova\u0027s bindep.txt\n\nlibrsvg2-tools [doc platform:rpm]\nlibrsvg2-bin [doc platform:dpkg]","accounts_in_message":[],"_revision_number":1},{"id":"ff94ea700ca0447ceb95a3bc48ecda5c233943f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 23:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5682e845fd3e4b5d8722c22f6d46509a : SUCCESS in 14m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd816e349144e7d82b14c45dbc146ba : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c173fa62185f4315b0b375e221ef7ee8 : SUCCESS in 1h 07m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/584163b245de4c34a05ebed1ee41f400 : SUCCESS in 59m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cd148fba14f481981cf2278e812ffbd : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d0679eded754707b5b3c96fd95e3fb1 : SUCCESS in 12m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c8d8f1102e74c3db2394dca4f812722 : SUCCESS in 10m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd8a497263d4532afed05b208988965 : SUCCESS in 11m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4eef4093b91e44fcac964c135eca80a3 : SUCCESS in 11m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b96d9b9cd3c74e2495061ccce90ba013 : SUCCESS in 15m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2286aed25fc94fe7bf325793c7bce270 : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3edcaeb8d8494f6d822e7f74bbedb8ac : SUCCESS in 12m 16s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/75a12c180c2b4ecab74f527eaf64be0f : SUCCESS in 9m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b89f3491451c43668911a3f09d3d73a7 : SUCCESS in 13m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e1c699ec38064d5c8c5939eeae5ece27 : SUCCESS in 14m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1b28e8c7edfb4a6db5037a28a8b1505b : SUCCESS in 9m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/addb399e04894725966f630060bcd8c7 : SUCCESS in 12m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/032e4434ede748958dbd121dee6decec : SUCCESS in 12m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1cd91fea88144baaa25a8e3ffa456090 : SUCCESS in 10m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/805315c5ef524ea49b265dc01a9f6049 : SUCCESS in 15m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbfb739c2c2a4583b54c7c8f195f48e4 : SUCCESS in 20m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6f28ca1ac3ff442485b5a0a2cc9b28c2 : SUCCESS in 46m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4db96ecfab984ab3be6453a2a3e69c59 : SUCCESS in 24m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ccef80a6aa2b4d2b9d87776f577c9700 : SUCCESS in 24m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b28ebe63c2b47f794637cb8e19ed821 : SUCCESS in 25m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9b594c2f74d470aa39c71f74adbfe23 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077742b320d344cbbfa4cd948a1fc180 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5f3f61816e748daa72bbe01e4b4b601 : FAILURE in 7m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1226ce50f9473dae22a395b72ee1e5 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4b0c89844dec5a01b6aea02508f8121efa067d52","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2020-09-17 01:23:07.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. That was suggested by clarkb in IRC.","accounts_in_message":[],"_revision_number":1},{"id":"191e3980a1ddda88cb5c26e74462e0de72a314ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 03:54:33.000000000","message":"Abandoned\n\nI guess the docs job fixed itself...?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3704cbc2bd2bfae3cb3a2ef655c9175cf5c5c8e","revisions":{"d3704cbc2bd2bfae3cb3a2ef655c9175cf5c5c8e":{"kind":"REWORK","_number":1,"created":"2020-09-16 22:05:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/752334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/752334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/752334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/752334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/752334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/752334/1"}}},"commit":{"parents":[{"commit":"8ed74c264605e12717c93e9b07f4ff50924e619f","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ed74c264605e12717c93e9b07f4ff50924e619f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-16 22:04:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-16 22:04:17.000000000","tz":-420},"subject":"docs: Add librsvg2-bin to fix docs build","message":"docs: Add librsvg2-bin to fix docs build\n\nChange-Id: Id0e8d64df0f9ea8642d63cb60e7fa7690953789d\nCo-Authored-By: SeongSoo Cho \u003cppiyakk2@printf.kr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3704cbc2bd2bfae3cb3a2ef655c9175cf5c5c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3704cbc2bd2bfae3cb3a2ef655c9175cf5c5c8e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
