)]}'
{"id":"openstack%2Fswift~752620","triplet_id":"openstack%2Fswift~master~I444e018b52279ffed96158a96cf66123a0e6267b","project":"openstack/swift","branch":"master","topic":"babel","hashtags":[],"change_id":"I444e018b52279ffed96158a96cf66123a0e6267b","subject":"Remove babel.cfg","status":"MERGED","created":"2020-09-18 06:11:15.000000000","updated":"2021-06-09 21:50:41.000000000","submitted":"2021-06-09 21:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752620-babel","meta_rev_id":"682c94c6765681fd12d28aec311f139d95fa44e2","_number":752620,"virtual_id_number":752620,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 21:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-25 06:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-09-18 15:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2021-06-09 18:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 11:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 15:41:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-09-25 06:44:00.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"dfc33dc8ec5fb86a9da689030184ab1a19319426","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-18 06:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a3a7cf06e9a40c3415f8228631591952b116c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:34:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/001796488dfd4d7d8f9abdd8eaea4d3e : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ecf63a17af48409811f20123ba6dbb : SUCCESS in 19m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/39a9f18f794041eabaa9066a11548333 : SUCCESS in 1h 00m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb9c0c8c51ca4192bf22af8743d592f7 : SUCCESS in 44m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d511d7f3a6af4cb088f351cb1dd85d6b : SUCCESS in 11m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/295823f1be7141199af583965248ce3d : SUCCESS in 11m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/21683c856e4a4e9a8c5822f457c18a71 : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e5c16cf5e2413884bb298757581a19 : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27d16949e73e42afb7107087bd397cfc : SUCCESS in 13m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d368b671d94a4e2786a40b946f40fd51 : SUCCESS in 11m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8048ead1c49f44c2b647b8617a575afe : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/790d31f522434dc5bb688b5ebdf32248 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/920c172949544bf89c4696c94a4782e6 : SUCCESS in 18m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6d052a82d5cc47998f4dd444b153665b : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7d725f773c1542efa2361c837f074a09 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8423763d8e95401dab03ad1177463b48 : SUCCESS in 19m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/314c1542ad374907ae20cd9282f1ea01 : SUCCESS in 23m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/be04f7323a8743b0b9ea5d85447b0415 : SUCCESS in 39m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/debaa64eeea847878a55a93bb1db77f2 : TIMED_OUT in 1h 03m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60e17561a9894320b8583fda20b478f9 : SUCCESS in 39m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b9331b3f9cf455ca27474505cfe45e2 : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f957b72fef4cdfbd8d38d1acac64f8 : SUCCESS in 8m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f9b54ec92bc45b8904f5a4f74afb836 : FAILURE in 10m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/399ba5ea03ad4282b984135e1d8c1416 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2577bb3ed7223d15f35652af8063e7c73fc9d282","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-18 08:52:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5c838a237fce618a8ccbc4637f7443627ebf3765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:07:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22a77c76c17447e5bc01dedfc9b0ec7a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437447929ffb485eaa51a75f4b445fb0 : SUCCESS in 15m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ee2fe6b8068849689156f089c8c989d4 : SUCCESS in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dda44fbe90634b6fa17ef1bda573df1c : SUCCESS in 48m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65d51aa0b4f44c94a0f6adacc46189e4 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95304a803ebd4eb895f5228cf1105996 : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1882e154ea0d48e7b7b47c4e8ab39c35 : SUCCESS in 11m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d76951b13ca24868905882eec39fb19c : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9cc5da58aa1244a8ac73b0738e8d0ccf : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed739492a7a4b8297c70fa082421973 : SUCCESS in 11m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2608e8fea7e4d978738dd83b1f02c54 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/798003c35d9948689c75086878a021a8 : SUCCESS in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d19a126e25a240058cbc428e41fcc0b0 : SUCCESS in 22m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/393e6c56fcee422fbe58974b2cad82d1 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a97163d85b204afd94e79381086ef5e6 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ffa90cf5d3904d7492e2b4b1d55adfae : SUCCESS in 19m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1597a8fddc8a47b5b9c1a1a8b7fcf02a : SUCCESS in 22m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d91905a991e54a2190599d6eb198d08c : SUCCESS in 40m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05882ec0adde428599b9f4eba6f29bf6 : SUCCESS in 44m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/704e1b46ae0f4b208905267288bb3e06 : SUCCESS in 39m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4ca21690ac246ca9d2b009847f2afb7 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29e4c8c9624c4e98a95fbcad029d565c : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/814d064902474608babd59da1748311f : FAILURE in 8m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41e3e71d409f4ed1940720a1d89d532f : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":2},{"id":"71ed7128c1f6bccaa43c618a19608a53b7f72266","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 15:41:38.000000000","message":"Patch Set 2: Code-Review+2\n\nAny bugs or mailing-list messages where I could find out more about the history behind this? I\u0027ve never really thought much about these bits, so it\u0027s probably just as well that they get removed -- but why were they added in the first place, and why are they now unnecessary?","accounts_in_message":[],"_revision_number":2},{"id":"2059d2385083312fba5aa8997c0226dcc5a72e72","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-25 06:44:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a7b70c31c6fac641e5de6813db0e1b8bf2c63015","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-09 18:53:03.000000000","message":"Patch Set 2: Workflow+1\n\n*shrug* Why not.","accounts_in_message":[],"_revision_number":2},{"id":"78e032bfb9b661ed799e104b8108c4c2f048fa01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:53:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3790bef87d6509effc3c31104a9fde39511394ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:45:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea37b47b37434288ae010b0a67c24989 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439e2a48313a416bb505362a407755c4 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6e7baa755b4645f2a2ef1bf1c0fac981 : SUCCESS in 1h 05m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02ad4724094d458d899d6c5ef3542339 : SUCCESS in 48m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c8dfde46fb1c4a7dbc7eea9c700bbab8 : SUCCESS in 8m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e8b66cb8f32e4524ba861662484655ea : SUCCESS in 9m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6e7c47c008747799ff08acc783f6e38 : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5be30a969a744844b49cde2c0d3eb34f : SUCCESS in 16m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2dc271a0dcf64c63be4bbdc90d3de94b : SUCCESS in 17m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ac09eb750342efae2fc8c2231317a8 : SUCCESS in 17m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8c9b9308e344152b7e2a34162a4c18d : SUCCESS in 17m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0681a99a99bd484181c801230c87e341 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64eff2b4b33944b886fd59b6e1cba285 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5c156f9c2b948fd991424b3d0476d42 : SUCCESS in 22m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/72cc4b0c19c54e52b69e11d31c271a8c : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/69e6478104b54b499f9568bdab574685 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4cdea5619f3b482a8a7a7059e10b3fe9 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4cfd52b933814a488795374e16d93438 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/24616d5c4ea04d8588704bbc42b8e655 : SUCCESS in 1h 03m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/435e92d418e54f97a7450630721bd149 : SUCCESS in 45m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cb9d18587a0432a96e5cbff01cc5c1b : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb666e66977d464ea47d76f69db24f92 : SUCCESS in 37m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8aebeed8af84b739d1a57a01c8cb344 : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec5b0a514f64c23bbd4d2f4cfdad9bb : SUCCESS in 7m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b85b402fec741429f034170b1e2117a : SUCCESS in 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8704eab7b4ed4657aff3438c89a13df4 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d1b55480a07d8a79449c2834e770c5d84f29cddc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:45:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"682c94c6765681fd12d28aec311f139d95fa44e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:50:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e2c6a4565043d7b82135a07812f579 : SUCCESS in 1m 15s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/edf98062f73f457fbdaf9f3c3eb79e54 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1c14bd1ffb2c4006b9eddbe0dc937e1c : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66da8eaeef49db675b66e92336c947e6522e5f71","revisions":{"d1b308ce84b317b850d7cae6119df11f57278cda":{"kind":"REWORK","_number":1,"created":"2020-09-18 06:11:15.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/20/752620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/752620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/752620/1"}}},"commit":{"parents":[{"commit":"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2","subject":"Authors/ChangeLog for 2.26.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-18 06:05:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-18 06:11:11.000000000","tz":480},"subject":"Remove babel.cfg","message":"Remove babel.cfg\n\nRemove babel.cfg and the translation bits from setup.cfg, those are not\nneeded anymore\n\nChange-Id: I444e018b52279ffed96158a96cf66123a0e6267b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b308ce84b317b850d7cae6119df11f57278cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b308ce84b317b850d7cae6119df11f57278cda"}]},"branch":"refs/heads/master"},"66da8eaeef49db675b66e92336c947e6522e5f71":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-18 08:52:47.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/20/752620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/752620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/752620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/752620/2"}}},"commit":{"parents":[{"commit":"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2","subject":"Authors/ChangeLog for 2.26.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-18 06:05:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-18 08:50:10.000000000","tz":480},"subject":"Remove babel.cfg","message":"Remove babel.cfg\n\nRemove babel.cfg and the translation bits from setup.cfg, those are not\nneeded anymore.\n\nChange-Id: I444e018b52279ffed96158a96cf66123a0e6267b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66da8eaeef49db675b66e92336c947e6522e5f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66da8eaeef49db675b66e92336c947e6522e5f71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
