)]}'
{"id":"openstack%2Fswift~752750","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I997710114efa4ebf470a4a49a9d3eb7c45a8c3e0","project":"openstack/swift","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I997710114efa4ebf470a4a49a9d3eb7c45a8c3e0","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-18 19:17:58.000000000","updated":"2020-09-24 02:28:52.000000000","submitted":"2020-09-24 02:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752750-1600914143416-57e98530","meta_rev_id":"4da313d4ae04382041e44fa2b51419d8aeb6bad4","_number":752750,"virtual_id_number":752750,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 02:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-23 19:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:33:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 19:33:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-09-24 02:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44a2ec5fb82bf5e1b1610db4ce87a748cf4b891e","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-18 19:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767b1f5729e65e61926646f38cfc2123b264dc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7ae4212916412083ef340df4c175df : SUCCESS in 17m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9010471d29ce4b81ad25c4508086d274 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a650dadacf5b483d829a47514f0c6f29 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ae2ebbfe06846d3a3d9863da2ba03dc : SUCCESS in 13m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a228c02b21494591e1c9dbfd32a089 : SUCCESS in 14m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/11005808986c4d7f8fb06816e101f33d : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3046635be43641ce9f49d1db8b756b01 : SUCCESS in 12m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/290a6d7de1cc40c7b2aeef4d72fbc55e : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947c8781e70b492895fea5a6fef465ba : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f0982b08f2592ac2e0f8e1d458396adcc057be60","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-23 19:33:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a505cec849f79ff49759ac0e1caccfb549dd875e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:33:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70aa23523cbea1baa2cc4b072eb3b888f0bd42b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 20:53:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7aa9e1723fc4a81bb5ca005805f9054 : SUCCESS in 9m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a8a9ba6d8e0845429fe8263eda57ba68 : SUCCESS in 12m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e2f76a9c4fac4a818b66de3f4595e13c : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70962ff51e99471391163b812ca1976d : SUCCESS in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c57f0a81386428a93915dfa853a2b48 : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a3a2b63cc2fd484ba053cec388b83546 : SUCCESS in 15m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b81bce7f88114dd69fa6516d4cab83dc : SUCCESS in 13m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7aea192066840e5a613a61b733730cd : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05aade9f7d5b4deab1a8ddfdcafc3c25 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9968704081004652af7fddb4d825bced : TIMED_OUT in 1h 01m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/849a964df0d54946bb2b5352e1f0b110 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c117c3951dd41748bd81f340037f530 : SUCCESS in 18m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f088b493eb47028ce731d01774e8dd : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0e5e47adef4d9caa88ca507113d9a7 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ba39428556f9d3d1781765f15f2476c18f2d9d79","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-23 22:02:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b271e8a8ad51bd14bba1fe4b9c1190c40b592f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 00:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e8a0a2ca02455a8dd6c65732b13f75 : SUCCESS in 7m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a956b2625ac4cd7ba426f79eaab23a9 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/66c9c08ddc174f58800dae26d1e3a78e : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c26afc80f1ff425ca3aa129040b197e6 : SUCCESS in 12m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6798b81fc15842c59a7de4ed66770d25 : SUCCESS in 12m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a7c737e489c472c80dbf3293f626478 : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b1bdda9477154531a6232ebab764f9db : SUCCESS in 12m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7365bd9fcff14de8857baad47b443df1 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05acf597f7a449ebb1c3967159376dd : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d018124d6a6ae54c32a67c02ebdc6f88815d6039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 00:02:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d88c757c687686d4b5dbbc568522306454bc25d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:22:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8827a1d9c3464952942220690eb0f47c : SUCCESS in 7m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1602f5ba643d46b884311d355eb78ef7 : SUCCESS in 10m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/42083f38f7bc4c1eb8d7f7eaf6f6642d : SUCCESS in 10m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c242f59d03f4e0e8e4589913e6cf306 : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7a0e39b5c914f2ea5e20104d8e2b9f2 : SUCCESS in 15m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/21941733bbd24dfc974518ae0d981be2 : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/912d23631b8d427f8fe112e6646d1f35 : SUCCESS in 12m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f872cc7b9ff84bbc954d30ada5b0a337 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de3cff4708a44d579c38e1defa2683d6 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8511615738f44331a78a210cb3bc1103 : SUCCESS in 17m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b5b2782e4c084961b0b3af74425d0da4 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4a64ec17142b4119987270a6ab6bf0b1 : SUCCESS in 17m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337fa5f1a65b4b53bb8fafde40e58e10 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5775b06ddcfa42ef87283b628703a736 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f4c16b730f4f94cd84f329448ffc5676134a29f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:22:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b86e2c2d9290f2c86ea176098a84f1ee426cd42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42a2e1b98a64cac8dda0d3b0cd336c6 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ca8e497783af480e92bc8dd2a45af4a2 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/63651df2e47540d38c12afe0b87e3ab6 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3095fb313924c96a53847dc595d9bec70cf533dd","revisions":{"3095fb313924c96a53847dc595d9bec70cf533dd":{"kind":"REWORK","_number":1,"created":"2020-09-18 19:17:58.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/50/752750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/752750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/752750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/752750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/752750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/752750/1"}}},"commit":{"parents":[{"commit":"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2","subject":"Authors/ChangeLog for 2.26.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:17:54.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:17:54.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I997710114efa4ebf470a4a49a9d3eb7c45a8c3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3095fb313924c96a53847dc595d9bec70cf533dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3095fb313924c96a53847dc595d9bec70cf533dd"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
