)]}'
{"id":"openstack%2Fswift~752751","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I94867b1a3c1b51e9bb21398593e198cd55fe5dd7","project":"openstack/swift","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I94867b1a3c1b51e9bb21398593e198cd55fe5dd7","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","status":"MERGED","created":"2020-09-18 19:18:01.000000000","updated":"2020-09-25 05:33:19.000000000","submitted":"2020-09-25 05:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752751-1601011867196-276e733d","meta_rev_id":"128d2386c34eb7c9427dc79b533ddc449b4e81b3","_number":752751,"virtual_id_number":752751,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 05:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-23 19:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 19:34:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-09-25 05:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c67c93df4509bff55de9ac7aff0a9f87830d1b41","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-18 19:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cdabaac00071e88da5c4d5a58107290373050a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:13:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9882a0f076b74e088e7231ed68a0fa53 : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a622db7777674aa4915545ea3dde71a6 : SUCCESS in 4m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8367885faab342e09874d988513bc036 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f79478e3c42048dca49a76feed75d80e : SUCCESS in 10m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce0ebc69ea474a158b599284aba1485a : SUCCESS in 12m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c409857154b45179b550b6cdd4ef3b9 : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c2b730cf58e4bfb994b050a8c61d728 : SUCCESS in 23m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/21976c2eef854749a35cdc9ee1442871 : SUCCESS in 14m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/336fcb47613c490d971281a5eb66762f : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9130d85ab98442718b88630668f390d2 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04e7e11b818840f4be6fbb9994c47227 : SUCCESS in 18m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4bccc223d88947fb92384c502eaa19ca : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4d44baba3de3498982107336c56c5af0 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a7888036c2a944c6971a0f1d7dee76c4 : SUCCESS in 23m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64038973438b47f798244ca64b4afb74 : SUCCESS in 23m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/79d19e39641a47108a9bcfccfb227848 : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d7dd9d3f0d6467dad269e648baf6814 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae2d52647614499ac6cf759fb9f1b3b : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59d948e6489e460188a4a1eb1cf82ae4 : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"901f7cab75ef8f53d1125b0584bae7d08ab1ef0e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-23 19:34:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20f4e5a1742359e4ccbb5d972c0c76b26cbffeed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:35:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48a7d6d2e94fe7b36d93e3ed773893f1cd1304e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 20:53:21.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"3d29ad6acc4a63051274c9b73f7e6a2eaa5bd1d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 03:53:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89bb80da349b50e14cff1a9b084570db1452ca33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:47:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a890efeea3d7449fa85d2bc2906eda1e : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab1b7442a944062964fb6f14322d542 : SUCCESS in 4m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33973a9612cc480dad5a766c67e94857 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a70d2460fa44dc6a782bffd06d8a52d : SUCCESS in 11m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1bbcc08351564641a6f013d26eef4b80 : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c8c24334e4147a392b2cec1fc516659 : SUCCESS in 42m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67cc72eac6914b52a24686eac1c658fd : SUCCESS in 13m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afa95680097b4b04961d18c1cbf3312d : SUCCESS in 12m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10dae9f112f94a4bb57e49b6bb8a8f53 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31e5b9fe892a4755917b6025095273b0 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5719981284334d52a32b042697af3d61 : SUCCESS in 52m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b875c1ae4e9b4f5bb71e290d12ba1e4b : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d25f73dc58942038891c459ae21cd5b : SUCCESS in 49m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/af5a3ebe90b74ff59605797b0bbc618d : SUCCESS in 19m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a54f6edb930e4d57b54ae7b4afd3d293 : SUCCESS in 23m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9705a78b6ed546a2a2f8fc71203b164c : SUCCESS in 46m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1edaeea36cc84632ad66c4ef7088f3c2 : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26480f2950134f7aad9692f280b0079b : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11059291464f4268a1bd78bd33479d45 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b36055b78ebf6ef7d4f0dea9dd96b7adad0f755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13e4b263617ec736209822201ead5bcc3fe70207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 05:30:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f71e60980b84df1933ef2317fcf935c : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d15638457e10418e853346864ccb6cfd : SUCCESS in 5m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/281a6c0112f24252a150e27a3c68b174 : SUCCESS in 10m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f0e3f2355fa349a7ba1585fe0dcc643a : SUCCESS in 11m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b2aefe053d7455cabfe627132d4f695 : SUCCESS in 11m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d018bca99af4954b02feba01c7b028e : SUCCESS in 19m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c1d1b26f0a942eaaf000db3f1fbe857 : SUCCESS in 14m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/801bea9b226c4c96bbbb1948e5dce6a5 : SUCCESS in 12m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/396c1d0d75f541afb7e26b02823b1e8b : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/823979e02a5a41e5bd9fc25a9029caca : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e60f686be3e3401dbe258895a85af2f3 : SUCCESS in 17m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3c59b57e8c5e413fb9724e626c0559b6 : FAILURE in 18m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6e43053b9ddb4967891fb7ba743c7316 : SUCCESS in 21m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/307a11f41cf24522a596845c5153811e : FAILURE in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39cc91732ccc4b07b634df54033fee95 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d99a9a792b24aa1a0cd21f992db9d42 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a23b86035a929ae20abc858b0e325b5c439ee98e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 15:00:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfc1b10f41838ccdad0a8ad14c8a1ff5e28ac13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 21:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d164aea87c4f8495222345c5cd0cfc : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0f1623fa6834a38b175bfbf73aa5555 : SUCCESS in 4m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a05af41c4fd44292adf52cb2f86ee3ad : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21f7c5f6c0074991b57f1980f6f6d92d : SUCCESS in 11m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80975b45e94b45228bee04ea782062cb : SUCCESS in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66dbc184e7604835907f3a08861660b6 : SUCCESS in 15m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5891cad404a047d98b7a0b95c37ec808 : SUCCESS in 12m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/66f46285ed6e40799e422a79e5797b1b : SUCCESS in 13m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9b055ff7c0143179b4352d28d0fef78 : SUCCESS in 28m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/006941b6015c4ed4aaaae2ad7020317b : SUCCESS in 20m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b499d97d0e604fdab5771bdde2b91876 : SUCCESS in 19m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07822f3714ad475390dc35bd29a857a3 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/83bba62693c5409db87cb3225d4cc2dc : SUCCESS in 18m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fe094cc2be3048209d308a79b0c207fd : SUCCESS in 24m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c50d0b31b37480bad9abba49ae5d1d2 : SUCCESS in 18m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fedbac6fe1d74425b1afe25b5b590c3a : SUCCESS in 43m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c168eb30a464338872103215c7116b8 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb94c92379a649a9801093ae6ac4e37b : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca5031bf1cdc4eb1bb980f8416590855 : SUCCESS in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e117e116aa96c3dcead094c77dd453667294a68a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 21:42:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"532600ef008933893fa5e0a5c27fe3d264417553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 23:02:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf13de6183744e1eae2de06efc419f20 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/569f46a396b44999a95626c8d756a867 : SUCCESS in 6m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/24e1818843b84b03b4abf06bfb9bca75 : SUCCESS in 12m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/053b130d975e4c46b12b2315c552ddaa : SUCCESS in 12m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c2cf2890d69d4fe18fffa9eb4177d907 : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e16ed1c07e1471ca75bd25315cb1b02 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7991eb9635664a3dba6f0075c4d633cf : POST_FAILURE in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/77555d5ccaf1492da1c143f503b7c8e9 : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5a2414193b644bf829a1793089fd244 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b99537af65c4a27b07e52eabd9d0c98 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df12d23a2e404f939c1cc99f43008c8c : SUCCESS in 17m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a653ffa093b54f2ba9ca51346f903175 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28e33ab62cf0459680594b6b6e8cf5b7 : SUCCESS in 20m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39f9774363044a618f9fa080d1491564 : SUCCESS in 48m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ca6494a15244d35bfc527d92fe588f9 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e4f21d1a1ff4dd3b99ec198a153d583 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d2ca6d7f24c51941896af0d619276eb6a360cc72","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-24 23:28:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb64354d3559e10e2065f1da58a5b13d7cd8e853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:30:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75dd8adc8af4b29b3cda5a53bf5e50c : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b2279e32bd34a358768487afb7920d6 : SUCCESS in 7m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d845fa7a0c194cb0b66b78c482b5e5d9 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/723b62709739436e985ffbdd6efbc888 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6dba68b4d9e4fb5a4b81118553e009e : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/436e3ecb930a4005b2025d9fe4e667ee : SUCCESS in 12m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9fcc1234bb73449fa364df9d4954a037 : SUCCESS in 14m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ee20c6181e40ada115f4df83207a5f : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6272a38e2ea64a05ba4752de76fe7528 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2efc4135cda4a1a9dd95e951d297d31 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfb9068e04074d6e8452b4201f47ef99 : SUCCESS in 16m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/efbf1d43016a4d5c8669b98dfbf0c9d3 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b00092d51afa455ab2625aee38b14489 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/95e4ae1741294eec912016c8e4183116 : SUCCESS in 17m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86114681198e4b3db98e4275c034fa77 : SUCCESS in 18m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8fedf45ff0474633af48ce4bdffaea07 : FAILURE in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed9bbe6f337a4e2fbfa0bb7f3448274c : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd89c733c6b44245995b10f2514c74f3 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53fb5adf8d2941cf9a1bd72985004707 : SUCCESS in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4af59ab25584b7f141884930dbabdfd3747b769","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-25 02:06:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4171b77809f8a7cfb9000eab7970237edb5508ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 02:57:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00683a20844640d3887a0cf98d7d2a74 : SUCCESS in 15m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bcd82c8f47044c99f0f9f7254230440 : SUCCESS in 10m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06e4c12c1bfe4e108f3e532cc446834b : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a10d4b50c5214d7a941027c3553c7f73 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1df1e8d4904d4eae9770238f8eaa5632 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba372468bac84334b116b476d0e04910 : SUCCESS in 16m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c510fce610ca44d9b95afbf474ff95eb : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c2316a48790246d8983ae87ee568f0d8 : SUCCESS in 22m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fe7d9b45a5cc4605a9b997d680486ac7 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa121d441af041c8a01e2f7a43d9f806 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff0fe414eab848468a994ba6855e16c6 : SUCCESS in 18m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/29b96485cd124b37aee1765b4eb80b7b : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/63393ac67bc145fb8aa2e689989c946a : SUCCESS in 22m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/36b86559efbc489e91d64620680d27f2 : SUCCESS in 22m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b851c5fddcf45798477b31156db9187 : SUCCESS in 30m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53c063bd478048e49f316b6722cf88d6 : SUCCESS in 49m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e08824d2ad1441a8e9bc1e4fd8e3ac6 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a4d545c22b4ec183dffa44daf77ccc : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3398220876dc470b8803afbd030c47f8 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c94d242280e0b7613356bcdcd4bbd6350909c56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 02:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2af4ffd60f0ce8d5852632e2708ff0e24a71abe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:31:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a414596dd9400cbe206f6499f422ea : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34160f74a62c491c9e36347caa862779 : SUCCESS in 5m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0b344e6f62c7416891fb2f869e5517b1 : SUCCESS in 8m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e6cd696a431547a0af70a87716c5b405 : SUCCESS in 11m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db451b4de1bb4f99823321aae6769921 : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24fb593cd93f409ebe089b322cc3c097 : SUCCESS in 19m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/389fe180ad0945bf830b3c3a64266a61 : SUCCESS in 17m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/97bdf044399e4216a1202ced96133c28 : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12ca7f1c175a4a4c9788ec876674dd24 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83a5368b69ab41b1bc5eba3d977f5861 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83a1843678b9476fa71314f9e4995aca : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08c43388ea60431bb67ab351bee9c7fd : SUCCESS in 16m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d8807fcf5174489a8ab617bd2dab291a : SUCCESS in 18m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6277814cbbd449409daad1c865de7285 : SUCCESS in 39m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d61f15d354c4564befbb1f1faf94c22 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb8b91a8f5e445fa980b7f49d4551a5 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f5937fea5a096be61d6044e003393209b12c106d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:31:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"204bd9faec20a701dd3de0b69a067eed9053a14f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:33:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3fb1abbf8c41a59eda79e8ad9f050e : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de04997ecb2f4ab3935c0063ba4741f3 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb0f9d7085f447c29207ce2bb4da67dd : SUCCESS in 1m 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7d0feaaf723942beb705f94bfc6d8cdf : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d80b2b590bf189042af64e74acf045bb385b158e","revisions":{"d80b2b590bf189042af64e74acf045bb385b158e":{"kind":"REWORK","_number":1,"created":"2020-09-18 19:18:01.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/51/752751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/752751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/752751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/752751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/752751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/752751/1"}}},"commit":{"parents":[{"commit":"3095fb313924c96a53847dc595d9bec70cf533dd","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3095fb313924c96a53847dc595d9bec70cf533dd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:17:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:17:58.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","message":"Update TOX_CONSTRAINTS_FILE for stable/victoria\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/victoria branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I94867b1a3c1b51e9bb21398593e198cd55fe5dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d80b2b590bf189042af64e74acf045bb385b158e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d80b2b590bf189042af64e74acf045bb385b158e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
