)]}'
{"id":"openstack%2Fswift~752752","triplet_id":"openstack%2Fswift~master~I229f74529a718be2dfd9627a9b60f677b67578c6","project":"openstack/swift","branch":"master","topic":"reno-victoria","hashtags":[],"change_id":"I229f74529a718be2dfd9627a9b60f677b67578c6","subject":"Update master for stable/victoria","status":"MERGED","created":"2020-09-18 19:18:04.000000000","updated":"2020-10-01 23:55:57.000000000","submitted":"2020-10-01 23:54:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752752-1601596452070-c7f4b8dc","meta_rev_id":"8859cd18c3318330cf75e61d59c077fe18aedc85","_number":752752,"virtual_id_number":752752,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 23:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-01 23:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-01 23:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 23:41:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-01 23:54:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"478cc7b5aa6f5f4d341b1c794483bc5ddc74f366","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-18 19:18:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e3e121ed0073434305da10ace44831309edad35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:49:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874d92e2112c4196886f77caa57734a9 : SUCCESS in 18m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48df5480d8754de4bbbc7a335faa7bfe : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d4677686b4a5cb61287eb8e463b1755fde2a6fec","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 17:29:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"718f1cf3fb5cb2eaf46fee291a2d8a006148167c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:23:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a3543381664345a10977c0e08d91af : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2123073b5274b5fa8deb701459351d7 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0a3aa4dca85228a33d219395029faceabf5751e0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 23:41:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5375deddbb48eba1f881a6aa391b8a0c8ba2c4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:41:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"51dd2d88d1d8c0d8353df0f4dc6ea171e836f98d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:54:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca3abee48564117ac7ad9b82fe79d84 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/662b7a2b7ec140af92f57a694d91f130 : SUCCESS in 5m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ff8e5c4d3234409da2991b87167890b2 : SUCCESS in 8m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e87c73d296ea4ebda55fbc886dcc1ce6 : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe62934a541be4022e4d78c2b10f98cbdf65b6ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:54:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"809f73504bb2ea3281a2909f05d38bd6e29ef431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1ce99bf3ed41e19d852a4c11677d60 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dd49cf69243449caeb3eea3fc815770 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7e1d2b99edda4080b6e955e940b87340 : SUCCESS in 54s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c57e3eaaa371409fb9adcce9cb3d7d4d : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"179a6ffb8f5a9c829a5f7db9df3749e4b88dc139","revisions":{"2253bf306db457a46d00a655b31997321ae518d9":{"kind":"REWORK","_number":1,"created":"2020-09-18 19:18:04.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/52/752752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/752752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/752752/1"}}},"commit":{"parents":[{"commit":"60530753b37e01a303825bb1ba289d241b21f45e","subject":"Merge \"Run swift-tox-func-encryption-py37 job in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60530753b37e01a303825bb1ba289d241b21f45e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:18:02.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:18:02.000000000","tz":0},"subject":"Update master for stable/victoria","message":"Update master for stable/victoria\n\nAdd file to the reno documentation build to show release notes for\nstable/victoria.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/victoria.\n\nChange-Id: I229f74529a718be2dfd9627a9b60f677b67578c6\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2253bf306db457a46d00a655b31997321ae518d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2253bf306db457a46d00a655b31997321ae518d9"}]},"branch":"refs/heads/master"},"179a6ffb8f5a9c829a5f7db9df3749e4b88dc139":{"kind":"REWORK","_number":2,"created":"2020-10-01 17:29:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/752752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/752752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/752752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/752752/2"}}},"commit":{"parents":[{"commit":"60530753b37e01a303825bb1ba289d241b21f45e","subject":"Merge \"Run swift-tox-func-encryption-py37 job in the gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60530753b37e01a303825bb1ba289d241b21f45e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-18 19:18:02.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-01 17:29:23.000000000","tz":-420},"subject":"Update master for stable/victoria","message":"Update master for stable/victoria\n\nAdd file to the reno documentation build to show release notes for\nstable/victoria.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/victoria.\n\nChange-Id: I229f74529a718be2dfd9627a9b60f677b67578c6\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/179a6ffb8f5a9c829a5f7db9df3749e4b88dc139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/179a6ffb8f5a9c829a5f7db9df3749e4b88dc139"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
