)]}'
{"id":"openstack%2Fswift~752770","triplet_id":"openstack%2Fswift~master~I52287bb273b955dec2ab96a987a7fc75b861d4fd","project":"openstack/swift","branch":"master","topic":"mpu-ec-disconnect","hashtags":[],"change_id":"I52287bb273b955dec2ab96a987a7fc75b861d4fd","subject":"Log error processing manifest as ServerError","status":"ABANDONED","created":"2020-09-18 23:08:39.000000000","updated":"2021-03-30 23:03:23.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"445165e4da77aeeeaca74dbf7dfe894c3e058feb","_number":752770,"virtual_id_number":752770,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-01 21:53:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2020-10-01 20:48:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 21:46:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-01 21:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30c41d907830879778c61c4b4d35ebd2da3bbe9c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-18 23:08:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca841baa17f11fb3a8df90c3b99c8a4cb3e417f7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-18 23:25:52.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nCome to think of it, if we can fetch the manifest but the first segment of an SLO 503s... 409 *really* seems like the wrong call...","accounts_in_message":[],"_revision_number":1},{"id":"c43e2c6b0cae3b7611abee5f87f99fff04067a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 00:14:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78b4a80f6e8416bb2919eefc3d8d703 : TIMED_OUT in 31m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/742c1981a70d4baf9da6350e5ec48486 : SUCCESS in 1h 03m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79889af6cbc24ff6ae54eea12e918108 : SUCCESS in 48m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ef896ac99df45fe8bc2e42964f6b315 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90ca58d105484d84822367a8b24f3c32 : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/541593107d0247dfb007a88a6deffaf7 : SUCCESS in 10m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4dd89513eb7d438a909e42182948904f : SUCCESS in 10m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7741fae599234cd59cbc0c5f10e279d8 : SUCCESS in 10m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b4dada857ec54af1926ebb2153660e2c : SUCCESS in 10m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/364fa39bab374700b1013b651342b15d : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ac1ae9c7bc94c75826d707c6bab17c4 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df44437621984ae9991e320a0d12aba2 : SUCCESS in 16m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/906cad3d5be447b88db856aed50fade5 : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bf531e3869514d20b8b1f43fc3cd485b : SUCCESS in 15m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ffa57883747f4eb9b4c044951eff6d29 : SUCCESS in 15m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be718726357b4d25aa141c24be579a26 : SUCCESS in 18m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f1004787b9624df1ac9ead5d60dffc05 : FAILURE in 38m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66a96c45b355419b94238c68b8e433cc : SUCCESS in 44m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/184ee8d85ea840aca3468c342c050725 : SUCCESS in 34m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25c41a363a964571985ad64e7f3472eb : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf916fdb932f4ba2993098388a02826f : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e1c7662ed6c41499babec5638e0978d : SUCCESS in 19m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b1eb3cc33343e4b4c93db85d7765ad : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1292bf37567f99dcb7d67e3de6c6528f6e9cc52c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-21 21:46:16.000000000","message":"Patch Set 1:\n\n:-/ This isn\u0027t working for me (at least, with my test setup). I uploaded a 4k+ file broken up into 3 segments, made both the primary and segment containers public, then copied the last segment out to a third container and replaced the original with a symlink. I go to fetch through curl and get my short read:\n\nvagrant@saio:~/swift$ curl http://saio/v1/AUTH_test/c/tox.ini | wc -c\n  % Total    % Received % Xferd  Average Speed   Time    Time     Time  Current\n                                 Dload  Upload   Total   Spent    Left  Speed\n 85  4788   85  4096    0     0  58514      0 --:--:-- --:--:-- --:--:-- 59362\ncurl: (18) transfer closed with 692 bytes remaining to read\n4096\n\nBut it\u0027s still logged as a 200:\n\n vagrant@saio:~/swift$ tail /var/log/syslog\nSep 21 21:39:52 saio object-6040: 127.0.0.1 - - [21/Sep/2020:21:39:52 +0000] \"GET /sdb4/128/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002\" 200 - \"GET http://saio/v1/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002?multipart-manifest\u003dget\" \"tx9aaccfe1e700415686baa-005f691da8\" \"proxy-server 3835\" 0.0020 \"-\" 3820 0\nSep 21 21:39:52 saio object-6030: 127.0.0.1 - - [21/Sep/2020:21:39:52 +0000] \"GET /sdb7/128/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002\" 200 - \"GET http://saio/v1/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002?multipart-manifest\u003dget\" \"tx9aaccfe1e700415686baa-005f691da8\" \"proxy-server 3835\" 0.0021 \"-\" 3817 0\nSep 21 21:39:52 saio object-6040: STDERR: 127.0.0.1 - - [21/Sep/2020 21:39:52] \"GET /sdb4/128/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002?multipart-manifest\u003dget HTTP/1.1\" 200 1689 0.003061 (txn: tx9aaccfe1e700415686baa-005f691da8)\nSep 21 21:39:52 saio object-6030: STDERR: 127.0.0.1 - - [21/Sep/2020 21:39:52] \"GET /sdb7/128/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002?multipart-manifest\u003dget HTTP/1.1\" 200 1689 0.003038 (txn: tx9aaccfe1e700415686baa-005f691da8)\nSep 21 21:39:52 saio proxy-server: - - 21/Sep/2020/21/39/52 GET /v1/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002%3Fmultipart-manifest%3Dget HTTP/1.0 200 - curl/7.58.0%20StaticWeb%20%20SLO%20MultipartGET - - - - tx9aaccfe1e700415686baa-005f691da8 - 0.0116 SLO - 1600724392.908993483 1600724392.920599222 0\nSep 21 21:39:52 saio proxy-server: - - 21/Sep/2020/21/39/52 GET /v1/AUTH_test/c%252Bsegments/tox.ini/slo/1600465773.917852/4788/2048/00000002%3Fmultipart-manifest%3Dget HTTP/1.0 401 - curl/7.58.0%20StaticWeb%20%20SLO%20MultipartGET - - 131 - tx9aaccfe1e700415686baa-005f691da8 - 0.0019 SYM - 1600724392.922414780 1600724392.924342871 0\nSep 21 21:39:52 saio proxy-server: While processing manifest /v1/AUTH_test/c/tox.ini, got 401 (\u003chtml\u003e\u003ch1\u003eUnauthorized\u003c/h1\u003e\u003cp\u003eThis server could not verif...) while retrieving /v1/AUTH_test/c_segments/tox.ini/slo/1600465773.917852/4788/2048/00000002 (txn: tx9aaccfe1e700415686baa-005f691da8)\nSep 21 21:39:52 saio proxy-server: 127.0.0.1 127.0.0.1 21/Sep/2020/21/39/52 GET /v1/AUTH_test/c/tox.ini HTTP/1.0 200 - curl/7.58.0 - - 4096 - tx9aaccfe1e700415686baa-005f691da8 - 0.0616 - - 1600724392.864268303 1600724392.925868034 -\nSep 21 21:39:52 saio proxy-server: STDERR: Traceback (most recent call last):#012  File \"/usr/local/lib/python3.6/dist-packages/eventlet/wsgi.py\", line 589, in handle_one_response#012    for data in result:#012  File \"/vagrant/swift/swift/common/middleware/catch_errors.py\", line 52, in enforce_byte_count#012    bytes_left,))#012swift.common.middleware.catch_errors.BadResponseLength: Expected another 692 bytes (txn: tx9aaccfe1e700415686baa-005f691da8)\nSep 21 21:39:52 saio proxy-server: STDERR: 127.0.0.1 - - [21/Sep/2020 21:39:52] \"GET /v1/AUTH_test/c/tox.ini HTTP/1.1\" 200 4583 0.063424 (txn: tx9aaccfe1e700415686baa-005f691da8)\n\nI\u0027ll keep digging, but thought I ought to give you a heads-up.","accounts_in_message":[],"_revision_number":1},{"id":"bc9dc2925cc5ed2ae9a56c27d42fbc86460b06ff","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-22 21:58:05.000000000","message":"Patch Set 1:\n\nsomehow the difference seems to be if you have an auth token or not\n\ni deleted the last segment of an MPU and when I try to download I see:\n\nSep 22 21:51:36 saio proxy-server: While processing manifest /v1/AUTH_test/test/big, got 404 (\u003chtml\u003e\u003ch1\u003eNot Found\u003c/h1\u003e\u003cp\u003eThe resource could not be foun...) while retrieving /v1/AUTH_test/test_segments/big/slo/1600453228.589863/10485760/1000000/00000010 (txn: tx49106bd258fb47db817f5-005f6a71e8)\n\nBut when I don\u0027t use a token (i did .r:* on both test \u0026 test_segments) I get a 200:\n\nSep 22 21:50:57 saio proxy-server: 127.0.0.1 127.0.0.1 22/Sep/2020/21/50/57 GET /v1/AUTH_test/test/big HTTP/1.0 200 - curl/7.58.0 - - 10000000 - tx840798b25b41409ea88da-005f6a71c1 - 0.1994 - - 1600811457.672647953 1600811457.872026920 -\n\nand if i use a token i get a 503:\n\nSep 22 21:51:36 saio proxy-server: 127.0.0.1 127.0.0.1 22/Sep/2020/21/51/36 GET /v1/AUTH_test/test/big HTTP/1.0 503 - curl/7.58.0 AUTH_tkf03a308ff... - 10000000 - tx49106bd258fb47db817f5-005f6a71e8 - 0.1971 - - 1600811496.125458002 1600811496.322581053 0","accounts_in_message":[],"_revision_number":1},{"id":"83591f97d6511d3406835ee13c1f500be9a83608","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-25 13:40:59.000000000","message":"Patch Set 1: Code-Review-1\n\ntim tracked down the issue with the status code on un-authed requests to be some old code in static web","accounts_in_message":[],"_revision_number":1},{"id":"88cff6a8685261a7b65ab2cf6e0199e7e7e65e7b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-29 21:19:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a49c9a1e561008bd5634e7240261b01d653454fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:31:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3436911e25f4397bdd6395c91262ee9 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a8070c8384044ceeac3f8a3cf1415f92 : SUCCESS in 1h 05m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a62bdf8a6bd647f381469cd4613f5328 : FAILURE in 44m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/876f0354b14e47f08b5efa886520cf40 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4eeb463a22ef4fe480432d6b8ef35f91 : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/afd0fb6404c14d759ad1bb9bbb9427ad : FAILURE in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a271ab49d946479f8310f9b47c8002 : FAILURE in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e3ec99cce9647ef92fbe8bb2b1f893f : FAILURE in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/612af50a5fa94c8bad42852bca33d46d : FAILURE in 12m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5db1dd9407cf40448885d168d45d8453 : FAILURE in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dcf239ae04e24514a1800ca91c03399c : FAILURE in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0fe51f1da6ad485881cb34e0296aebc6 : FAILURE in 19m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff3a76819d544bafb769aeca07fb553f : FAILURE in 16m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/da6700de232b4b8aa3de22ce7b648833 : FAILURE in 19m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ff1a9eaea2914fe3966f6f6c54f3501b : FAILURE in 18m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac2f4cca86404ceebb99a2f12c3444c1 : SUCCESS in 23m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1a963e6d77404675919cf16362a97466 : SUCCESS in 42m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d75e357811445e281e6b7ad2f92a437 : FAILURE in 33m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/721bbcc7a4814df396ad460e0604f949 : FAILURE in 30m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0e2b1d1d57940f7a79c4001b073a184 : FAILURE in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbeb1f3d0f14fd4898410fc06141769 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/402d21a800c04ce09103b57eb126c171 : SUCCESS in 21m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a31bbfde6801497b8320c27a58edc65e : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b3584071fea5a221e65ce94f6b29bb8e67a2e296","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-30 17:47:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f0c0afef05742b9cf549a377c8deace2c57281c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:01:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d93fac279e4eb3b972fe968aa8d9d5 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/424aaea386b546049870684b5002850b : SUCCESS in 59m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29e118fa4f3f41598b9b5bb6027bd805 : FAILURE in 27m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5a3f947457e40428a9d01a837f90439 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46858974830a4cc6aeb3efe77f45b260 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58d28c27d9a048d5b6e65ec88fe0ebe1 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0c12bdd482444829baf3cbf16c598be : FAILURE in 15m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d895264204a5423ca6b9a7d520ced481 : FAILURE in 13m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d84d2f17f32044daa66002b42cd85393 : FAILURE in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/043487e9505e42d4a9a3c5093c023999 : FAILURE in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/37b661b93f75465ab09f41cb1b8de278 : FAILURE in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f422a25338f741d996431729e5a990c0 : FAILURE in 17m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc83540b890f43899d138c216ca73d23 : FAILURE in 17m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/47ab779d35c24cfa81154503cb5f43cd : FAILURE in 16m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e73444444fb840a187000b87562e4795 : FAILURE in 20m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3548d3d7135b4f86bb569705b49e0598 : SUCCESS in 19m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/80dd0b72951841dea915632dd45dbf5e : SUCCESS in 40m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1303a581f24e48e08bc065b18d5451a7 : FAILURE in 30m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b17e6657a8a240349d4e9873fd7127c6 : FAILURE in 29m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9d078cbef524ccc97d219e368e29708 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b140dec1e1f47d0a0d984a08cd9e691 : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa813f36ee2e4b83af86220397554841 : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd74fee1cb364b17851441d705bdee15 : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":3},{"id":"50f10ab3a1e4847f5aca4d9a3ae8efaebc2960c9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 14:18:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"fa052eef21c6c01f83fc284e4aea828d44b33942","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 14:20:13.000000000","message":"Patch Set 4: -Code-Review\n\nanyway, I\u0027m actually thinking about moving the catch errors byte counting exception catching iteration wrapper into proxy loggings iteration wrapper and working on https://bugs.launchpad.net/swift/+bug/1896518","accounts_in_message":[],"_revision_number":4},{"id":"718d3b08f469302ca0a6c4d4b7c90905b14c5ac7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 20:48:16.000000000","message":"Patch Set 4:\n\nmay not be needed with https://review.opendev.org/#/c/755639/ (if you add a byte tracking mw, like a second catch errors, before your proxy logging)","accounts_in_message":[],"_revision_number":4},{"id":"85db2c007243267ecfa7cd56eecf6d61999883bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:53:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173641a79b1f4577b3376a3107daf411 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/908783ba7bcf4590bc2c3122d1f9a212 : SUCCESS in 57m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f364b2b11064423b4dc8899426776fd : FAILURE in 45m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9db28e898af4e90b074430e2759b7a5 : SUCCESS in 11m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1a69f42a66547afad201e35f0e3b52a : SUCCESS in 9m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c4543d2d84f46899f89a7741e3e94bc : SUCCESS in 13m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bcb3db3a994491f868c0dd236ac09f8 : SUCCESS in 21m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c1f368c106dd45308ed0de91d49bcecd : SUCCESS in 16m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/540a9859b5324ea1b86acd4c1170e38f : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa92e99114d4425388a4ba2b01ae9fdf : FAILURE in 3m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b8924fd609db4aa88b7abb2390e23ccd : FAILURE in 4m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65f0c9509d9c49aaa921c0db59c4413f : FAILURE in 18m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f05f767e7ae14e34806650ff5dcb3dce : FAILURE in 26m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f6b62ed08544ed8af32bc00ce878fdb : FAILURE in 19m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6ed14d9e870b41a1b31116d0448fbec1 : FAILURE in 16m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7939e97723e941d3ae40664ae1fa3672 : SUCCESS in 19m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/159025bcdf294899891a14ef8526c966 : SUCCESS in 41m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51a6339bcd234a31b7b8e1ff93a45e03 : FAILURE in 29m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cc57358b082464b80291bdbb35df844 : FAILURE in 37m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bd3f28abfde4f3d8aab7ea41c18706e : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2a0c4ffae74763b469d403852729b6 : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcb2c8c881514b71848915a1d608fe9a : SUCCESS in 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7705e057cadc4d1a80299c40d6975025 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":4},{"id":"445165e4da77aeeeaca74dbf7dfe894c3e058feb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 23:03:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3e1243134b19d1b70c8f10bb7363732f47a8f23b","revisions":{"d8abf2be4f4d6538bb03d5c85589b48e57b90c87":{"kind":"REWORK","_number":1,"created":"2020-09-18 23:08:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/752770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/752770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/752770/1"}}},"commit":{"parents":[{"commit":"d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2","subject":"Authors/ChangeLog for 2.26.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c6ab764e4c0e737b99f19b2495f1b9aab8c0d2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-18 20:58:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-18 23:08:22.000000000","tz":-300},"subject":"Log error processing manifest as ServerError","message":"Log error processing manifest as ServerError\n\nChange-Id: I52287bb273b955dec2ab96a987a7fc75b861d4fd\nCloses-Bug: #1895691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8abf2be4f4d6538bb03d5c85589b48e57b90c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8abf2be4f4d6538bb03d5c85589b48e57b90c87"}]},"branch":"refs/heads/master"},"f79c188decea74fa8470831b1c26ad58217ef6c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-29 21:19:56.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/752770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/752770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/752770/2"}}},"commit":{"parents":[{"commit":"dddd97ba051bc63cbdb482a07ec877942d825984","subject":"Add staticweb to default pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dddd97ba051bc63cbdb482a07ec877942d825984"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-18 20:58:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-29 21:18:41.000000000","tz":-300},"subject":"Log error processing manifest as ServerError","message":"Log error processing manifest as ServerError\n\nChange-Id: I52287bb273b955dec2ab96a987a7fc75b861d4fd\nCloses-Bug: #1895691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f79c188decea74fa8470831b1c26ad58217ef6c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f79c188decea74fa8470831b1c26ad58217ef6c5"}]},"branch":"refs/heads/master"},"f56e82d994f8144e8a8d0f762f7cd78900c32e4c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-30 17:47:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/752770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/752770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/752770/3"}}},"commit":{"parents":[{"commit":"4ff14b7b55fc8a40dfb0a5536a2d035972b32c06","subject":"Add staticweb to default pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff14b7b55fc8a40dfb0a5536a2d035972b32c06"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-18 20:58:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-30 17:47:51.000000000","tz":0},"subject":"Log error processing manifest as ServerError","message":"Log error processing manifest as ServerError\n\nChange-Id: I52287bb273b955dec2ab96a987a7fc75b861d4fd\nCloses-Bug: #1895691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f56e82d994f8144e8a8d0f762f7cd78900c32e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f56e82d994f8144e8a8d0f762f7cd78900c32e4c"}]},"branch":"refs/heads/master"},"3e1243134b19d1b70c8f10bb7363732f47a8f23b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-10-01 14:18:57.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/752770/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/752770/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/752770/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/752770/4"}}},"commit":{"parents":[{"commit":"f820dc8dc7ba3af40da461c97ea5801bd46cf6fd","subject":"Add staticweb to default pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f820dc8dc7ba3af40da461c97ea5801bd46cf6fd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-18 20:58:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-10-01 14:18:57.000000000","tz":0},"subject":"Log error processing manifest as ServerError","message":"Log error processing manifest as ServerError\n\nChange-Id: I52287bb273b955dec2ab96a987a7fc75b861d4fd\nCloses-Bug: #1895691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1243134b19d1b70c8f10bb7363732f47a8f23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1243134b19d1b70c8f10bb7363732f47a8f23b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
