)]}'
{"id":"openstack%2Fswift~754412","triplet_id":"openstack%2Fswift~master~Iec579ed874ef097c659dc80fff1ba326b6da05e9","project":"openstack/swift","branch":"master","topic":"slow-connect-ec","hashtags":[],"change_id":"Iec579ed874ef097c659dc80fff1ba326b6da05e9","subject":"Use bigger GreenPool for concurrent EC","status":"MERGED","created":"2020-09-25 14:48:55.000000000","updated":"2020-10-03 06:57:15.000000000","submitted":"2020-10-03 06:54:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754412-1601708056752-5cec80e2","meta_rev_id":"b1badef888a24ba6327bdfc77b7c4367a9e58619","_number":754412,"virtual_id_number":754412,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-03 06:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-10-01 21:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-02 04:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-02 04:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 04:25:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-03 06:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94f479bbbb41ef3749a0f796c7e4dbd925a9528b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-25 14:48:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da00b8a7fd5d5e9980d27a0b10eba32d75119ad9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-25 15:41:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nI need to think through it a little more, but I think it makes sense...\n\nShould we just get rid of the pool, or make it unbounded? It seems like we mainly want it so we have a logical grouping for this batch of requests -- but why do we want to cap the number of concurrent requests?","accounts_in_message":[],"_revision_number":1},{"id":"154902a436527dc95d756b2ea6a717d1563871b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 20:30:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7203da1f94949358c29b67d6c9ea9bb : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/92b7e3a858dd42ac9e1451c5aff47bb4 : SUCCESS in 1h 03m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55dfdfbd7627408b982bc24857eac37e : SUCCESS in 50m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/851ceb3d39a240fd954ce25659f57ed8 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ace1fc2a89df4885ba33530199968a92 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d262907e19c347deb3abe74c5f3af1d8 : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9aeddb7dd6fc45b296767e134478a33a : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8440912961b46b08cc280c9cdd06057 : SUCCESS in 15m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc2c1f12d4174cb78d1a6224a4d63f71 : SUCCESS in 12m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2a8c71e1b9c4659b40ee5cf90e6061c : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78f676e25bff477c9d3137d5a26ce52f : SUCCESS in 19m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d9365829f2445b2bb7bd0493eea0f5b : SUCCESS in 18m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/593500b26e464e389cf22112d6ccb9b9 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9c9a2274e6514f1db99810b9bc75f322 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8db6c2d384784587bf0612ce4190b60e : SUCCESS in 22m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fa8c52a877b4a74b0b0d224f0fc8ed0 : SUCCESS in 25m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/70ac1ff2116c4530b31f8df891dcfbfc : SUCCESS in 49m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c623650422874f6794f16762dfe4a396 : SUCCESS in 40m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41d800f3b36a40ab8b86dc94f85d0331 : SUCCESS in 37m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490bbc157dc0490782bafb1708929e7a : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47acae863daa4bceaa98d4c9bdac0c70 : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13ece193cd3f4f54a6371424271cf8f1 : SUCCESS in 23m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81f083445ae74ac7825cb6648c439ee9 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"226fc6135204c6d79d604f85e2c3264d96952f6f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 21:08:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\n\u003e why do we want to cap the number of concurrent requests?\n\nI think it\u0027s just a sanity check against bugs, I think it\u0027d be weird to have it unbounded\n\nBut this does bring it much closer inline with replicated:\n\nhttps://github.com/openstack/swift/blob/master/swift/proxy/controllers/base.py#L1373","accounts_in_message":[],"_revision_number":1},{"id":"18dc46244dd2a7fbb80280aeda84bb7487616d51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 04:25:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\n\u003e But this does bring it much closer inline with replicated\n\nIDK, is it? I see three axes to play with:\n\n* replicated vs ec\n* GET vs HEAD\n* concurrent_gets on vs off\n\nTrying to remember how to do a Karnaugh map, I get maximum concurrent in-flight requests like\n\n           | concurrent |   not\n ----------+------------+------\n repl HEAD |   replicas |     1\n ----------+------------+------\n  repl GET |   replicas |     1\n ----------+------------+------\n    ec GET |    nunique | ndata (*)\n ----------+------------+------\n   ec HEAD |      ndata |     1\n\n(*) limited not by the pool size but the lack of a feeder thread\n\nI\u0027m struggling to see much symmetry -- I feel like bringing it up to replica count across the board seems like a more \"obvious\" fit.\n\nStill, this is a step in the right direction!","accounts_in_message":[],"_revision_number":1},{"id":"401faf571ce0b3e9b267ce9b3038efbacb1aa94f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 04:26:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b238933344203c94593704497cc7b76ef312bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 05:59:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd2de26cb534d848651d7f5bfaeadd7 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/ca4f5accf35c464ca2879524731a8f0b : SUCCESS in 1h 09m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/193cd63c31c54566a7f8c65cee3d9396 : SUCCESS in 40m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/60779de44def439fb761794b5c6f950b : SUCCESS in 8m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c50e657fe6da48169ed27779865ae9ea : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49ce84d1343048548908c8435c7a6c57 : SUCCESS in 13m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9730366c34f0421ab0951690fcea0667 : SUCCESS in 12m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d1f93de1d96438f845fb63505a22da9 : SUCCESS in 12m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c177924f4f084954bc69c446052d8557 : SUCCESS in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/65edc569caae45889264381c217eb12a : FAILURE in 3m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c7f0643a1c6405f8e7287e5c5fbcf1e : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7d4659d95904f30bd2aee99d505d7a2 : FAILURE in 5m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/82bf90a8eea5498291bde9f6a51ec511 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a6138f83fbb4715b011b8d86ec935e2 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/09a3efd86c5c4ceb8bf83882ce052bd4 : SUCCESS in 17m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/636b56ac499e4574b181a5c8d5efdb61 : TIMED_OUT in 1h 31m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ad9c601577d4b31b61e70ea47f3c105 : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f1a6f4936514103b8dd4427a4bb9de1 : FAILURE in 13m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd2cf600920f46d4b7a9655dda63b88c : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501d06c341f14947bc8b50514129fca8 : FAILURE in 6m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28a958891b59434c8d51b2a277b2466b : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2670625ec848260cd14e7a6c818d783ecd52188f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-02 14:59:39.000000000","message":"Patch Set 1:\n\nYou might have convinced me that multi-region EC could possibly want more than n_unique_frags pending... I mean in theory one could send out n_replica requests to all the nodes and just rebuild from the first unique n_data that come back\n\ni honestly have tested (or even really considered) concurrent gets with frag duplication at all","accounts_in_message":[],"_revision_number":1},{"id":"cbcf0db60cb84ae6c35c62bef8bf5ae45ff9cf30","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 15:39:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22397d62b71d1f550784a0b78899671563324dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 20:14:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b356d5dbde452e87465b9fc83f9cd4 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c6dd8d835f6640529f1276c7fc8681ff : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4335cdc979434a7f9eb473241d457565 : SUCCESS in 57m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9806693bd174890a8afbb191542e59c : SUCCESS in 11m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05c4fbefd05e4cb3950b5342ef34299d : SUCCESS in 12m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3ed9000333a40d7ba85583028029994 : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c782df726f44460a97319e67a484521a : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6b7bbfeaa99f4c34af12908b857d397d : SUCCESS in 12m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d961fd66fe2547f7bc3f838753cac125 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40f9602538fa40499aee85d95cb18a1e : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/639a0ce7d660471693de03d2e9049791 : SUCCESS in 22m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf680a7d3a13407d982f388f716a7610 : FAILURE in 18m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/285b28536a9c413b8b530de37e139518 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/26a3f8884468406bb300c3f80a26520c : SUCCESS in 27m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bdc3a87e0c4544a9bb6b242ecdda1464 : SUCCESS in 21m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02946416ce5a442aaffcb9371f1184e0 : SUCCESS in 23m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/958625daee4546b78da99999b48f938d : SUCCESS in 43m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/593170cf51614c6080cec5ee1c816dda : SUCCESS in 43m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4eb09334b1745ef8d3c7c813f318956 : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8545ddc769664f8b8a8806c442aa5546 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d403a95077b4c8096d5319d00b60ec2 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d83d225944904b47b0c1375af834228f : SUCCESS in 21m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b10f063edc24203ba17e0e05e0d521d : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6d6ff474e4ec88e28246dc2f9e593780e7d7ea63","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 23:33:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65a2d4d61ce88a4aef1e958ec2b73b97e4ca0da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e024139cb154af48a48f7b546be68d8 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/e3e3e8963ba8415c896656cff742f6b2 : SUCCESS in 1h 03m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33c568b1eb764044b9a6787f58ef7d53 : SUCCESS in 46m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94a890309c634aca9139fb237a57a90e : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3fba5c5d310477e9c324af9e98e0a22 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66d4bd43811b4244a5967812e7d0fe93 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0516252fbc6c43bd9c24490cea72becb : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0a02e1dad3e4eb38d7c232c139c6c24 : SUCCESS in 12m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f294295130b4184b8ee1643ac69f011 : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de62b3ba02e546dea302b574e9adc443 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1c6c64bfb05f4fb6908ea6fb2b55ed40 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1bdfc5fcaa59401cabbd0ca5e4992b2e : SUCCESS in 16m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b4b427852c964b6d815bcd49cd3523db : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a41218769c640fcaa69b2c1f382b6aa : SUCCESS in 17m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ec4e626e44d54c09a6f6dc4b30457831 : SUCCESS in 17m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb411042e75543e0acab8f5593724071 : SUCCESS in 25m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1cbf6781af294994a480aac33e41b715 : SUCCESS in 47m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/debc4ddc732240419485b8c5441ec0be : SUCCESS in 46m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/258f237259d849a68792316be577a059 : SUCCESS in 40m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5e0396913d845cfb8e01206818f1157 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340da2d959c74dc7bed9c1580ecd58e0 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb0052af205f4ac1b9015b5b4be1ad7a : SUCCESS in 21m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be6cdef7a9bc4ee48e7fc66ed7a08782 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0926d325aaf9026381effcf88b6cd990fb1a8178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20df0264c85be332dad5d47e35d4cc8f22a951af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 01:50:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e81ed0e944f4d1b98d1a6fc84aca5d8 : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/bfd1103ec3ae416a8867c5e35f8ff550 : SUCCESS in 43m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f2be60633f24d96bcd7f91603295c2b : FAILURE in 52m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4bac2d5169134c0bbf03159740bf138a : SUCCESS in 9m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7bad13de94094f7baf77f06519159ffe : SUCCESS in 10m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f2519be553fa46d19189ffab4c3abe05 : SUCCESS in 12m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/37e59e91a82a4155b4615a52a6ed0c28 : SUCCESS in 12m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d406648986174cfebf05e380c74afc8d : SUCCESS in 24m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa4f02effca84ce6a52932faf711912d : SUCCESS in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95ef596920554db69260fffe747182e3 : FAILURE in 24m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/768e0aa5881f40bba2883ee074ea0f78 : SUCCESS in 24m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e5debf101ab649778a4cfab4fa15f157 : SUCCESS in 16m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6c5aed3ab4af4a4ba85b00141937c70a : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a8286f9bc87b45b381a24d19d9d617ff : SUCCESS in 18m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2ccd9cbfea81405080682cfd0d371b29 : SUCCESS in 22m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0a234b9d7f84838adc1f3ae42ee0a17 : SUCCESS in 40m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dbb5565ccbe4f888e17207a96e0ff52 : SUCCESS in 41m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae41e42f93e04d5b9e0e736813a7ec7f : SUCCESS in 40m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7edbddf47cf4a0dba5231df35869a47 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c84d5eef8b14aa29c6dbbc956dab59a : SUCCESS in 6m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b86ddfc81647cf86549ddbf901e0ac : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c6991574360bc1d80001a3ca6c09b72b2ddba2a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-03 04:17:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5f6e70517452bb358315192fc25e3392cfaa001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 05:33:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb9011c5fda4bf5b1c24de46490d345 : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2b7424fa621846f892891849b040d7f3 : SUCCESS in 1h 04m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe903d0c05dc43dfa7d62e6da2d14c74 : SUCCESS in 50m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c0458b74dcd4114a1eb9773118e7710 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6dc3c415813f4989be79911d65183c6a : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1b0d51309ea4f8e9f3cc5f26451a0c2 : SUCCESS in 13m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6cd05b6f7e434b0eb4e413146ce7dab9 : SUCCESS in 12m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1aa2a78fca7046e8a84f9d716738afc5 : SUCCESS in 12m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30d9bf4805f645fe9199bf81c588f2a6 : SUCCESS in 12m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90b92f0dddd24f27896e363d2f81a67a : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d511189253847c7b414c91e45312dd2 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97cdf0e6451a4fc5aa86a9965ff4c6f1 : SUCCESS in 16m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46e3d9e6ee5848fc93ac74f8afe8256f : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/394928542f614e6396a031958c09dd8e : SUCCESS in 17m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/df590df6ed2f46799d4d0a81f39cb6ec : SUCCESS in 17m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c19053618cb4c9f91ef2465240d48dc : SUCCESS in 19m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c63afc72f3f4a96bf5eb52c061a1bb9 : SUCCESS in 44m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29237c5afadc4717956936b106841add : SUCCESS in 46m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48da713054314ae9bb07ba4222f8d70c : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f85811192a94b0e8cb9f242e99f4cdc : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4d6164de4c4acfadd112f590f4c970 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fbdfdcd94c3f4f1a8843a110019ea5f2 : SUCCESS in 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73666558036747d3b8721d8d86f3d065 : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9597eec1d25edf6d265bd976c300537c5cd2d072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 05:34:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c869856b121a2e006ab06b199425329071586ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 06:54:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf3523f1799443181033168de7e5f1a : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/744187d099284f9390bec013664ff93a : SUCCESS in 1h 14m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c70ba9c2d8042e38d290dd5674ab4a6 : SUCCESS in 53m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d3a1b4b6f2f841d3b0faa140b288e653 : SUCCESS in 9m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6c1e7ff680cd4bbf949249d18da5e9b9 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e536d5173a9643ebb019811bfe881d22 : SUCCESS in 11m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7b1f74e84e1418f98933eb157b5d6ea : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67e9cf6e0f394facbafffb4150d31ec6 : SUCCESS in 12m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5bb9e7f110094dc6bde1b0a1a1d125fe : SUCCESS in 13m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9bde2beca45f4a7a8328288ec31874a8 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0ddd5ab16fc4f48913ef40bf3d225ad : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/905afcde418647edb9a040030cb9b1af : SUCCESS in 15m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0331efbe4ed24468905ba5e0fd847966 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ee24cb0a01a945ea934f675c6d834916 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/603e6b1b37ea4f4ca34e2d52c6ae0935 : SUCCESS in 18m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4bf85b4aa80649d483f7cb07dc2391fd : SUCCESS in 47m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc8e338623da4d60b3b8ceb1b8ee2484 : SUCCESS in 43m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58b3d13772254fdf902b4422fb083183 : SUCCESS in 53m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5c456dfafb1455eb15725bc83dadf16 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01675a8d76434e85b61eb63bd9dcb4dc : SUCCESS in 6m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0af0c2a6fd7f4d2187deae2ae875941b : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"aec164bbd6f800fc20df636f86754066b31f2a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 06:54:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b134fec5a1b6df90b7908ebf323b17a19746f27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 06:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9112070991242eba13e2d0b17153e06 : SUCCESS in 2m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb3880a3ebdc4cc4a07e111d5da60a56 : SUCCESS in 1m 13s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7cf7f08e80e649eca4a88403cd3b6905 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f95e1becec55f422bc27cfbfcfef43854c76be7","revisions":{"5f95e1becec55f422bc27cfbfcfef43854c76be7":{"kind":"REWORK","_number":1,"created":"2020-09-25 14:48:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/754412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/754412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/754412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/754412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/754412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/754412/1"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-25 14:15:55.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-25 14:47:40.000000000","tz":-300},"subject":"Use bigger GreenPool for concurrent EC","message":"Use bigger GreenPool for concurrent EC\n\nWe\u0027re getting some blockage trying to feed backup requests in waterfall\nEC because the pool_size was limited to the initial batch of requests.\nThis was (un?)fortunately working out in practice because there were\nlots of initial primary fragment requests and some would inevitably be\nquick enough to make room for the pending feeder requests.  But when\nenough of the initial requests were slow (network issue at the proxy?)\nwe wouldn\u0027t have the expected number of pending backup requests\nin-flight.  Since concurrent EC should never make extra requests to\nnon-primaries (at least not until an existing primary request\ncompletes) ec_n_unique_fragments makes a reasonable cap for the pool.\n\nDrive-bys:\n\n * Don\u0027t make concurrent_ec_extra_requests unless you have enabled\n   concurrent_gets.\n * Improved mock_http_connect extra requests tracking formatting\n * FakeStatus __repr__\u0027s w/ status code in AssertionErrors\n\nChange-Id: Iec579ed874ef097c659dc80fff1ba326b6da05e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f95e1becec55f422bc27cfbfcfef43854c76be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f95e1becec55f422bc27cfbfcfef43854c76be7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
