)]}'
{"id":"openstack%2Fswift~754814","triplet_id":"openstack%2Fswift~master~If238725bbec4fc3f6c8d000599c735a7c4972f7d","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"If238725bbec4fc3f6c8d000599c735a7c4972f7d","subject":"Fail short reads in SegmentedIterable","status":"MERGED","created":"2020-09-28 16:35:22.000000000","updated":"2020-10-03 02:18:54.000000000","submitted":"2020-10-03 02:17:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754814-1601691466597-c1ef629e","meta_rev_id":"1e0d080ad757523bb9026107253836e7f29de7d9","_number":754814,"virtual_id_number":754814,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-03 02:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-28 20:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2020-10-01 20:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 17:23:23.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-10-01 20:39:38.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-10-03 02:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13ec3cdb14f0cf9e4cfc2109a613ca1d64f2c66f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 16:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b4fe0b49bf9bf152527b3eabddca8597bf00e33","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-28 16:52:49.000000000","message":"Patch Set 1: Code-Review+1\n\nyup, just needs a test","accounts_in_message":[],"_revision_number":1},{"id":"12941695d9387419ebd20086a33fc42c0c5be244","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-09-28 17:23:23.000000000","message":"Patch Set 1:\n\nYou\u0027re just doing it for the purpose of better diagnostics, right? I\u0027m sure the MD5 of a truncated data ought to differ anyway.","accounts_in_message":[],"_revision_number":1},{"id":"a57e43c010e8bdf81847479fba75bfbc5f2ceef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:43:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe698f4e4864c998185aa4b61380f72 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f5d980948806434a938bcb591b2fc11a : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33129396627d47dfae9cafd75c55f0a7 : SUCCESS in 47m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f9fb60e78264107b07cf12bf68fda61 : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/55900a8c7b0a4b4c9d930e53b8c55d8d : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c42cee2050e045fabc41a0a7dbe0d203 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed63ad646e84ea2a37450fb22054efd : SUCCESS in 15m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b615a55f82b549e587598bc3fdddc73c : SUCCESS in 12m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b15bcc03f3944b51bca4283e79eaa068 : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c013127b825944b58a7b981a6eaa51fd : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4eb31753a6d64d679baafa04ad07474d : SUCCESS in 59m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/717cd7babe4a47279f66d3aa0bfa74e2 : SUCCESS in 19m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a5c2db68d2774bc8b2c4c4cc2c643441 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d15aab5c232449129312bc2dc3b222d4 : SUCCESS in 22m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c8529b4dbf054ff1b57e1c8b30e2fdcd : TIMED_OUT in 1h 01m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f109132969ab4696aefbcc5a1f4673cd : SUCCESS in 24m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce6b8637c8fa411cb555a17b82fef1b7 : TIMED_OUT in 1h 31m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d9f08c363634625bf847e475f03a815 : SUCCESS in 42m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d446b3797b664d64b3f30621a2d25821 : SUCCESS in 46m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/015052764ffe4981bce88ddcfc6cebe5 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26eb871b6714993b294d07bb838a65f : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3ea24c492d34dda98d293561069ef8d : SUCCESS in 22m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ddea53bdd040d39f175667f1ea128b : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4fb3dffee5238b854c57d4379a07919246eff4de","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:11:22.000000000","message":"Patch Set 1:\n\n\u003e You\u0027re just doing it for the purpose of better diagnostics, right? I\u0027m sure the MD5 of a truncated data ought to differ anyway.\n\nYep -- just want to make sure I don\u0027t waste time thinking about bit-flips when there\u0027s a much more obvious explanation.","accounts_in_message":[],"_revision_number":1},{"id":"fba177c5d61808c6ffce604003b585a0c4f40113","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:35:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ae4d06823fdbf7d3767d848a7cb4b0ab1d8aafe","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-28 20:44:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e8cbef631a28b0901f85ecb8c08459ce5b1ebf9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-28 20:53:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"78ad928ebb6a127d8641a86ef9cb50c60b8cd296","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-28 20:54:42.000000000","message":"Patch Set 4: Code-Review+2\n\nI\u0027m pretty sure this is great","accounts_in_message":[],"_revision_number":4},{"id":"0240426d3ea5214de4c3cc0cd9ad8e0393e9d301","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:56:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"56f6b85cb5308806303fc71eac8296d59d168249","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:57:11.000000000","message":"Patch Set 5: Code-Review+2\n\nBest of all worlds now ;-)","accounts_in_message":[],"_revision_number":5},{"id":"c8410c61a203524156a7a2343211f32fb20a3c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 22:05:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84a78c27d06340fa9bf449b4c4d104ce : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/4ca40941d533488a9bd9d538f61f60a6 : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/576e00ec87ee4ee4b30a6ecba6efc494 : SUCCESS in 38m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f170af464e64c0886c8343bf8de8ac9 : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/172f9f61ca0a428d94d458ed91744658 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/615a4b68334148ca8ddf3036c8d1412d : SUCCESS in 12m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25434834fa2143ecb78f4fa6772ae03c : SUCCESS in 12m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/855997b7bda84e53ae6eb88020b75a90 : SUCCESS in 12m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/112c1cbad82a46cfa1606d511d86ed59 : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a586a79f57841be8e66c52d89c3e32e : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f99eacc548f4f5580150fb927bb7e4d : SUCCESS in 15m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8201bab3b164ef183cf5fe0ab110d73 : SUCCESS in 18m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bcaf56caaed544c79ed1c12d47806a48 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e791ccdca99a402b95067c466ae8beb5 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2b209fd79fe146c38b6f34a8bae26c96 : SUCCESS in 16m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c11c689a86f6456a987183360428d4f3 : SUCCESS in 25m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f0c9b499d794a30b5f8900fd429a509 : SUCCESS in 40m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b38d5878c83d4cf98c373d7f5b0f4bef : SUCCESS in 47m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/625fcec977194e248bba610de9b38063 : SUCCESS in 36m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c16200f4b7544536a39761e81eead666 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b77a5d0da5a447318d17d8e56995b8d4 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8917a68d23014ed38d7e9cc8be75e1af : SUCCESS in 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a47cab1e26a447f8ba78497afdf0467b : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":5},{"id":"0f9ef775d0f2a047a5d7ec865f316f48c55850af","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 20:39:38.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"74c82e7764729b45b1de39dfb903206496b55e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:39:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"824698c60b889c381d208b65d407c5f646db08ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:06:38.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa992e9761794f0ab50ab0168a231bb0 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/a74d237c946e4050b98be3df79a8ba40 : SUCCESS in 1h 05m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/834fa4b6cc9b471d81bab02081da7e1b : SUCCESS in 44m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c1ec9bf47c584e40ba50d12b77b200bc : SUCCESS in 9m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/be1a2a199f364027a51e86c98b1f77ad : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8357882878bd4a9a9b578c32d596cc58 : SUCCESS in 12m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/368b64b4408941adb7447181c75f753f : FAILURE in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15993de059924ce5b513db4d932cb2da : SUCCESS in 12m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f88e62287555428ebc98381cca2007c6 : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bcb5d2d98694774a77cdb648bcfad23 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15caea4457364e0281990ffa0850047f : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98e1f310c7464301b169f1e04d6061d8 : SUCCESS in 19m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0a732da82aae4c16aab81dbf3b210b5f : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a520b745ccd140bf9de9eb1fa41f0872 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1b5e9437fb094de1863824966f4275d3 : SUCCESS in 18m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5fda873ff52a4e3c8ea306d739be8bf6 : SUCCESS in 39m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23efb4e7bc3544f08ea15babc3cae18f : SUCCESS in 37m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80a3385497ba4f80a5e8cb245911c366 : SUCCESS in 42m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b113807173d84672a299456286417939 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfc3864805e42949a615fc8d4db10a8 : SUCCESS in 5m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d13112f35a934900be4d13d3ef6db4c7 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":5},{"id":"f27c329739b0acd59a23cc4f285a7ae16116b499","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 23:41:03.000000000","message":"Patch Set 5:\n\nrecheck\n\nIDK what that failure was about, but this doesn\u0027t touch the reconciler.","accounts_in_message":[],"_revision_number":5},{"id":"4ec69dce1039387e3d3a238aea5b02e1acb4da2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 02:04:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925fd07e98c24f7f9722622b1213da84 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/7558804a14bb4b99a804697c052bdb6b : SUCCESS in 1h 05m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e347bbaaab334e0aace61221f9652d1a : SUCCESS in 49m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/212d0543d45f4dcea4555a8648a32e78 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d8d1bd1274c44d1493f40dadad1135e5 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58f3c9a034944378890aade43257cb39 : SUCCESS in 13m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/190a4168f69a4bca8787a00181ccec8b : SUCCESS in 14m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13fe5c56e4494d5cb146bf8998f22977 : SUCCESS in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3ac5b6b2b0401193dba77a671e9ea1 : SUCCESS in 13m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc0ccec3753b406fb2943e9782a469ea : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98ec9643a74045469a0b82ca48b8180c : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfe760025ee947adb44ccdf064b30122 : FAILURE in 7m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/61200c28239744a392dd0cdfabd7d09b : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5193d0bbad25486296644ca891bb1a71 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd1b4ea9d90543b1aaead9d994342f99 : SUCCESS in 20m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87f9d1c92c1e4874a220edb09ef927d4 : SUCCESS in 30m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/76194e5e29bc4f86a3c0d7d323b0d3f1 : SUCCESS in 41m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fbd7d4122752450cb0c5c75d124a618f : SUCCESS in 44m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/245b4b0481b54c1ebedd893850f4f0ac : SUCCESS in 57m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d55243c1ff4b08a2cc55fcc5daaa61 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b975a0881384884b883bc59e04cacae : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2bc1f4e6a2646b4a9948deabddf45e0 : SUCCESS in 25m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c63b40a368c94cd190af70ea5b06af9e : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":5},{"id":"6f79ba24364ceedd8e2158405efbbb518cd64752","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 03:53:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a09d84b2a407811f07f2b77e8b9aeea210d4e95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 05:05:13.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1746f3ce67b94f39b0b2192c4a8fb6c3 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8fd6b839209b44269f1df153f92e7f3d : SUCCESS in 1h 03m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63a398084c124d21b6277fab2b8dc831 : SUCCESS in 49m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/081f342292584cdca6b98c9e5fc817de : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b793b2a302b64091908e224f7dc772db : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e14fec92ee854c12b553129ca4647f05 : SUCCESS in 12m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29cd9f3ffd5f46f0958a1497fd98efe9 : SUCCESS in 13m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b826760fc03b41a19d2ec9c265dac344 : SUCCESS in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/440070011831471393e1f1edfb288f5a : SUCCESS in 12m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ddd297588a45480f93ba79fc0f93163f : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e49d14335074f2aa1789cd447a9f0a3 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22684cde4e01418d8433fe185262dd5e : SUCCESS in 16m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/51812ec39fb94a40bee75efc03dec839 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/92f011e955984d20ae6b5be475d22a51 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/75a3d86ef56c4ef385bf64143431c5ef : SUCCESS in 17m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/288b9f02621b4fe09864041fe38fe16b : SUCCESS in 24m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb8deeb8070a4c00849cc82626a604f0 : SUCCESS in 42m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1ffb9b09cd6459c92a06a7b5a89597b : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d333425a8b4480f8b712519e9f104c5 : FAILURE in 15m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/008f5a96bb364de783a46e9fa7c11ddf : FAILURE in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489373378d3940de991017e6d2d50fdf : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/770cc5ed9c474199845019838fddae3e : FAILURE in 7m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8922acc0e5a64e659824f802fea9f83f : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":5},{"id":"be7000ddcd91ed939f9f8b4f5905ce05a425a4a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 15:40:04.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"123e41247e034669138dbddd8ca2831a011800b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 20:36:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8ea0c900684aff87f749a9580436ef : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d0e64347f3d24a1192e1e8e44609196e : SUCCESS in 58m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e46b90f8841d4754b34b4680f36ee453 : SUCCESS in 45m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/593ef82f668c4af0ad904f8f0aabcb89 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce0bf79dc77b46c297c55d6d1df7e0c5 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5c4a4e39dbb4214b3b9615416ab3431 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ca4021f44824d9d85a4ab3772b5de8f : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bb765e8afea4b2ab84c76031c855e13 : SUCCESS in 12m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4403fce611fa4501a5158e71f660b0e3 : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/549721c6e1f34f338085c6920a9444e7 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef6395bf110e423da427b8790ca3d6d6 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b855ead3fc164a029cdea06af75f749f : SUCCESS in 18m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f837c79b012443d9a6415ffe6ac096bd : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ffe93625bb1a42dc84603d5767d81827 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/81c535e2643248af984f1ed2e0e3f9c6 : SUCCESS in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f3dc404dfac4fadbd1862ef17b64410 : SUCCESS in 25m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/588e61eb2dbc49d6aa4ff7812386e593 : SUCCESS in 46m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/33cb3305603344f18b8490a6a5ca77ac : SUCCESS in 50m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f30dd9f90c24fc5a589f3dec7d38eb5 : SUCCESS in 40m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd053419e0a46c0a29568a7bdb5b62c : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d932328c30483893abd7f31904eff4 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74969ea7c96540b4861698a2216d1be4 : FAILURE in 31m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8154c11453f746d2bbc739a9b38bf533 : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":5},{"id":"70b6f0e1a97d67f06f746298e01ce4015b5df03e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 23:33:26.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e1d9b3527637997ab81d6713cc7fff4838798779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:46:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a641749009423db9c094fb8e7c12bd : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/aac6229cade743579267c2bf7f812380 : SUCCESS in 1h 06m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49810abc62b64efd89f23f050a844809 : SUCCESS in 50m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30211e2ab2c54b14ba7d8211a562a6f5 : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/681c5f1348e84633a59714e0285b231c : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5da6367cb617427da359241606263710 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cd000a5a2024803a4f695b32e0f361b : SUCCESS in 13m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7dc3413e128e46d8b9bca4aaae5c33fe : SUCCESS in 11m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/22227ce9313441d08ae69953bca4cc0b : SUCCESS in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64801787758448068e99e60e0bafde0d : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/916b8035c13649bf9171a17b99d26423 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/083c52a506c5457e9c5dece9d9ea6934 : SUCCESS in 21m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e96541ebbe348fda10b2cfb01c38b1d : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a06a60254c514773817ad7a80885e17e : SUCCESS in 19m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/10df6ec1d4a249f9b44018c3c2f30950 : SUCCESS in 20m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/008ab3a2c8104fe2904f93e5217037b9 : SUCCESS in 19m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da58c669e66c42048c3ceae4ceac4d2e : SUCCESS in 47m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c398413ffe84f6aa94d6c887f7056da : SUCCESS in 47m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c877681104f14a418b16a4de718b7c0a : SUCCESS in 39m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1b158e56064895a3191c7ac897938d : SUCCESS in 15m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5bb21e279e4dc4ae39578d04cfa366 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bfef90b311a443eb0db63259c489687 : SUCCESS in 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b99ebb7019c54ed9aa321e4cd3110022 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c7e69e9b1ed7b281b7b01a5b5f2f9fca2b8e41ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:46:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"06ba00f615c45cc8fcf99bf2799ea7649fe6fdcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 02:17:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca21ba5e3c149fb8d18af53026efc73 : SUCCESS in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/7bb5f972d14b49b09c1c4739c8381465 : SUCCESS in 1h 03m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b677e781367440bca09babe1cc159536 : SUCCESS in 36m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/464b789efed24601b8f0d24d485729b7 : SUCCESS in 8m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/df066608714b4d5ab2b03db141833e11 : SUCCESS in 8m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e41c7294857d417c9c3489c82c7df32f : SUCCESS in 12m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a0ffa67cbc14dd0823510a932e8990e : SUCCESS in 14m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f120762ce7c463a83962acc5073086c : SUCCESS in 12m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e49c95c044cf4ac18b13fb183494011a : SUCCESS in 12m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f8140b5f18b433faddbceba01ff1dde : SUCCESS in 29m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86ef86ced7a24e609166ef435741563a : SUCCESS in 22m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d63b2d63dd194df7a479c95e1116402b : SUCCESS in 18m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f2e2b4e073f8482ab61961d5c432d3f6 : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e306af445dce4f6dbe6338168ad17a16 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/72576addbfcc4ccdb80a3895b333f49b : SUCCESS in 30m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3a02a8e6486491c9baf9fda6017afb9 : SUCCESS in 53m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64688730ee8b4208bac17054d479fac8 : SUCCESS in 37m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e1b2d0e69284e4caf6a993ce47a0550 : SUCCESS in 39m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e19b8f9e87d498bbd893697530353ad : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcac773a9b26457aa2b4b5e484e2a5ac : SUCCESS in 9m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aea33b7498ec4325a650267f062d3b36 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":5},{"id":"5459421646caf57c077e3816299aa0d5ae900741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 02:17:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1e0d080ad757523bb9026107253836e7f29de7d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 02:18:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e05299e78f24bd89f984cceadcdb326 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3196bd3e989c4681a74637ec9185be05 : FAILURE in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8127c7ad2db54677af273ad0a95784b2 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9a0bac4ceb15ba0e6b24d0cd021972341e97e60b","revisions":{"4e627353e8f140aae3605fd1fd53bfec29863d5c":{"kind":"REWORK","_number":1,"created":"2020-09-28 16:35:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/754814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/754814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/754814/1"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:34:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:35:11.000000000","tz":-420},"subject":"Fail short reads in SegmentedIterable","message":"Fail short reads in SegmentedIterable\n\n...and check for it *before* doing the MD5 check.\n\nChange-Id: If238725bbec4fc3f6c8d000599c735a7c4972f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e627353e8f140aae3605fd1fd53bfec29863d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e627353e8f140aae3605fd1fd53bfec29863d5c"}]},"branch":"refs/heads/master"},"3193e841489d6b6c814ad1e70a270ca03bc5bcc6":{"kind":"REWORK","_number":2,"created":"2020-09-28 20:35:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/754814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/754814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/754814/2"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:34:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 20:33:56.000000000","tz":-420},"subject":"Fail short reads in SegmentedIterable","message":"Fail short reads in SegmentedIterable\n\n...and check for it *before* doing the MD5 check. We see this happen\nocassionally, but as best I can tell, it\u0027s always due to a\nChunkReadTimeout popping in the proxy that it can\u0027t recover from.\n\nChange-Id: If238725bbec4fc3f6c8d000599c735a7c4972f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3193e841489d6b6c814ad1e70a270ca03bc5bcc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3193e841489d6b6c814ad1e70a270ca03bc5bcc6"}]},"branch":"refs/heads/master"},"03beca68a86e842a48ab7eab9f519e473d5b50b0":{"kind":"REWORK","_number":3,"created":"2020-09-28 20:44:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/754814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/754814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/754814/3"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:34:58.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-28 20:44:23.000000000","tz":-300},"subject":"Fail short reads in SegmentedIterable","message":"Fail short reads in SegmentedIterable\n\n...and check for it *before* doing the MD5 check.\n\nChange-Id: If238725bbec4fc3f6c8d000599c735a7c4972f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03beca68a86e842a48ab7eab9f519e473d5b50b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03beca68a86e842a48ab7eab9f519e473d5b50b0"}]},"branch":"refs/heads/master"},"a8a7514dfdfee5156ac3c5d6b4b1c2286ed5c57f":{"kind":"REWORK","_number":4,"created":"2020-09-28 20:53:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/754814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/754814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/754814/4"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:34:58.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-28 20:53:26.000000000","tz":-300},"subject":"Fail short reads in SegmentedIterable","message":"Fail short reads in SegmentedIterable\n\n...and check for it *before* doing the MD5 check. We see this happen\nocassionally, but as best I can tell, it\u0027s always due to a\nChunkReadTimeout popping in the proxy that it can\u0027t recover from.\n\nChange-Id: If238725bbec4fc3f6c8d000599c735a7c4972f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8a7514dfdfee5156ac3c5d6b4b1c2286ed5c57f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8a7514dfdfee5156ac3c5d6b4b1c2286ed5c57f"}]},"branch":"refs/heads/master"},"9a0bac4ceb15ba0e6b24d0cd021972341e97e60b":{"kind":"REWORK","_number":5,"created":"2020-09-28 20:56:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/754814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/754814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/754814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/754814/5"}}},"commit":{"parents":[{"commit":"004052dc65406e012822312f267d8374e760d34e","subject":"Fixing S3 credential caching region support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/004052dc65406e012822312f267d8374e760d34e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 16:34:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-28 20:56:00.000000000","tz":-420},"subject":"Fail short reads in SegmentedIterable","message":"Fail short reads in SegmentedIterable\n\n...and check for it *before* doing the MD5 check. We see this happen\nocassionally, but as best I can tell, it\u0027s always due to a\nChunkReadTimeout popping in the proxy that it can\u0027t recover from.\n\nChange-Id: If238725bbec4fc3f6c8d000599c735a7c4972f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a0bac4ceb15ba0e6b24d0cd021972341e97e60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a0bac4ceb15ba0e6b24d0cd021972341e97e60b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
