)]}'
{"id":"openstack%2Fswift~754851","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I2d22e087a88c9e8003621feb26460ab6e5ce2a57","project":"openstack/swift","branch":"stable/victoria","hashtags":[],"change_id":"I2d22e087a88c9e8003621feb26460ab6e5ce2a57","subject":"py3: Fix swift-dispersion-populate","status":"MERGED","created":"2020-09-28 20:42:27.000000000","updated":"2020-10-02 21:40:02.000000000","submitted":"2020-10-02 21:38:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754851-1601674708490-48b48ccc","meta_rev_id":"1f9705345c84a2d2378db96f57b0cb858c4c2a3a","_number":754851,"virtual_id_number":754851,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 21:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 15:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-29 15:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 21:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be3ad300804cada2efc2997ca8cf883e7a582769","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-28 20:42:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e01b539231e42edbd3ea61595225a148df66393a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 21:32:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1bafe48a384f9e9dd27a5f6ae3d59f : SUCCESS in 7m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5765f7880f24ec1a36331d0b9a6acaf : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e376f71e0c904802b2488a331e39c1c6 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5995578fac0549bc8b6ae177625ac2e4 : SUCCESS in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aeff535f6dd54eed9644e63d4281d5b5 : SUCCESS in 13m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4ee92dce5dd949b8b58ac1cf1dd5b032 : SUCCESS in 12m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad33693294cc4955a7d5e6edbf8af29a : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e27ab01b308d4cde984baa2848c4243b : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef281cc14d2e40ae962733202981a151 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e45b30f830c24365ad9dd79b0d6464d4 : SUCCESS in 16m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/73fa22390aa34750b0f8df26ea254fe0 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/abe06f494ba141d9aaac068c0895b9ab : SUCCESS in 17m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a3e758e32c7f4a9c9a8e830711ddc14b : SUCCESS in 21m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83d734bd3c1445c29a6c822c810f4646 : SUCCESS in 20m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c01fc8c770ca49c3be265b05bc2cd84f : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d63d66c2f09d4e5899a4de5038231080 : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e96e8d92a646f4ab64ea1a042be112 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86281a40970f4e3c91e05a1be436edf6 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95a810aeed0ae1043f0ec4221bf51e2d8df000e7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-29 15:14:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17396e86e37338a1c7743ea2aba0e6546eeac090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 15:14:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79f3d4951a61ce3e9e58328a187df27ef631e60b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 16:54:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f08f67895b43d699f6f3db86248e40 : SUCCESS in 10m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a862abdd7b8641838767aa5581bdef84 : SUCCESS in 46m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/891576582c1244bebe35934882a49044 : SUCCESS in 13m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9d0f897d4b7f41d99d75861bd27d63c2 : SUCCESS in 12m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a0311e85c0640ca8581313e65ab8147 : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fca083cbbf247dcb0ccb947cba99dac : SUCCESS in 17m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f455a1862c5641c78e5928640f2d975e : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3d4ba2e3072742c48ace387ae314627c : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dfb898cec785452f8ef390934bc07fd5 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1613da93e02a47c3b286014e9ad30c1b : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dabd8ec6ab3d4e43bae9a8f6f3a9506c : FAILURE in 18m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fb82002b96054b99af99bc5edba88172 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/04a034d301374833839daaeea25bf23f : SUCCESS in 18m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f67f0ae7bc12499a9db3e3bc72fe4f57 : SUCCESS in 44m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe45cd8e3735400a817b5c26711ce279 : SUCCESS in 37m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/535d7301f7844eb8a924ce07bcbfc48e : SUCCESS in 40m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81088960ce2e41ef9e1378aea7a4becc : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6932a1ad67a0457fbd8bc1afc9ac1a65 : SUCCESS in 7m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64e9f438b4a140af96f351567af675fe : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"dcd5a4d5bb7f7331d7f256404d3cdeebc8a48495","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-29 22:23:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4489cfd2bebd85d7eb7f5b52b042500369ccee96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:19:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ee0707f947493197c8c7dd19d8b21e : SUCCESS in 8m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71423f1055cc44d0b2b3faff31a4c56e : SUCCESS in 44m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ce004881b2b42949ecede9bd01d31ad : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59a6d8243de44725975bf59b079d3b95 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb9412d32f3e4f249ceef9e79d4625fa : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7138686f696d4ed8a1cc918c0fc07241 : SUCCESS in 14m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e79a4f33931b463ab2ceb7bd091ca187 : SUCCESS in 13m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/148422da1e64400b88f176ba8c1728d3 : SUCCESS in 13m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8117939063b74137870626cdaabaf1aa : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f43393ca47e842fea772a76aab0c4b70 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a0ca3c61942436e86b60535dc4a135f : SUCCESS in 18m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e39e147b0c6c4f5f84d4cbdf70d8ac94 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4ab91fe1a3174cd09b388f9cdff9e6e7 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aecc1d1a8ca44fe3b163f685c547b4eb : FAILURE in 18m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/928fb788c45748d9b751ba4cb767e071 : SUCCESS in 24m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e98a13e7eaa2453e83d39d6e8cf7d773 : SUCCESS in 43m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b963d086d2b44b6834210310cab256b : SUCCESS in 41m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0528c62cd984f2a8affda1c564842ba : SUCCESS in 41m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b73001467ce4fa6900576c9dc034575 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a820196d65bd4958b3c45d313511aecc : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa5ea282d43b4f80a9608d640895969a : SUCCESS in 26m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5700f539a4f4e54bf52717ce5ee692c : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a41d9b4ceb7059f551eea6bf76cc038a6d560b01","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-30 00:36:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f646568192c33c54e9f4e24064b1ad3185bef069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:38:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff72b52485b47af98d36497895645ce : SUCCESS in 8m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/743759aab5cd4d70a92735bada620ff4 : SUCCESS in 46m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8621cb9f492243e0a3e120c168bfb149 : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6f41be517bdf47969879c7b9e3ae8b2b : SUCCESS in 10m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4433f9950f2040b59735cf330bf6b902 : SUCCESS in 14m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c5772164b4f49dc92eca411e620cbcb : SUCCESS in 14m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84cde0c3fe2b4206b28bd639a64054fd : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6d9f8f38d9df420096e608a3beb8f0c7 : SUCCESS in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c024d9721674f2dbe067a028e3a9a8f : SUCCESS in 19m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d9caed7108748f9ab3cadc928e296d8 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b895349991e49e6af1b84f5630cff9b : SUCCESS in 18m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b288c1ecb54245b285b88da44c56778a : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/686b9bc4f87c4dc4a393aee42bbc7172 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/058ce6810e4442229e21ea8a362d808a : SUCCESS in 18m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea3d89fecb81451ba8bab5ed2173499c : SUCCESS in 22m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/686641b1df8f44109928cd6c45792f2d : FAILURE in 49m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/710c3a48c2484318abefe07c31775b47 : SUCCESS in 44m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df776038854b4753b47b16be184cf6e3 : SUCCESS in 43m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36f96e0590db40cea43b54d361ddee15 : SUCCESS in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ddf6c808964d70b476aae7d03d2e0f : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1dba7636ff11417c969a7a3dd49880f6 : SUCCESS in 25m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3206180081ea4c83a5d8938fd30e0648 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a75f8ffac74892cc7b94a712bb1e253d28ee1b63","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-30 04:13:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89e34346f4d404895b2e362ca106f2114e6b3812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:24:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2b74194eeb45e78a8450bd3ccfc66d : SUCCESS in 8m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/359c9e75ecfd48c79acdbbd430dd16c2 : SUCCESS in 44m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a3751533b4841068d66773a6e7b634d : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b6b5b87bbf0a4204b056442526717327 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/049198a2ee1b4e6ca6bc198d48ce979f : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1558edd04b7c48cba325a4d967def563 : SUCCESS in 12m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/12b55972e2f547edba007b4857d87c72 : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/947645d756a44b02b2dde20f541aa2cf : SUCCESS in 14m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d43a6ef96374977835b9781b21a5336 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb5757ff5ae042fe8865cb48db134291 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f35a78d4b5c4b32abbe87c67f61c798 : SUCCESS in 18m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/356e2d3c80314f379f509d86535879ef : TIMED_OUT in 1h 01m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ae3518c5bd8040feb0674f5b6548c5a1 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f353c70fe7b04311ba2e796388dcadb1 : SUCCESS in 17m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9ff6e13f0ae24fc2adad189da42af14a : SUCCESS in 23m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/66043431fc4a4fcf8aaaa7797f30329b : FAILURE in 42m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9d499736b1d465d8ed063fd09e6b729 : SUCCESS in 44m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58c2204e738e4561840a685eebeb3d24 : SUCCESS in 38m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f7981f792ac4fe6a05f864b94decc26 : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e4a7f7c2564a1fb3349865fa9db8cd : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08667ee848ce4f2e83ec5c5a3fe34f8c : SUCCESS in 20m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a1a4dd99692447989ae80849668bb22 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7d7fe27c2574615f941cda2d2a13be846452ae03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-30 14:50:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a7481374cda19b7f0691b9c06d8b984c822c0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:44:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa36ea3206840a798d6c5ef3783d0eb : SUCCESS in 7m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7b02ed8173042248ea0cbfa64824c48 : SUCCESS in 50m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c518f96caae94747b09f5cb60b7eef18 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/409d7511dac340b2ac4d595062e19694 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46c47ee8d33848979d56fecbb6a04094 : SUCCESS in 11m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47637660c47849fea5e1f575beb288b7 : SUCCESS in 12m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e5d3b28edc084c728792dc0239ee7cfd : SUCCESS in 12m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/019c6baaad46485b95817126c94cdc77 : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0f6548a92494baeac347848e37de084 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c898fc37ece442f7aee30d40256bb8e8 : FAILURE in 49m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8adde2fb957e43f88ab8da36bbaac9b2 : FAILURE in 19m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1f45b30d2fd449a6ac204c2a27696ea2 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/847645d5876a4017b9d7847477cc0334 : FAILURE in 18m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0884ec912c1f47f29f9f2aa59a5f37f4 : SUCCESS in 17m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd1ee74b580843fbb8bb191a64f40eef : SUCCESS in 19m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b48c84d5302c437fb017e8f8f583e490 : SUCCESS in 47m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f8e7af700374305806cd02e0e5701e1 : SUCCESS in 43m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dd63593b5d84779a816c827cdc18e94 : SUCCESS in 40m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2970f8bc0d7451284a6a3424b651366 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbeae34454a4dc3b14b9c99d56be33d : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94e56d5f59d24987b4877bc2c8a378b0 : SUCCESS in 22m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fea8810961f414e81a6110c458d7964 : FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"440b81d3e51c1254d6abb39d9cb8c873832c55b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 05:48:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1931818c9f0e2aa4e7e137956eb57ddd5c7093f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 07:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab440770d7e84b2f8d977111f7353b5c : SUCCESS in 7m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3b8937c4e2045d28255d0f18c8ccbfc : SUCCESS in 53m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbb883a71d0a40828a0ba4f3617181a7 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24746d0e52bb416ba5053f837a91ab05 : FAILURE in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccda414743eb4fc18805ad201f03400f : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30c58f6f102e41f6861e3abda1a7874f : SUCCESS in 12m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3901a8dc705b4640b22488c39f7524c6 : SUCCESS in 13m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd59e160c3dc40bcb162aa828f9b07d2 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fe046a4a6cc457ba54c9cbd6ba8f786 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d00dda39e77c44909045df21934c8ed3 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf81eb07351f49fcb9d91163ac20023d : SUCCESS in 18m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/15ca4523ccf64a9d82aece31fa379a59 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ae1890c756c245eea1b898fb036d42e8 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a777d7348f1041be807d551b87ed10c9 : SUCCESS in 18m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe9269e257ae4dee8dc19110514ed7f0 : SUCCESS in 18m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92a1679cf3fe461faaff706a7ac01653 : SUCCESS in 45m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/beb11f21f75e4414ae8140db6441718d : SUCCESS in 46m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/315bfc76ec4b44aa8c0c625ce33baae1 : SUCCESS in 41m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2e2ced1fd8f45b8a3cf8c5465614ecc : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a440c2d9674896a58134c5f5176cbd : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c2de6d0bdca4968aed6ec6bcfda8e06 : SUCCESS in 20m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ccf362da4ed43f882ac9589539d9d5b : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4fc959d74307aac93aed5ae6e6f69f32fd7f333f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 07:11:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa56a99f18a89992c0abc520d8d98e52d84410c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 09:46:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/decd3955b7114cb7a60baa798b75b6ed : SUCCESS in 9m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/246e0ec99d1d445dab6b40b31a8afcdb : SUCCESS in 51m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c54d8b7dcd434833a6ef5e61bc664427 : SUCCESS in 10m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/12607cc7eb7f48f8af11fbc1b8dea5e5 : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dce6b95b640349f993fc871983f35017 : FAILURE in 3m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b290bb79eca4e87bf4537326a561868 : SUCCESS in 18m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bf53e6d75bcc4876a135cab3e144f004 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8403772c1e8846ba855213f25b25ed31 : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/581d4a6bb44e4d61993d07730733a692 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fecc4917025a460fa37a1d5c43a0dcf4 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ffd558b590d4932a8e6183357d367fa : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6739911527cb41a0a564b839c17daca1 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6f8ecf6da9214b488f37ea5642c8728b : FAILURE in 16m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/510a7b885b4b4e46bb9182457e9cf103 : SUCCESS in 54m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44ee258812a24d4f9811d1819fa9bf0f : FAILURE in 13m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b37dbaa9e1c0416ca2b18e608d4b50b8 : SUCCESS in 43m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24824ad124d044e991a7ad6195fe987e : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865a26d27e9a412ba018db2c0059f07e : FAILURE in 4m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be48179ecb94fba8d21dda5eb9dc479 : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":1},{"id":"04b5fbeb7bf0f0eb21edc32609a445b627c7b126","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 16:46:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f4a4141c6d0628109c289d8cfdde48aa96ade33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 00:20:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521d3948d2dc4f179fdfd555c04513c1 : SUCCESS in 21m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/fca3d697258b44a4b0340c2e506e0ff1 : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59f9885975484c848a942e3b50b4573d : SUCCESS in 45m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39f9f60250604eec85a11dd98a287cf8 : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/165e40172be446df981cd9248e05782d : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a3d9b71ff4c45f3bc5f251aa6668c4c : FAILURE in 3m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd9bec090484194966048350f58ef03 : SUCCESS in 18m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63ee8219a66a44eab35b0f89869da5b7 : TIMED_OUT in 1h 01m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f22474e500224763872670099507a27d : SUCCESS in 11m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5062abb330de47929ab3a01865b19bbf : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/371c8af7c0304d248c13898a8c252490 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ffbd13238d842be822db56cc3a5d219 : SUCCESS in 18m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1548a709994f4edf855fc40f61fe771e : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5e24d3db503c4d50924fefb15ea45006 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/024d90e5616b45438a1fce2d6a531d37 : SUCCESS in 20m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83fbd005ba904de28c91458c99b73a8f : SUCCESS in 23m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc90a09ea42b41019603af0bc2810b85 : FAILURE in 48m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d22a198f0e924e859ecca980264470c2 : SUCCESS in 44m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33cc528b170b4ed196da3df510475400 : SUCCESS in 42m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb57a5603daa47d2b570cfb0073ebecb : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa3cb58ae1b4b7fb1f7e8f7a95e4cf0 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d9898538062401bb6a9a41c496931a7 : SUCCESS in 25m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30f6b2de89e1468da2859759fb2bdc54 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9c23977e532ea639b625e375446b2a802583b6dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 03:53:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3442c577c33bb8749eb999bc5a16ae5aac765bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 04:58:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edde74bdd89d460ea6db0faa13a9e725 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8430c15602b44498a85c6cd6fa79fd2b : SUCCESS in 1h 01m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6eb37daec9041d58943a39803d34731 : SUCCESS in 39m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77ec399a978944ffa9f074dba545635b : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4aabf043049c4c618d3ae69f1002811a : SUCCESS in 9m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c54e8f717fb4693aa7c3ab5875d07b2 : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/967bcb2dd9a3491fa6e5048c8228d3b8 : TIMED_OUT in 1h 01m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/505a2b69ebc94ef486316f03709608de : SUCCESS in 13m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf5be4aca9cc4f508bcf9d8e76c3811a : SUCCESS in 12m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/312085e98bec4f2c984512b6dfc15b35 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/691b05be5af24a21bceeddbbacac43ed : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/efe95dbda51a40948c2e674ff36479cd : FAILURE in 3m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/faafb65bc386498880402e6c9f7dbf76 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d9a7b3539fba4c2fa7feff26124bae9e : SUCCESS in 19m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d7e4daadec8407ab35d2031f0f64f01 : SUCCESS in 18m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec677c913a3d4304ac6349e52435a4fb : SUCCESS in 21m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e8ed541744b41208836e8b5f2e93960 : SUCCESS in 47m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca346c309bb3484b97103ae07a724a47 : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21f8d676f5544caeab91f55ae9b30c54 : SUCCESS in 41m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e22d206663a453a8c6b5b0013cbf982 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b2324bbde84c4aa1afa4d33863b7d2 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/667775cd97b54b33940430e2fa865247 : FAILURE in 8m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764110c42d39468fa571b77fa6f7b651 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":1},{"id":"76c9b3a13904b02fc827338c0e65449824c3f255","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 15:39:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b1d417bd628d455e2acb79c99244321d7963583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 20:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87652b5fae8d4b038279c1a80f7d5854 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4ddfb92c33074f2d905638ff66f28c85 : SUCCESS in 59m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70d8b71c60ee462da285883943b72fa7 : SUCCESS in 44m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b453fb23ceb24e2b96576f1432891b14 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/40f088a6c756476799ca49056495a7fd : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5e5e686654943948a6738acdf94853f : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/773389676b704f23a8537964e7d194da : SUCCESS in 14m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aefe3cf7814845429070b2024591ef99 : SUCCESS in 12m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e78aa672eb4143b8a99769b332badda8 : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/962d2b467bd542fdb18c7ac33f6a9cad : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/612c3421a34f429fa1ebecd8c9fdff80 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1385e73b99b64bf9a612d846d80cf493 : SUCCESS in 17m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0bdcb27cb3194cd99780e2c937f0630c : SUCCESS in 20m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c10129469e974f2eb17234349d1854eb : SUCCESS in 17m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1c47209e931d431b8c93b55e2be96d4e : SUCCESS in 19m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5068136f9ab472cac8b1340ef36e369 : SUCCESS in 19m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/019c30b70ffb472696ca22482b2a9eea : SUCCESS in 39m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d12cd3b6da14b37a37584c2b2dc70c3 : SUCCESS in 40m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2c1d6548e744156bee89390cbcbf27e : SUCCESS in 38m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df9a34fa6bb44944942b96b7c976bd55 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2ab3e3728945f28a941dbebca5568b : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b50f1428af364d96a2f4bca9a327bf4d : SUCCESS in 22m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99a204e49225496689a0de41b08dea3b : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"74e18087922e9cc1e3a1d7942f49e17b153e951e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 20:18:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdf3d05b57bd6269a83596a4c626daeb5be709d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 21:38:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46247a939d454c3380e7c74bd27ef1f7 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/52efdd31d6fb4ff7a5b1270131755c0f : SUCCESS in 1h 00m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c37e06a1c05441696744b6e6192ee69 : SUCCESS in 1h 00m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/215cc97a0da54249bf6111240766ca55 : SUCCESS in 8m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/95c68e934afe4f0baf34839962afe447 : SUCCESS in 12m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69be4dffc9c84f029fd413609b14b2b1 : SUCCESS in 14m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1440079f5b24c76a42abb69cde47633 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/43f630450d55492795d0d79618202f30 : SUCCESS in 17m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/49420dac1a264629a0ac23e2944c6a4d : SUCCESS in 12m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f09c73b53c004349acd53b0001e6174c : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16bd2ecc74844d309fbce91106805131 : SUCCESS in 20m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7355ce9b5aa8461ea93fe5d69a23bd97 : SUCCESS in 21m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/694333fa155f4dce9b28e8d784954140 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/783b3136ccc14c279739db38d158a176 : SUCCESS in 21m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f9a0d8eb88448b58b5d848b1541775e : SUCCESS in 43m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/575c25d3d43844a38db4614a70fef721 : SUCCESS in 44m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef3b35a23328413c840ad7a9be8ac7d6 : SUCCESS in 41m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c6883e1f7045c38ae6eec12cd92e0d : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb8f4bb0a8644df9dc33f38927f986b : SUCCESS in 9m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a63d22018f784322901b7de4d4fe79f5 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9c242515ca8075676d60cc5523a6c185bf19c070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 21:38:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f9705345c84a2d2378db96f57b0cb858c4c2a3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 21:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f09dad5cba46ceb2d696782119d9f4 : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dd92ed6964bb4119ab92e56a8e0696b8 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/22c3be37539842a5ad93688dff38ecc9 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"508c444697e9d0d721adc0bcc0b863b43ab3768d","revisions":{"508c444697e9d0d721adc0bcc0b863b43ab3768d":{"kind":"REWORK","_number":1,"created":"2020-09-28 20:42:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/754851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/754851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/754851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/754851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/754851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/754851/1"}}},"commit":{"parents":[{"commit":"d87b333b4627d8ca8b0168cf4f4fc04d43e93e91","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d87b333b4627d8ca8b0168cf4f4fc04d43e93e91"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-11 18:46:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-09-28 20:42:27.000000000","tz":0},"subject":"py3: Fix swift-dispersion-populate","message":"py3: Fix swift-dispersion-populate\n\nWe fixed swift-dispersion-report already; -populate needed the same fix\nor else it\u0027d hit a \"maximum recursion depth exceeded\" error.\n\nChange-Id: I2d22e087a88c9e8003621feb26460ab6e5ce2a57\nRelated-Change: I24f4bcc3d62dc37fd9559032bfd25f5b15f98745\nCloses-Bug: #1895346\nRelated-Bug: #1863680\n(cherry picked from commit 5b8073c7703c42a0b2b07536cd84e16ebf811b04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/508c444697e9d0d721adc0bcc0b863b43ab3768d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/508c444697e9d0d721adc0bcc0b863b43ab3768d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
