)]}'
{"id":"openstack%2Fswift~755132","triplet_id":"openstack%2Fswift~master~Ia8d3fefbe19e144ffed0c979e4f12b1910dcf711","project":"openstack/swift","branch":"master","topic":"env-pipeline","hashtags":[],"change_id":"Ia8d3fefbe19e144ffed0c979e4f12b1910dcf711","subject":"Add staticweb to default pipeline","status":"NEW","created":"2020-09-29 21:16:46.000000000","updated":"2020-10-01 21:40:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cb0804beaf264fb3eff9b025026889aac82691e0","_number":755132,"virtual_id_number":755132,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-01 21:40:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-2,"date":"2020-10-01 20:50:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 21:34:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-01 21:40:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8430600cbc69f9dab17a4bfa33e7e2fe709ebcf","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-29 21:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f671cfbe5ecba6f4a308821e6921734d43dd0e63","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-29 21:18:09.000000000","message":"Patch Set 1:\n\nI\u0027m interested in this behavior for https://review.opendev.org/#/c/752770/ but I\u0027m struggling to come up with something that I believe would reasonably help us maintain it.","accounts_in_message":[],"_revision_number":1},{"id":"60407a5bd9fa254d45603167b4f2eda931ff5650","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-29 21:34:50.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fbe4719f584d698b3d6d6e0a706b6951c15bf314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:38:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a921574433754c17bd70e8874009960e : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/fb911bd1b61b4b62a266700d0ad5887f : SUCCESS in 1h 02m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/daea3cef87b64c7cb9c5f26a25f7e35b : FAILURE in 42m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/943f7743379a472894762026cac3af03 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be5a677fbcf64ddcacd9e3f308f7441c : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea213b6bf831434bbcfe82a0f2ba9476 : FAILURE in 13m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/372f8bf5348b4bebb6861bca0d88f4d2 : FAILURE in 13m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6032402a0c0e4ebfa492ed9ab5181bcf : FAILURE in 13m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48ebb852d9364bf68e52448d283a0318 : FAILURE in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc66b935e5654c67b07339857f92f812 : FAILURE in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdd2656925444531870cad8b39501240 : FAILURE in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aae9cc1049f94644871573011258bc3d : FAILURE in 16m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1896153070b14888995d47002de5a888 : FAILURE in 18m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0051af6fe2ff44a2bb1dce07b149f64c : FAILURE in 18m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8a1e3179925c411f915087d635020b70 : FAILURE in 16m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57feed9c23304eec814f7a23b6e5fbb3 : SUCCESS in 18m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d6b8305ac8324a5eb336f0fd061840e5 : SUCCESS in 44m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a9ca7833247459faa8dab9518f183e6 : FAILURE in 31m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e30d9357b3ce4391918c17d377942f46 : FAILURE in 34m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a78c0be7ba3a45c286f869b43e22cd97 : FAILURE in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590b8ad643e2422799d82e597a7ac13a : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c78891531ad4776ac6608945a3f8e0f : SUCCESS in 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d09d93c8d4e425798a0387a25916cae : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cfa1372fd172bc1f767a70ee85eb21b209a8758f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-30 13:01:13.000000000","message":"Patch Set 1: Workflow-1\n\n(4 comments)\n\nmaybe I can swap out those env copy\u0027s for subrequests while i\u0027m trying to figure out how to fix TestProxyServerConfigStringLoading\n\nfeels like a yak shave - I really just want https://review.opendev.org/#/c/752770/","accounts_in_message":[],"_revision_number":1},{"id":"cbcf253a5c0c783c9cc9a4e9e17dc609bb3925da","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-30 17:47:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"036a6ff9933603871657a4cfe2575c5b78721178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:49:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61de9598852445b3a94e9e2c737aea70 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2bc5bdb975734436b95a506fab4664a8 : SUCCESS in 54m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d35333e61d164de6933a4fd1929eb5fa : FAILURE in 35m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56114f0fb82c48ceb39dee089309f558 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e09ff89aebfd4bde8751775dc8a62133 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e122a2943a34c5db71905fbdfb58876 : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/900d8977149b4e71ac34b7a00abbe4c3 : FAILURE in 12m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c6c6d3a7112747a6b6302111cbb80669 : FAILURE in 12m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2d2f7735ed440798cff9742136b1cd : FAILURE in 13m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6cf8bb2c675843059327bf8de20c0720 : FAILURE in 15m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac3d187898a74333a01b83394c93ee82 : FAILURE in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5b83731fc94402c890b9b3f5416c804 : FAILURE in 17m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/707cc6cbef88429795f73ecb3e6afaeb : FAILURE in 23m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d3323a63516428cb21e8bf9b41ee632 : FAILURE in 18m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3f183807568d42ad8b38c812c963b9f5 : FAILURE in 23m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df615df1d90d4bb7a1f04d2a16a5ae30 : SUCCESS in 20m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3fe6b6243fc341f09ee7de6626ab4722 : FAILURE in 44m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/526af76edf8c45d8a95bc40d3b1b18f4 : FAILURE in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5fcb45a225d9461d971927ba53050f56 : FAILURE in 27m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e9535a55a74e3191197844940e0a28 : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8478e9c206984fb184a503bbe8f09074 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b38005060ed74acfb860a5b6feeb61e5 : SUCCESS in 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fd70613844d411aa3b4da3881a3000e : FAILURE in 34m 32s","accounts_in_message":[],"_revision_number":2},{"id":"45a7a7de346ea3821066d35fa5b33f4162935f76","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 14:18:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb5303aacb1a42e35b29c99bd72e21f2421dc764","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 15:48:19.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nmaybe it\u0027s NOT a good idea to put staticweb in the default pipeline - this middleware is inserting itself into a significant number of requests","accounts_in_message":[],"_revision_number":3},{"id":"360ec072cede106c05cca3ef9b2f44b5ac877f59","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-01 20:50:07.000000000","message":"Patch Set 3: Code-Review-2\n\nstatic web doesn\u0027t seem great to me actaully - maybe it shouldn\u0027t be in the default pipeline - at least not without the fix to the env mangling of all requests.\n\nthis change is not a pre-req for better error catching when we encounter problems download SLO bodies:\n\nhttps://review.opendev.org/#/c/755639/","accounts_in_message":[],"_revision_number":3},{"id":"1937654c7206e980b6dd05047be662744bb700af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:40:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a9e5b2bfc047b9bd4364ea73e91f65 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/4f20e0ba2be341df82e855d68d9f3b07 : FAILURE in 41m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b2e8ddc137b41f9ab7ceca910c357dd : FAILURE in 42m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0852b355561469797790ff82620d818 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edb2c51872f147808cced2a16b8b1739 : SUCCESS in 12m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c3b2630f97b4b6597629f65ee828727 : FAILURE in 3m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe85e8bfd48248f0b1294ea325f6b6cd : SUCCESS in 13m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08ab2531a01143a99dbf9bb2b2c5462f : FAILURE in 14m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/016338b707c0496aa66642d99083a14e : SUCCESS in 12m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/358e1e39971c49fea4a84f459b781159 : FAILURE in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1207ebba96fd425687c6c325b8fdb8ef : FAILURE in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b3def2a00a92413199483464f753cae7 : FAILURE in 17m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/478cdc7970194eeca33357ddf6d43f67 : FAILURE in 16m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b9a3cc0a29aa49dca05dddd07f9156cd : FAILURE in 17m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7dee2c739c0349fc8785e59620e908f3 : FAILURE in 20m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0005c950e3e74defb9ca88dc836fdab6 : SUCCESS in 19m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bea1891f2a184fecbdc43e5bebaecb92 : SUCCESS in 47m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ab7a4923a0b4dc09e169c936b0b6076 : FAILURE in 17m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02481cccea38444381058a0bb2efef53 : FAILURE in 18m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6af5c72840e94c21a140d6b06535c2e7 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6194841188f6481c98603a96cafcc10a : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56c61baea9444856be04ca17d83d58de : FAILURE in 10m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be1f639cffdb4641a85f1f7ca4759225 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f820dc8dc7ba3af40da461c97ea5801bd46cf6fd","revisions":{"dddd97ba051bc63cbdb482a07ec877942d825984":{"kind":"REWORK","_number":1,"created":"2020-09-29 21:16:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/755132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/755132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/755132/1"}}},"commit":{"parents":[{"commit":"63d04b953a046b4d497c59f98d9ca696992f9c3d","subject":"Test proxy-server.conf-sample","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d04b953a046b4d497c59f98d9ca696992f9c3d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-29 21:15:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-29 21:15:32.000000000","tz":-300},"subject":"Add staticweb to default pipeline","message":"Add staticweb to default pipeline\n\n... and add a test that at least some basic requests with the default\npipeline are preserving the request environment all the way through to\nthe app.\n\nChange-Id: Ia8d3fefbe19e144ffed0c979e4f12b1910dcf711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dddd97ba051bc63cbdb482a07ec877942d825984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dddd97ba051bc63cbdb482a07ec877942d825984"}]},"branch":"refs/heads/master"},"4ff14b7b55fc8a40dfb0a5536a2d035972b32c06":{"kind":"REWORK","_number":2,"created":"2020-09-30 17:47:40.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/755132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/755132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/755132/2"}}},"commit":{"parents":[{"commit":"4d48fb6737c15ce9aa3319010b055413662b32dd","subject":"Test proxy-server.conf-sample","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d48fb6737c15ce9aa3319010b055413662b32dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-29 21:15:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-30 17:47:23.000000000","tz":-300},"subject":"Add staticweb to default pipeline","message":"Add staticweb to default pipeline\n\n... and add a test that at least some basic requests with the default\npipeline are preserving the request environment all the way through to\nthe app.\n\nChange-Id: Ia8d3fefbe19e144ffed0c979e4f12b1910dcf711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff14b7b55fc8a40dfb0a5536a2d035972b32c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff14b7b55fc8a40dfb0a5536a2d035972b32c06"}]},"branch":"refs/heads/master"},"f820dc8dc7ba3af40da461c97ea5801bd46cf6fd":{"kind":"REWORK","_number":3,"created":"2020-10-01 14:18:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/755132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/755132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/755132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/755132/3"}}},"commit":{"parents":[{"commit":"f0d406127c87fcf1791bf4c0f4132c1e086f1f34","subject":"Test proxy-server.conf-sample","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0d406127c87fcf1791bf4c0f4132c1e086f1f34"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-09-29 21:15:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-10-01 14:18:34.000000000","tz":-300},"subject":"Add staticweb to default pipeline","message":"Add staticweb to default pipeline\n\n... and add a test that at least some basic requests with the default\npipeline are preserving the request environment all the way through to\nthe app.\n\nChange-Id: Ia8d3fefbe19e144ffed0c979e4f12b1910dcf711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f820dc8dc7ba3af40da461c97ea5801bd46cf6fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f820dc8dc7ba3af40da461c97ea5801bd46cf6fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
