)]}'
{"id":"openstack%2Fswift~755651","triplet_id":"openstack%2Fswift~master~Icd09ed4d5fb4e2b9b84ddead21313ea1c0a87c91","project":"openstack/swift","branch":"master","topic":"apache2-deployment-guide-update","hashtags":[],"change_id":"Icd09ed4d5fb4e2b9b84ddead21313ea1c0a87c91","subject":"docs: Switching to /srv/www from /var/www to be more FHS 3.0 conformat","status":"MERGED","created":"2020-10-01 21:06:05.000000000","updated":"2020-10-01 23:08:02.000000000","submitted":"2020-10-01 23:06:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755651-1601593599322-c1ef629e","meta_rev_id":"3c69290f2a7293eed4bde76b4cea918a904027e3","_number":755651,"virtual_id_number":755651,"owner":{"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-01 22:04:03.000000000","_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 23:06:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-01 21:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-01 21:16:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 21:16:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-01 23:06:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"912ba896c624b49b4aea2fcf07b8996eaef53e14","author":{"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},"date":"2020-10-01 21:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b1a48bd2851080772adc9d35216744722565393","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-01 21:16:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFunny that only SUSE seems to follow that convention (according to https://cwiki.apache.org/confluence/display/HTTPD/DistrosDefaultLayout)\n\n*shrug*","accounts_in_message":[],"_revision_number":1},{"id":"cdef204f017b5d500ac6090c417241e18881c614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:19:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9271b6965d4feba4b8e86d8c15ffe1 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6d2cc539fb47885da356680fe9012245d8565a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4ec60a8b1b8fc1a44640796dab3316d82343801","author":{"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},"date":"2020-10-01 22:04:03.000000000","message":"Patch Set 1:\n\nWell, one could argue that Fedora/RHEL/CentOS do too. The selinux policy includes the path extensively.","accounts_in_message":[],"_revision_number":1},{"id":"1500639775b50c2479a0c74457679dbf9ba419c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:06:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78eb88a34d24264b6e9cdecc9e9612c : SUCCESS in 8m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2303f01f0dcc4eb3990dc7db824d9577 : SUCCESS in 11m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b942bdd9bedb4860bfbf532744c307ce : SUCCESS in 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f1276394bc966dd55d845292ea159bf79c9bc3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:06:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"825fb948d75dbe0b508b968bd00d231cc1f4aaf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:08:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e0d164b8964050bbd4e34929fda8a3 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e93acd0dfa864f7cb717e60fa223b7cc : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b567811d700649aa8db51764611f0bcc : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38ab4593f471c4e9d9b2fa086fcf4be29b432b14","revisions":{"38ab4593f471c4e9d9b2fa086fcf4be29b432b14":{"kind":"REWORK","_number":1,"created":"2020-10-01 21:06:05.000000000","uploader":{"_account_id":32375,"name":"Renich Bon Ciric","email":"renich@cloudsigma.com","username":"renich"},"ref":"refs/changes/51/755651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/755651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/755651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/755651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/755651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/755651/1"}}},"commit":{"parents":[{"commit":"d17ee8ba14cc88e832da80a939a4129b9eb908f0","subject":"Pass thread_locals through to frag queue feeder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d17ee8ba14cc88e832da80a939a4129b9eb908f0"}]}],"author":{"name":"Renich Bon Ćirić","email":"renich@cloudsigma.com","date":"2020-09-03 00:12:37.000000000","tz":-300},"committer":{"name":"Renich Bon Ćirić","email":"renich@cloudsigma.com","date":"2020-10-01 20:55:40.000000000","tz":-300},"subject":"docs: Switching to /srv/www from /var/www to be more FHS 3.0 conformat","message":"docs: Switching to /srv/www from /var/www to be more FHS 3.0 conformat\n\nIt\u0027s more modern and well supported to use /srv/www now in place of\n/var/www.\n\nChange-Id: Icd09ed4d5fb4e2b9b84ddead21313ea1c0a87c91\nref: https://refspecs.linuxfoundation.org/FHS_3.0/fhs/ch03s17.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38ab4593f471c4e9d9b2fa086fcf4be29b432b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38ab4593f471c4e9d9b2fa086fcf4be29b432b14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
