)]}'
{"id":"openstack%2Fswift~755906","triplet_id":"openstack%2Fswift~master~Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822","project":"openstack/swift","branch":"master","topic":"log-wire-status","hashtags":[],"change_id":"Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822","subject":"squash: More logging refactoring; stop needing an extra catch_errors","status":"ABANDONED","created":"2020-10-03 05:35:39.000000000","updated":"2020-10-06 20:19:00.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efdf81316e88272e37e7c53868a7c1862be9b348","_number":755906,"virtual_id_number":755906,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-05 23:41:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-05 19:31:15.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-10-05 19:13:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 19:13:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-10-05 23:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7079fe0f783a3a0b55e8c37036c75f9e4d51d499","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-03 05:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9374fa9341d8bf4c2c7c13c1b9ac2af688858cbd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-03 05:45:05.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nStarted to turn into a bit of a yak, but I think I like where it landed.","accounts_in_message":[],"_revision_number":1},{"id":"5409fd7ec74ddc966a3a8c2c74e4697ff7a6f94f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-03 06:21:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a21c68349a7324fbdef16ef211c3a9f1b964b0f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-03 06:45:25.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n:-/\n\nThis is better than the first pass, but I\u0027m still having a bit of trouble with test/unit/proxy/test_server.py:TestECObjectController.test_ec_client_disconnect","accounts_in_message":[],"_revision_number":2},{"id":"61b095d3873bcd6052bb6556b693a4381d6b2c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 08:02:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512209866d22470b96782312b8e5d665 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/c5e96a831c0a43ebafcd483f64fd25fc : SUCCESS in 1h 00m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32e6a4990f0c425db64a2102845aebe6 : SUCCESS in 45m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b23483a4e6c340d584ba06cd891dd385 : SUCCESS in 10m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ab49b71a55a45569fa17a817335bfb0 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b5100b05b714327aae53a7ffe58299b : FAILURE in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6dae9e89f13a4967b5db593c07c8033e : FAILURE in 20m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5530391b23bb4f74a28bba1370cab065 : FAILURE in 22m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c1b3ebee358431581e5797c05a7475f : FAILURE in 12m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5719ae11a557496e8146494833f2113d : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f29c1a8102eb4a2f9a79db1cd657150a : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a16b301ada914908838a3c703ac39c5f : SUCCESS in 17m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c4706cfcc9904f1ca338916abeaea2cd : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f19a4b60c4634eca8516ae416f15094a : SUCCESS in 16m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a20ccc390dd047d1a1f75f20b09f8073 : SUCCESS in 17m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e9c2dec26a64348880730ac0cc370c0 : SUCCESS in 21m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e67417b454db41c9b6636bec8b8f3a01 : FAILURE in 44m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13432e10ee9d4831995e5857def85d6b : SUCCESS in 36m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/482618e812e440f1892d6c59a8c714fa : SUCCESS in 48m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb169dbb22904b68838362ef79da36ed : FAILURE in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095390dcfbdd4e8ca1ac7253fff932e6 : SUCCESS in 8m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b93780d47da24644963ebec72a331425 : SUCCESS in 28m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a54e583a5f941afa6425f5eda5b1fe8 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c6d4bf8e955a7e2c8cc45f04770a472fe60da92a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-05 16:56:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80d39909501f9cc5b8c133756197c033fb6e8846","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-05 17:01:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c577db6ae8015a66276a4e99c91a43fcc1c4e77","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-05 18:46:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"970132671ef8041d005668bab5dc562ef2d29944","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-05 19:13:25.000000000","message":"Patch Set 4: Code-Review+1\n\n(7 comments)\n\nthere\u0027s a lot of churn with iterators and next and I don\u0027t quite grok where all of it is really coming from...\n\nbut the meat of the change in proxy_logging seems quite reasonable, let\u0027s squash everything down and keep working on wire_status and unified byte-tracking!","accounts_in_message":[],"_revision_number":4},{"id":"899010ec15836ba7cdc834b2f5248d709eaf784e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-05 19:31:15.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e there\u0027s a lot of churn with iterators and next and I don\u0027t quite grok where all of it is really coming from...\n\nYeah, sorry -- I found a bunch of places where we\u0027ve got an iterable (python obj w/ __iter__) and we seemed to expect that if we closed an iterator over that iterable, we\u0027d close the iterable, too -- only that wasn\u0027t how it actually worked. This seems much more hygienic.\n\nDefinitely one of those times I miss Sam, though; I feel like he had a good head for making sense of these things.","accounts_in_message":[],"_revision_number":4},{"id":"bb2e7a96d47a4671ceeda1ff32322ad27e698a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:41:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77318f9ef0da489ba2ef01f03c9b7f66 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/fe8e8f92d5614da28be3364a7439fef8 : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0279791cc6584574a2b14b443cc2bc65 : SUCCESS in 46m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4f9e3db386b489fb2c2d8fa363781a5 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b42ba0ec295844c58badfb8c67211b42 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9e4745d220994db9a0ecd1f2d4201f44 : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27908114dc674505a29706ebb7025837 : SUCCESS in 17m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed2fbca4bbf54214bb4d64e81282d89d : SUCCESS in 13m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c762378d91d64360b36554796ae7bd21 : SUCCESS in 12m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b3626382f4d4d0db6157ed378e6cb3e : FAILURE in 21m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16ece30c8d06492e8972278bd15ecf27 : FAILURE in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/931080d3f6f449adb4a9dc3ee8501bcc : FAILURE in 18m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d621700cf19c4bf984ef1370fc65245a : FAILURE in 17m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/74e59db0c85e4bffbbb85087c52d1042 : FAILURE in 20m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5cf7f77198ba462aac3a51d16fd291b1 : FAILURE in 18m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce8b035715154626b71849dabc52fbde : SUCCESS in 27m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9979dfa83179487e95474d92bcb3601e : FAILURE in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9b9cfd0a7bb42bea11cb5111038961f : FAILURE in 39m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c4a0c5443554b70b3e05c8e817b4649 : FAILURE in 30m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc1843b55f1641269b05787afa9025a7 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee0c06b8da640708deab7d82e1facf2 : SUCCESS in 8m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb6f7a7b106040b3bb6271f29ea95307 : SUCCESS in 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc3edc462c324dd889ed76779d3fe3df : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":4},{"id":"883b44ea89546522b37e9ec867b68465fc29d8e3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-06 20:19:00.000000000","message":"Abandoned\n\nSquashed into parent, along with a fix for the object-versioning func test failures.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1aa0058d566e531cf3570bf2a5308bf402d5c40f","revisions":{"43f0ed071cac3647f45bd73192c83108e63a8324":{"kind":"REWORK","_number":1,"created":"2020-10-03 05:35:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/755906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/755906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/755906/1"}}},"commit":{"parents":[{"commit":"a52aebb4c8df885206c5fa14ee0c6e36a351786c","subject":"New proxy logging field for wire status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52aebb4c8df885206c5fa14ee0c6e36a351786c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 05:04:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 05:31:52.000000000","tz":-420},"subject":"squash: More logging refactoring; stop needing an extra catch_errors","message":"squash: More logging refactoring; stop needing an extra catch_errors\n\nChange-Id: Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43f0ed071cac3647f45bd73192c83108e63a8324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43f0ed071cac3647f45bd73192c83108e63a8324"}]},"branch":"refs/heads/master"},"ffa2d445faeec8fecf8b52af272071f97fa5df6d":{"kind":"REWORK","_number":2,"created":"2020-10-03 06:21:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/755906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/755906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/755906/2"}}},"commit":{"parents":[{"commit":"a52aebb4c8df885206c5fa14ee0c6e36a351786c","subject":"New proxy logging field for wire status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52aebb4c8df885206c5fa14ee0c6e36a351786c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 05:04:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 06:21:28.000000000","tz":-420},"subject":"squash: More logging refactoring; stop needing an extra catch_errors","message":"squash: More logging refactoring; stop needing an extra catch_errors\n\nChange-Id: Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffa2d445faeec8fecf8b52af272071f97fa5df6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffa2d445faeec8fecf8b52af272071f97fa5df6d"}]},"branch":"refs/heads/master"},"054de06150fdf053352f3f046b0c7c90ee4d8a6b":{"kind":"REWORK","_number":3,"created":"2020-10-05 16:56:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/755906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/755906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/755906/3"}}},"commit":{"parents":[{"commit":"a52aebb4c8df885206c5fa14ee0c6e36a351786c","subject":"New proxy logging field for wire status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52aebb4c8df885206c5fa14ee0c6e36a351786c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 05:04:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-05 16:56:22.000000000","tz":-420},"subject":"squash: More logging refactoring; stop needing an extra catch_errors","message":"squash: More logging refactoring; stop needing an extra catch_errors\n\nChange-Id: Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054de06150fdf053352f3f046b0c7c90ee4d8a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054de06150fdf053352f3f046b0c7c90ee4d8a6b"}]},"branch":"refs/heads/master"},"1aa0058d566e531cf3570bf2a5308bf402d5c40f":{"kind":"REWORK","_number":4,"created":"2020-10-05 18:46:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/755906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/755906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/755906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/755906/4"}}},"commit":{"parents":[{"commit":"a52aebb4c8df885206c5fa14ee0c6e36a351786c","subject":"New proxy logging field for wire status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52aebb4c8df885206c5fa14ee0c6e36a351786c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-03 05:04:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-05 18:46:39.000000000","tz":-420},"subject":"squash: More logging refactoring; stop needing an extra catch_errors","message":"squash: More logging refactoring; stop needing an extra catch_errors\n\nChange-Id: Icd7acda90ba2634ffae4fbd35ae8fd51b9dd5822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1aa0058d566e531cf3570bf2a5308bf402d5c40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1aa0058d566e531cf3570bf2a5308bf402d5c40f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
