)]}'
{"id":"openstack%2Fswift~756163","triplet_id":"openstack%2Fswift~master~I97e3e86a441c105f71a36cec96da308b70ab3fa5","project":"openstack/swift","branch":"master","topic":"p35_quar_on_close","hashtags":[],"change_id":"I97e3e86a441c105f71a36cec96da308b70ab3fa5","subject":"DNM: follow-up for Dark Data #35","status":"ABANDONED","created":"2020-10-05 21:48:06.000000000","updated":"2020-10-14 01:10:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"005445d7ea86b2692200265ce35a806dea711d1b","_number":756163,"virtual_id_number":756163,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-06 00:59:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-09 13:31:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 00:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 13:31:44.000000000","updated_by":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"reviewer":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"state":"REVIEWER"}],"messages":[{"id":"8ba182ab715dda849f3ae6b070b4c5a58491abd8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-10-05 21:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2c5dc2bd39358c43c4668ffbb1dedfd7dc9e4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 00:59:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150d04fba5024eb7a3db01a04ab164bd : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/caddd58c801c4cab903dad724551de87 : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7bf3b89d2fb04000bbd96d7c1ca0324a : SUCCESS in 46m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01a7aa353489401a87d79b3788e90992 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0b688ff53cc4c3887db61c94f3ca865 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e6b745e3ced4053975682f8b357bab8 : SUCCESS in 14m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8622e41b933d44ef855c9821748ad055 : SUCCESS in 16m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/073c833e0a6047b2b772c5e0fbf783d7 : SUCCESS in 21m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8e508e273546fa943233aad6bdcd3c : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b70f4d332ec04acfadad2236a76f9bb8 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/016871d7d31a4114992d7a3f60a5d357 : SUCCESS in 23m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/41a9bc308fa14aafb32f87513c9d7de5 : SUCCESS in 36m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7bb34420fa7a4f2aaff6c071b1aa76fa : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b27f123e3a10454b84ce3214d397b2f6 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/639b58b54ab943338f888f33fe2e3773 : SUCCESS in 22m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c7a3f2931b74e2791f81cd6a7fc6a4c : SUCCESS in 22m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9adcdacf7f3e4c17989c0e24c1cbfd56 : FAILURE in 48m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1a2da1056634a1ba107bd1953512f6c : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc20a81eb6f7480095b4334739d352f2 : SUCCESS in 46m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c99ac56920d442db95ed51ee75834b0c : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9ca7748da04f799455f00f218fc49c : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4426e676166e4967b213d43941df5369 : SUCCESS in 24m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f155696c588b4960a484c97595ba2538 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"65feb65b1e4176fd4d4fa78579cfa51d0e3a8275","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2020-10-09 13:31:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8522972def4c0fb702f86497f630a0406ec01eb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-10-14 01:10:29.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62a80e7b6cc72c4899fecc922fc2ee04865bbfc9","revisions":{"62a80e7b6cc72c4899fecc922fc2ee04865bbfc9":{"kind":"REWORK","_number":1,"created":"2020-10-05 21:48:06.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/63/756163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/756163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/756163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/756163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/756163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/756163/1"}}},"commit":{"parents":[{"commit":"cd7fbdf36c74de4246ee1fbec64d8dbce266aba4","subject":"Let developers/operators add watchers to object audit (simplified)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd7fbdf36c74de4246ee1fbec64d8dbce266aba4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-10-05 21:45:39.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-10-05 21:45:39.000000000","tz":-300},"subject":"DNM: follow-up for Dark Data #35","message":"DNM: follow-up for Dark Data #35\n\nThe quarantine-on-close is not possible because we\u0027re calling\nsee_object() when df is not open. So, let\u0027s leave this alone.\nHowever, I think we might want the following:\n\n - Do not call other plugins if one plugin forces quarantine.\n\n - Do not open df if we don\u0027t need it open, which is always.\n\nChange-Id: I97e3e86a441c105f71a36cec96da308b70ab3fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a80e7b6cc72c4899fecc922fc2ee04865bbfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a80e7b6cc72c4899fecc922fc2ee04865bbfc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
