)]}'
{"id":"openstack%2Fswift~757025","triplet_id":"openstack%2Fswift~master~Ia91f64a89b174a02d528af9c081300f46da7ffc5","project":"openstack/swift","branch":"master","topic":"remove-py2","hashtags":[],"change_id":"Ia91f64a89b174a02d528af9c081300f46da7ffc5","subject":"remove py2 test","status":"ABANDONED","created":"2020-10-09 08:37:38.000000000","updated":"2020-10-10 01:07:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa8ceafbeb04381ac28833ace3090188bc38d5a5","_number":757025,"virtual_id_number":757025,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-09 12:57:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-10-09 17:18:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 12:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 17:18:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"11def72f2bd67f9f40adbd8c320df30effb277a9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-09 08:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dd120dc53a710f7c52129faf7c66a838f22ea5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:57:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ecba592af0e45d8bc328ce6ae4bf7a1 : SUCCESS in 3m 16s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00479df9ecbb45b581c0bc57bd32fbf2 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c302c10f9b1d4a0ba9f5af633253986e : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222f3e1b505947e79e03cfb8b75a1d0c : SUCCESS in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0a081a4c03c44633a7f85672a237aed6 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6c4627dca734125a9bc7b0cb839982c : SUCCESS in 49m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a6fe12e26c64d22bc43308075997fb1 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c23f4317a54146bdaf33b28160a9e731 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a3436bad6804efab188267b22be3532 : SUCCESS in 14m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d44943b45a24b199dc81a2933166e4f : SUCCESS in 28m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/64bb7e430d7047519f1fbab28b9f667c : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ffbcd5f5a1947bb9c77642876f3c0d4 : FAILURE in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8145eee16924871b1bd6fa997abd0ca : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a04420b9f6804344938480eabd521a77 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb2539e3ff8643c396420163e9789e7c : SUCCESS in 17m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ade615a65e5c4f389ef3d2b3c8b98b99 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ffc69290afdc4668b20c516dc77ff4d7 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c211d926ac7443349e75f9e6bde6048d : SUCCESS in 18m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b1cd9958ac24c5aa13c8af32347bbd5 : SUCCESS in 21m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fe64ce104e294c058de15e8a200b86b5 : FAILURE in 46m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1067c0927a774346912e83e09e266aef : SUCCESS in 40m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c77fa2433d49478ab0b8b9da8d8714a8 : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4df274c39ba14f5ca03549abf152f3a0 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f6b8dc50e6415fbb3feca296631ba6 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f87324fa3d1d4b238feb155e846a23b3 : SUCCESS in 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2aca39c479ad493e87c8702b729f876c : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5c95421d153fb172ec48e13a3593d0d98d854185","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-09 17:18:01.000000000","message":"Patch Set 1: Code-Review-2\n\nOpenStack as a whole has moved on from py2, but Swift has not. There are several known multi-petabyte deployments still running on py2 that are understandably skittish about transitioning in light of the numerous py3 bug fixes that come with each new release. That said, deployments *are* moving to py3; indeed many of the recent py3 bugs were found by transitioning clusters.\n\nBut we will continue to support py2 until there is a clear, viable, and battle-tested transition path to py3. Anything less discourages upgrades and decreases the likelihood that the eventual upgrade can be done as a rolling upgrade.","accounts_in_message":[],"_revision_number":1},{"id":"4ae52c04fee8df565f5a2e55f6dcceb7ef321b34","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-10 01:07:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"160fb8df4e4ffdf70130ea1364e6b4dec9e3dac3","revisions":{"160fb8df4e4ffdf70130ea1364e6b4dec9e3dac3":{"kind":"REWORK","_number":1,"created":"2020-10-09 08:37:38.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/25/757025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/757025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/757025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/757025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/757025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/757025/1"}}},"commit":{"parents":[{"commit":"44ac59ddc38298b5c9ce215fa74e44055d9fced5","subject":"Merge \"py3: Fix non-ascii chars in staticweb listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44ac59ddc38298b5c9ce215fa74e44055d9fced5"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 08:35:18.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-09 08:35:18.000000000","tz":480},"subject":"remove py2 test","message":"remove py2 test\n\nThe Python 2.7 Support has been dropped since Ussuri.\n\nChange-Id: Ia91f64a89b174a02d528af9c081300f46da7ffc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/160fb8df4e4ffdf70130ea1364e6b4dec9e3dac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/160fb8df4e4ffdf70130ea1364e6b4dec9e3dac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
