)]}'
{"id":"openstack%2Fswift~758505","triplet_id":"openstack%2Fswift~master~If2b35925d790429968d8d68780072acdc6dc9631","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"If2b35925d790429968d8d68780072acdc6dc9631","subject":"Optimize swift-recon-cron a bit","status":"ABANDONED","created":"2020-10-15 22:06:30.000000000","updated":"2020-10-16 19:54:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be46e3f93d9ad73f7f97ec5cd1783263d35c3228","_number":758505,"virtual_id_number":758505,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-16 19:43:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-16 18:33:17.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 19:43:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4942420e520c1c31bd33957cd39bc2be9bb229b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-15 22:06:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25a511b1d45c24a9408eb792bbfedd2840111890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 23:40:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96449e59a52f4a8bb62ef7243acce483 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0737fd7dd33444748a853b0fbf963423 : FAILURE in 47m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7683527325ee4c879a487b9e294e8148 : SUCCESS in 46m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2b11946a15e4f59b4bb71f58319187c : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69338cae5dfc4c3482d2103c77fd691b : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b7f0e72cafe4fb4950e2fdceb6508f2 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b37322a0da64fd98946bff4d825337a : SUCCESS in 19m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85d4694823ec4238b312b7163594bb91 : SUCCESS in 16m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45f84b2b2a1743cfb492a34b82dfbae3 : SUCCESS in 18m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41892588fd09434fbfdb75e659b8b929 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aaa3048420eb4d448923957aa9a2da59 : SUCCESS in 26m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab0f046e83744eb4a301e8db32047b0b : SUCCESS in 20m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07997cada98b447ab2bd9512cc8be96b : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d7b857440404ec09d8eff5cf59277ea : FAILURE in 59m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3028f172bee849898073211e3556033d : SUCCESS in 19m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a79c7914539c4457a2d9af271e9636b2 : SUCCESS in 20m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/891f2f5c8eeb40d2890874cc2b8515bb : SUCCESS in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4900e1621b1c409395ccf1c5706bb257 : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8e67b4fa6e94c3285faa30330813da8 : SUCCESS in 45m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7867e83a30e4895995b688af2a01dcb : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1395d20d73764bf992619b34d788e51f : SUCCESS in 8m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e4061da46434b0f8102049904636f94 : SUCCESS in 48m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f257f9e2ae184676bbdec21ae2a1319a : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"24940b5968aaa20ba2431261ca3adef3d45cf924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 23:41:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96449e59a52f4a8bb62ef7243acce483 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0737fd7dd33444748a853b0fbf963423 : FAILURE in 47m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7683527325ee4c879a487b9e294e8148 : SUCCESS in 46m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2b11946a15e4f59b4bb71f58319187c : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69338cae5dfc4c3482d2103c77fd691b : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b7f0e72cafe4fb4950e2fdceb6508f2 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b37322a0da64fd98946bff4d825337a : SUCCESS in 19m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85d4694823ec4238b312b7163594bb91 : SUCCESS in 16m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45f84b2b2a1743cfb492a34b82dfbae3 : SUCCESS in 18m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41892588fd09434fbfdb75e659b8b929 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aaa3048420eb4d448923957aa9a2da59 : SUCCESS in 26m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab0f046e83744eb4a301e8db32047b0b : SUCCESS in 20m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07997cada98b447ab2bd9512cc8be96b : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d7b857440404ec09d8eff5cf59277ea : FAILURE in 59m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3028f172bee849898073211e3556033d : SUCCESS in 19m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a79c7914539c4457a2d9af271e9636b2 : SUCCESS in 20m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/891f2f5c8eeb40d2890874cc2b8515bb : SUCCESS in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4900e1621b1c409395ccf1c5706bb257 : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8e67b4fa6e94c3285faa30330813da8 : SUCCESS in 45m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7867e83a30e4895995b688af2a01dcb : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1395d20d73764bf992619b34d788e51f : SUCCESS in 8m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e4061da46434b0f8102049904636f94 : SUCCESS in 48m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f257f9e2ae184676bbdec21ae2a1319a : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0ffdf8bd39e54ee836f5dd989764752fab819da0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-16 18:33:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"01e05c6b7c6da434ce4a745d4a46c74453b22613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 19:43:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f9df77e35947ee8723393e674b3a30 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/cf1cb87bd55d44d58bc01cc33a0a8970 : SUCCESS in 54m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5e7696b6cca4341a879bb8abab20ba6 : SUCCESS in 43m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4cfa4b4d5a640b9b72c7a6ca9179775 : SUCCESS in 9m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30e3306433e4490f856230d237b57a24 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9b559d2d480456585844cbf22c82320 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5372f6e5fc174dc59f1526cc345ed879 : SUCCESS in 18m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d9546714b63045f6af0063464efd088a : SUCCESS in 17m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/440b5323afc9461f9160988897d7ec53 : SUCCESS in 14m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b20a83d573784489a5404cc2434b4c72 : SUCCESS in 24m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0f44afcde8c41d7a27f27fdb3f87e1b : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4577611ea00a4a41aa281294d88a0fa0 : SUCCESS in 24m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b473129aa5314fdc86200ff66fb629d4 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a3397edeada740b980669071706717c9 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/45ae470fb3904edf851bcd62ecf544e1 : SUCCESS in 25m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4dc506ca519b43b09c2b93b05e5db786 : SUCCESS in 23m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/061a54d1e54b4901b87579cc03722006 : SUCCESS in 43m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86af9649b12d4dbe9a68f412195c97fe : SUCCESS in 43m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c7698e1dfdb4b229c5f9e692297bfed : SUCCESS in 56m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/716326e8d7c647e58e2d1eb195f53960 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fe82f5851645a1b3e24e01b3c12ee0 : SUCCESS in 8m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3725621c012d40d1b14eddc3521186f5 : SUCCESS in 20m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b3498e3fa39442d8819562f9b7e0eef : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"be46e3f93d9ad73f7f97ec5cd1783263d35c3228","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-16 19:54:12.000000000","message":"Abandoned\n\n*grumble*\n\nI\u0027m being dumb, that can only help count subdirectories, not files :-(","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"982d314664828b045f6687e12f866e78bc3800bd","revisions":{"982d314664828b045f6687e12f866e78bc3800bd":{"kind":"REWORK","_number":1,"created":"2020-10-15 22:06:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/758505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/758505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/758505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/758505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/758505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/758505/1"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-15 22:02:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-15 22:06:22.000000000","tz":-420},"subject":"Optimize swift-recon-cron a bit","message":"Optimize swift-recon-cron a bit\n\nWe don\u0027t actually care about the files in the suffix dirs, just how many\nthere are. So don\u0027t count the names; look at the link count (deducting\none for \u0027.\u0027)\n\nChange-Id: If2b35925d790429968d8d68780072acdc6dc9631\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/982d314664828b045f6687e12f866e78bc3800bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/982d314664828b045f6687e12f866e78bc3800bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
