)]}'
{"id":"openstack%2Fswift~758603","triplet_id":"openstack%2Fswift~master~I3e5e904ffc0eba7a7e141e1c2d9f9840e4952041","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3e5e904ffc0eba7a7e141e1c2d9f9840e4952041","subject":"Add async_pending_last time to object.recon","status":"MERGED","created":"2020-10-16 17:39:40.000000000","updated":"2021-06-18 08:13:18.000000000","submitted":"2021-06-18 08:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758603","meta_rev_id":"b2247ba7eb198a31958960ff36b36daf079f90e5","_number":758603,"virtual_id_number":758603,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 08:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-06-18 06:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-06-18 06:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 21:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 06:39:53.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"47cfc2ec4e227f4f4ef4b16ac8bc98724b46b1a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-16 17:39:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17a998d80da1760f8f4554ac96b2ead8ecc5d465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 19:08:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8d67d8ddb846ec9ad8ce47816e2376 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/d74017de8e134b8bb99ce1983f20db12 : SUCCESS in 45m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/548a96be64d84e17b8ca02eafb9bf2df : SUCCESS in 38m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/189a59a73e8c40aba398a6f6bfa1a034 : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eed5cd9e301c4f6489c3aa0084405d2b : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba0365f9c4f24cbc96428f2c1cf76f3a : FAILURE in 14m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/936d3e29be2b4c188c47274c15a619f0 : FAILURE in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c386f366c5ca45709e5cd7789f2eb1ae : FAILURE in 42m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/99f707a2f40e4da59dd957cbf983c929 : FAILURE in 35m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6eebf4a3c14e48e8a9a15eebec1e2579 : TIMED_OUT in 1h 02m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a279a34a0bd9476abddb405551035464 : TIMED_OUT in 1h 02m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/55db52298a524ec9bbbf72b5380d90d8 : SUCCESS in 29m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/19132add756e476288ffa5557873eeec : SUCCESS in 33m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a1b64151ecca48daa951d56260de3b57 : SUCCESS in 33m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8794e39bc2a44321aa96099ebdfd1840 : SUCCESS in 20m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b2383ae5a46455c8f611a90c09a8d5a : SUCCESS in 35m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9fdef1b9541b47388250a9182a779460 : FAILURE in 30m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d68656abc60d4e13ac990c9a11fa96b9 : SUCCESS in 50m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6f8cd596e10457e84d897f9e0087546 : SUCCESS in 49m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c483049d581491096ace4a75fcc6b6c : FAILURE in 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb9d491f52ea4635b6467759fb93c62d : SUCCESS in 7m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39f28a5401524914a98c499594f1ceb2 : FAILURE in 1h 00m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c34bc03dcc23452ebceadf04cd0b8a1f : SUCCESS in 1h 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5877a98a4d41fd4fac2c63ae7c34d23df5cc5c96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-16 20:20:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0be6ab35294ff5228d9375ef729a8f96077c511d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 21:24:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/940efb36f22746759264aa104202a1cd : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/602127b9bd3849b982f01fc53f5fd5ea : SUCCESS in 1h 01m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b7e9306fb2144f42923efdd12f4a8231 : SUCCESS in 47m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/990525b7d5ff4191a850da5aa63fa1b0 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97d1611c4bfc4d7ba6a60e32f0d90b21 : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb2b8bf5d0644e0981451302ad6e5d32 : SUCCESS in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/912d5cc6774643a88ebb9a3791db4e63 : SUCCESS in 13m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88e49ce10f424ccd84723623a328c899 : SUCCESS in 15m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/852ba59cb808416ebc4b426084017fef : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba8586010ad7466f8a305191d3d4ca50 : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d412953552d24eb5abdef1ae9be031c3 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba87c0d576ad4448bd91dc0f46a72aa0 : SUCCESS in 19m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b173c1d83d5041e8b563dac23c9e6fbb : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67a08977cf4345129de9b03c44660c11 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1b93d373cd4f44b48da58633cf160f5b : SUCCESS in 19m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/412f61dd21eb421199ffba85c3be2d17 : SUCCESS in 24m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8e63c5350f23488ea419032442102085 : SUCCESS in 43m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f2348c7256a484aad46ec91b423a371 : SUCCESS in 44m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0860c625a75649c0877ba90939e5104d : SUCCESS in 49m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8e93d125ae44f13b6c3e7f38655cb92 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502dc0641e7443d2a21dff723d3c78f2 : SUCCESS in 8m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e85f61056c344464860f3f561baef897 : SUCCESS in 22m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39cdae761a874552a621b84ecec7adda : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":2},{"id":"aa0c5aaa82b555d6056325ade84356fca0eb5060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-15 15:12:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b2988a7f82fc46c8763716d6bea49222fc34747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:33:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7221696b81a4b2c9843687effda1bb8 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/91c3df880b6e4e4ca013133c0cd9da43 : SUCCESS in 1h 08m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2670d80123f641899dec5e059864e5fb : SUCCESS in 1h 07m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2f8025b73894991b285bdf31c0ff9f7 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6d880a812ed34a8d896cc0d4ac380673 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/922a8dcce28b420691104f5bac97f0b6 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40945957a84445729dc38f661bcf823a : SUCCESS in 17m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/572bc97a9bb34ed9b162eab8394ffdb7 : SUCCESS in 16m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f7cc7b8f967b403a8b5683d2226127ae : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3581c875d371407d8f0e08806689fbf2 : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9085a44e6ac8447a84f90c4dcc793737 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00eb88e63f8e4ec9961d17585ec70a82 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f1de2c6e6f548a9a8c77dc1ef23f027 : SUCCESS in 24m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1fac8a25e47a4200aee11e11f839ae53 : SUCCESS in 25m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/32c12aba6fb74eb3b50517c847ad07a2 : SUCCESS in 21m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a705dba131754ae59a140e20db772de8 : SUCCESS in 31m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ab1a0dd7fa44deebdc6c244399408ff : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f37bdd1138c344059a3a0b5bd1e83ba5 : FAILURE in 20m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e0f1ce60e134a41b8381fe9462e876f : SUCCESS in 50m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/cbf12ba07ce0472c9a41f85f360a4561 : SUCCESS in 47m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37dce8ef55ac4334a7d5f859c228a2c6 : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/169044f4e2994a559bcc32dd55e950c6 : SUCCESS in 45m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/164dda7e84974a889055e198e84d8340 : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7245bbc7572e42b4a77a67bb2641a2c3 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44af3eed2b184d9ca1be13d964d90bfd : SUCCESS in 31m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34955ccbd7484618b1e77b554fee8027 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":3},{"id":"4a0088568e405a40f41e894f712bfe391c7f5176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:12:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/339db84c1aa44ec190381c9774eb6c3a : SUCCESS in 36m 17s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8486a85cbcd641748804d40d0f6cd9b6 : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af5df6329d057469839390905211254aac5e98f4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-18 06:39:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6366acfa9c994dc531de84bedd704cd22096bcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 06:40:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"078c84b720426cfdcc4c7bee633b9c356e8bb336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:11:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefcc4b334b3459ca65de496c4c06d2c : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/79c74a1b72ad40ba82a573d74c52120c : SUCCESS in 1h 08m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d422d52d50c4d3a8610484e240a0520 : SUCCESS in 46m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c9c2fc831921433ab4cda93ca6089d6d : SUCCESS in 9m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dfa9387e4bc2478ea94095eb6c076471 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a676b902dda45dc9b3e658283e0d952 : SUCCESS in 16m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3783665df524892ac234576a2dfa5df : SUCCESS in 17m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f56e9ed03e194a7abe2db9a1310c78f2 : SUCCESS in 17m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c767cb9c50114ef3bee85d9f5e1be87c : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/503457060ee44a06ae67bc6d3ee8fe8f : SUCCESS in 21m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a450fd2ef7de42b2af473ea11a65e58d : SUCCESS in 22m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49a8defb6655437289657ceaf4d47468 : SUCCESS in 26m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6604da8ddf6445eb3396fc38968d6c0 : SUCCESS in 27m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a39654447fa44e9790202801187e2e58 : SUCCESS in 20m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8fcf29ef7624b908309456f72086bcc : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a43a3dcb61a94a388c4d571d226de955 : SUCCESS in 22m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5614ad0b6d84c89a7f22f305090d9fc : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8705ec8a4ee64d9e8aff68dedc136336 : SUCCESS in 49m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b905807df0a2468d800a96d9b7f629db : SUCCESS in 52m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1057dfc6ec6a48a494c4c89caf8a4f74 : SUCCESS in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a513f52a52de4820a3f5c1570745ccc9 : SUCCESS in 38m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f845e7d869884f1aab2eafc7c79a1aa6 : SUCCESS in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b080499126da4888a18199e7046aa0ae : SUCCESS in 8m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d387cb4bdb3049138b903d9d796a870f : SUCCESS in 30m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2594dcdf08e41798fbc467b81427129 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f177cbedbed3bc463210bf9994fa9deb97467968","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:11:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b2247ba7eb198a31958960ff36b36daf079f90e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:13:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed79bbe32fa4ebc9cc580696eaf575a : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/df1363cf09304193a629bc8af22777b1 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8c60e5e5cb41448baa683e3209030098 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39ad468dfee7ceac1456a67604af774528341a80","revisions":{"9c4e6f2491d33b7e664c14b24cd374cc3cacf547":{"kind":"REWORK","_number":1,"created":"2020-10-16 17:39:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/758603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/758603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/758603/1"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-16 17:38:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-16 17:38:54.000000000","tz":-420},"subject":"Add async_pending_last time to object.recon","message":"Add async_pending_last time to object.recon\n\nThe async_pending count isn\u0027t near as useful when we don\u0027t know how out\nof date it is.\n\nChange-Id: I3e5e904ffc0eba7a7e141e1c2d9f9840e4952041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c4e6f2491d33b7e664c14b24cd374cc3cacf547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c4e6f2491d33b7e664c14b24cd374cc3cacf547"}]},"branch":"refs/heads/master"},"37b6db2e979e1a75ffe94f2dd5f44c36e252ac60":{"kind":"REWORK","_number":2,"created":"2020-10-16 20:20:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/758603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/758603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/758603/2"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-16 17:38:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-16 20:20:33.000000000","tz":-420},"subject":"Add async_pending_last time to object.recon","message":"Add async_pending_last time to object.recon\n\nThe async_pending count isn\u0027t near as useful when we don\u0027t know how out\nof date it is.\n\nChange-Id: I3e5e904ffc0eba7a7e141e1c2d9f9840e4952041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37b6db2e979e1a75ffe94f2dd5f44c36e252ac60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37b6db2e979e1a75ffe94f2dd5f44c36e252ac60"}]},"branch":"refs/heads/master"},"39ad468dfee7ceac1456a67604af774528341a80":{"kind":"REWORK","_number":3,"created":"2021-06-15 15:12:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/758603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/758603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/758603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/758603/3"}}},"commit":{"parents":[{"commit":"1a5b42da9e43f4fef16ceba9eb19aec9344d7132","subject":"Merge \"Remove babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a5b42da9e43f4fef16ceba9eb19aec9344d7132"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-16 17:38:20.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-15 15:12:05.000000000","tz":-420},"subject":"Add async_pending_last time to object.recon","message":"Add async_pending_last time to object.recon\n\nThe async_pending count isn\u0027t near as useful when we don\u0027t know how out\nof date it is.\n\nChange-Id: I3e5e904ffc0eba7a7e141e1c2d9f9840e4952041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39ad468dfee7ceac1456a67604af774528341a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39ad468dfee7ceac1456a67604af774528341a80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
