)]}'
{"id":"openstack%2Fswift~759075","triplet_id":"openstack%2Fswift~master~I6a2ec3739ac92affb0f3dac248f26a1fc4ac9493","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I6a2ec3739ac92affb0f3dac248f26a1fc4ac9493","subject":"py3: Fix status comparison in ECGetResponseCollection","status":"MERGED","created":"2020-10-21 15:25:57.000000000","updated":"2021-01-15 20:54:04.000000000","submitted":"2021-01-15 20:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"759075","meta_rev_id":"c7618f19348c143a7fe4d52d4923f7f518290126","_number":759075,"virtual_id_number":759075,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 20:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-01-11 16:15:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-01-11 16:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-01-11 16:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 17:54:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-10-29 10:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 22:27:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"9b3e8f0e6f073a88d9b5be43489705968b96582d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-21 15:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fdf9ea7d5053e892cc72b54ab8975b020785d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 17:31:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62a3728d76c4009951293c8980bd4a4 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/15c0e29a2e64412dbca688114ac435b4 : RETRY_LIMIT in 23m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83204afa8ce844be8cb6bc8926ed2b7e : SUCCESS in 49m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e85e0fdbf32437db7535e0ef9d924fc : SUCCESS in 11m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/23e7528013c44bcaba87ddd2fc6051f7 : SUCCESS in 11m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36225b2aee064ef8beb98060c76add78 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/862b4ceca1954077a6b396819bfceccc : SUCCESS in 17m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65da27f9635344f1af256125658e79f1 : SUCCESS in 15m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f56db4c87534b39be324d765ce589e9 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8148f933811d4610aff4d732298d926c : SUCCESS in 36m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f9d835fdd7cc45cbbc7f0e6f24edbdfc : SUCCESS in 23m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da725390f7f04938ab16ebd9db96a9f5 : SUCCESS in 23m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/51e5e4ab36e74bac98c15fb3b6318359 : SUCCESS in 24m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d6eb4a49396642459b374ae264a66890 : FAILURE in 24m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a94bd895ba634b36ad9d63afd40bfbf4 : SUCCESS in 32m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f35587c3d6e45b19843be5854d03413 : SUCCESS in 24m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d8730f9ce7c4cdda0aeb746960add95 : SUCCESS in 50m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a2938455eb04f61a795935c411eeabc : SUCCESS in 50m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56323bfc4f7746cd89aa2267a90deb69 : SUCCESS in 42m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6769aa2f064d419f9ae71ce951554aae : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af9c99fcb294d339ca4f3a12afa0728 : SUCCESS in 8m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94321ff4b83c41a890f7a74befa3adeb : SUCCESS in 27m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d233d04ccd84e00b64925a9ac425cf8 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bba5c772de29eb9477c8a4945ad55e56f1a25116","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-21 17:54:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d06a321ca4d8c0be652c8062eda17b01dc99b90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-28 17:27:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14ae9bf36e3c96fbc6dea89ce4db1539f81cfeef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 10:37:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217d97f9353848f89c253c22ec3737cd : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3eec6f9630634bb2a85beb9dd16705a0 : SUCCESS in 1h 04m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd24f97ea4304bdebb2cd0a9acef3d4d : SUCCESS in 42m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/957ea08c29bb4c74b99ec785d04c7391 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e1c2ad7a23f4a02920597ba50973b3e : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea754903658242119a64797d2dbac0ae : SUCCESS in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/532e404d7e324f42857ecdfed12826d3 : SUCCESS in 12m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c1bab20bd8104e5c9cf6a3c391af383a : SUCCESS in 13m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2b982d73c63f41b6bb0ab5844d799f80 : SUCCESS in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b88a15545e849339eeb0d66a72c1803 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/55f6a765c7e04ffb8b9093b90222dd7d : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/feb9f6432d494f9d90da7dc1e19e0caa : SUCCESS in 22m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/52ff1ee754d84d64b46fdf039eb614e0 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/db0356fd50b842b19813b1125a4e3bee : SUCCESS in 20m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/415cf0520d864b1fb76bf2616f956a49 : SUCCESS in 18m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/598476ef367c46fe90e77165b116e05b : SUCCESS in 20m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ba4c5bf084849938b4bc54393d4582c : SUCCESS in 49m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c53dca497f4408399d835054f83b1ac : SUCCESS in 45m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9920836ad55a474daed760d16eb4e039 : SUCCESS in 47m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0574b2696d640c1b39e2e4d5d3a315a : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8daac3c0bb6e48eb808132db47bbe9a6 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a06705eaa0d47e293ea71b1db9b01f8 : SUCCESS in 21m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc1c23b153e043b0a6cde99d65c23403 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c63d56aed3fed3432645a9c1356c0de88c0bfc3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-17 22:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d9d8851d9a2cc5ff3b7a19f87b78e187f46a59b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-17 23:47:55.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThanks for the test! Let me know what you think about an alternate approach (inline)...","accounts_in_message":[],"_revision_number":2},{"id":"5c161ae7d9c61578700e1733634d5dbc1a147829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 00:31:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcb6acfff1748ffb325d0642a31fe63 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2b86d232632a4baea74576b0de7db7eb : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dec6248aa294ce8b6cf45f20802c35d : SUCCESS in 48m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b1d645159eb4e98b64f90e2fb5a51f5 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/119d81d659904f65b20b23a974081302 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00ac01cac87045e9ab75ebd62fafe002 : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ebe9688ce54862ac3b7c7b67c7c0df : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a6e47f313af6455489587e053dda0298 : SUCCESS in 17m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c006bc3f7bc44e2a86badcf86c98d6f : SUCCESS in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7381b978c76b4d4dacb25e79298b49bd : FAILURE in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d26e3961dab24cbf910bcdd6fedb97a0 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea1a986f492f4ab789a9cbe966b9097c : SUCCESS in 19m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a7df208045af4038a71bc237e288fdfb : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/26d6ae4412544b00a384d683fab49ebe : SUCCESS in 17m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/872b0855f59544f5b2274b70e0dea1ff : SUCCESS in 19m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49f881b405c94e779835e50d8e31cf5a : POST_FAILURE in 6m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/21891463257247cb83b189fd089003c7 : SUCCESS in 54m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b6862771f6a47d99222b0c9f6967b89 : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6552e8f944204b9caacc375fff9282ee : SUCCESS in 44m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/646bef0549f74968ad05b8a7d8399968 : TIMED_OUT in 41m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd6062db8bf4065979c615c164cebe5 : SUCCESS in 11m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c1b6b267a5f4326b180fa3685589a53 : FAILURE in 7m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4266d0fd717140f5bbac4c396b0eef82 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ed5ab4695c891cb2a6a98340863f8eda248e88ba","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-18 09:14:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85871ea94635e86c48bf425435cbd98f5e02d6b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-09 00:57:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba80473a9426a9bfe71d75c21a3dddcebe621c7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-09 00:57:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"026947fd87a10689898256a091c866b4967d92c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 01:59:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7fdc7712634fe0a0b9978102fce5e4 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/f86a9ccea66f445cbca681b5ecff9bd8 : SUCCESS in 1h 00m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f95408015804b179ad1a07d925a5e27 : SUCCESS in 45m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff672aba86c346f1900a40dccd9c29a7 : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/078728ce451c4fabaecd5d10eb572b03 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d6434e8e7ed43e394646ab7c998d61e : SUCCESS in 13m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16870ffeb73445cfb44cfc3d82629081 : SUCCESS in 13m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/140921e66e614c62a3c62d3ecc305038 : SUCCESS in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8a879b4f7372434a98c6d0e9eeb961ed : SUCCESS in 18m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abb57686ec3c40ddb58aa629667aeeb0 : SUCCESS in 25m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66731a0230684160901780101db7b650 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aee9faad41394d7185af383183a965f9 : SUCCESS in 17m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3bb8a40a710344168eb1511ac910c40f : SUCCESS in 28m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/041f52ef5d0c400f8c16ae6695f48c29 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd15ff7289c54fa78b22ada377cf5582 : SUCCESS in 18m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a95a30541c4c48cebffbb473665acd0a : SUCCESS in 33m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6483a6f92f094694b3a991c2295ed25e : SUCCESS in 39m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2650f82220284e67a96213e185028b42 : SUCCESS in 39m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26ec5fdfa20f4f2ea13b16b8259b3cd0 : SUCCESS in 41m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40e1096bb27f406d973b49783144e68b : SUCCESS in 1h 00m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a38feccd12fb4e428ca657449bfed72d : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e56901ea5d4437baa24ba15bb074283 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf1896fce6b5468a9b2b6f1903e446a8 : FAILURE in 10m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5bcf7b6c1ce4c45adbcfb011e387bc5 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":3},{"id":"d06d1ff58e7641ac3cf4b2df115cc7daa7f8818c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-11 16:15:40.000000000","message":"Patch Set 3: Code-Review+2\n\nmaybe another core can approve given my level of involvement in this","accounts_in_message":[],"_revision_number":3},{"id":"a6bda5b544facd120ab22842efe584c87a3b089a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-11 16:38:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b09693fa40d0bb9676413f5e67d35271d58c9d16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:38:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"572fab814205f3cd7887e465bab143e194936308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 18:19:00.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03584f79129540f7a52aba7efdfe07a0 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/28cb43a51d7548f39cc10b1f83450b2a : SUCCESS in 1h 13m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69b3dccd6a82489dbb256686f0badfad : SUCCESS in 45m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e078548ebf3b4276bcd7f4128ce295f5 : SUCCESS in 10m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/55eaf381f8bb42b2b01fc15237ad840b : SUCCESS in 11m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5500eae92ee64acab62a454129ceda39 : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01a78906eba34c859cdee41abab33d0e : SUCCESS in 13m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e866beec71c64019ac1f8511ff9fab51 : SUCCESS in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f37a67415719405e82bfe1d7bb4d2dd0 : SUCCESS in 12m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdd48c4aeedf4330bf3c10feb275eacf : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb51a71436fe401e97b8a478533c5237 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c4cd6db07e146c893da1f779c760d52 : SUCCESS in 16m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/34990df80ab84189a088da1018570a0f : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b310954211004b0d815f4ee6bb800272 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2de051f2a8404e6e81eb9217b645146c : SUCCESS in 20m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/11b8015de5d24febbfbcd71e35dd2ad4 : SUCCESS in 52m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/37e609a598b941bead7aace9e8e43e19 : SUCCESS in 1h 00m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c25975aa25ed4cbaa4f7f21afa5f2c7c : TIMED_OUT in 1h 33m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fde655f5b5144f868d777dbcf53c7a20 : SUCCESS in 39m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ae9ee0f957f41b1b2b6196deeca1489 : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc5c9b227a64d40992fd24e670aa905 : SUCCESS in 5m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3afb3159c37d4264a1eff2c0d0b99189 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":3},{"id":"aa070165eeca65e2e642a8b14f443b4d33126f13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-11 22:33:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e45e36a67b279e785b45705036b41b9c01f7aab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 01:17:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6236bf8696234b3c9847fd5ef1f77ce8 : SUCCESS in 20m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/c090f3b6e65e47f4b5adb40f09be0b1d : SUCCESS in 1h 11m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9a2429cac6d44b79425998a1c19de33 : SUCCESS in 45m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b287426d2e341c98a4ce3a4223b1e59 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b4417745d704f61b12723603027de5b : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/280c4b7b50564e728d9df75e4db50ef4 : SUCCESS in 12m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/610aeeb882dd40b0b7a927d6d51adda6 : SUCCESS in 11m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5189e88adea942a0a5e7fe8c57b13797 : SUCCESS in 13m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d44ea134ed6e41d281aa0a4d3f74557e : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5afde1f92e7d4325ae775306da54aa9a : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fc83879ab4245569edb12571248b6a8 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0879f876d8014c3390697d39ac54b189 : SUCCESS in 17m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2681e3e5918441b78ab1a346fba39d73 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4ef5c350634740138149f3c388cee804 : SUCCESS in 21m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c9e82698c14a461fad48f5e4b836684a : SUCCESS in 16m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/14a16f3d600f46a8b2703e3c5a0364c6 : SUCCESS in 21m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/daa0eba77ee048fcba67e88f07e23b03 : SUCCESS in 1h 07m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d2206ef2172c4c3e9bea2a410a8898f3 : TIMED_OUT in 2h 00m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec009e634f88457f82dd69a430785fdf : SUCCESS in 39m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0765f27117e4844abb946dec928aad9 : SUCCESS in 39m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5e488fc01384e7d9ac9a4a04b9095dc : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f35960fd0d4da6937451b9470bf7cc : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f505b12bf564f37a26ebe659d2a0ae9 : FAILURE in 8m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a08e8c309cc146eca7d4f8f59fa7f577 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8d901aa434de392c6614cdce52887ef05278766f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-12 10:50:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4a25a645b6896c6f0e147baf1e1c83e131ead7b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 17:24:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70b3c71854d44f38cfb21c4b6ec554e : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/24fbf578b4da4a25957186795d1bde02 : SUCCESS in 1h 07m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f34de97364694ea39a8a56702a217d05 : FAILURE in 13m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f17f4766d15e409a89f552ea99de8381 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/440bb52d2b39437090ee77947a521ac6 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1ff24cf0d0349a782dee28f97eb853b : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4acef42df5d48e0937ce3d744bb2926 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f656ca4b51e4d589b8ceda5d81e4899 : FAILURE in 5m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd888e4209624b5aa20c10b06df89fdd : SUCCESS in 12m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ebfaf304f94e444caae166ad7db9cb2a : FAILURE in 4m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44396f5a72604eb1b405f5399fe2af59 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87e2e61ebd464458901763157b281076 : FAILURE in 3m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5cf2a9aee8fd4ff3bee3e7b324b0ac49 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/34e112d9481b467c95ae962de31bb6de : SUCCESS in 16m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c8f84c99b5734acf82b2e367adb545c9 : SUCCESS in 17m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89cb9d4f73d64907b4d118acd1352bcc : SUCCESS in 19m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/326917aeaea34e33a68e0dc7b0d846df : SUCCESS in 43m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1dd3cbb4e0874d69879db92f33057894 : SUCCESS in 41m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f20da59080b4f30a37dd9e06811aaea : SUCCESS in 38m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5de7ae8e0794a9f8f857919e2e2c762 : SUCCESS in 37m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4758a4c0aaee4e379f6805bbfeaa481d : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e11e65c84ce40bda3f16288edfa9e90 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/524ee8bb289d465bbb7b005d735ba9ea : FAILURE in 8m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a86076a4c7e48968bddf8760335529d : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":3},{"id":"14277aefe6ea1ee7ad116abe269d2491de35d1ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-12 17:29:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"09e73a73276190f3881d7a58c705203aab7375a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 05:18:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf61669260d49a49248c6234a3a43fa : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/d6a039da8984450db07472e6be3ef050 : SUCCESS in 1h 19m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cd684c667df4420b1a37fbdf0fe5409 : SUCCESS in 57m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/975627626d0645e4939795aae9b4ae4f : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/857b8fe697f44c7f9ad54c7b3b90a386 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cc48a9fb0fb4a75b4efac5bf7cb099d : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33e2668683ad46c09b6fd9fb3275f262 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ccd0dbb0e55649b0821feb39cfb55d8c : SUCCESS in 21m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba53ef1440c44880a5396d749edce27e : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9d6872d1424430c994f3a2cd9f919d2 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/344404e87ab34f71b9e142ba9efa2131 : FAILURE in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ef87f7b772547c48910b17370c96b65 : SUCCESS in 21m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d150fc377557412e89e8dcf350376144 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4df75470f6314f6a9af26cc0c3bfa7c0 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ea43e85e06cc4429bea3af7cae30dd4f : SUCCESS in 18m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5de2017541b14960a92d9738215c6407 : FAILURE in 31m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9f054ed27c374ade9ed69dd7db61842c : TIMED_OUT in 2h 01m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a478d759d8b84f84bc012f2d14b21525 : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c7b2546667647debb112ae73ae4e712 : SUCCESS in 44m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de3c178bcc0a4e35a6239231ad96084a : SUCCESS in 47m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b6830d30bd34564a8c69f84881d125c : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44472d5128f340c6b1e7cc1d5d2543c3 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52fadd4d19d744118b7de644cb3f1916 : FAILURE in 7m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6ec75941e044f09ca810b585549f4d : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":3},{"id":"db7d260db6b27599e5e4e404a26778a53b976c81","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-13 22:45:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c17c59a1ebf46d7471df613ac0f2069528c15d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:42:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ecf705c26d4648ba83394d26a91aaf : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0304018bf1714889bb86e89efe6e5a0b : SUCCESS in 57m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0382409bb0634c2380f05f09697c8996 : SUCCESS in 54m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff594953170a485fab44490353472fd0 : SUCCESS in 8m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d52ebaf1b5c4d19bc48178a11f7ca02 : SUCCESS in 9m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d8802bbccf0f437ebbdf403bff86303d : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0717b5a0bc484991b542559ac0a4a777 : SUCCESS in 13m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dbcc752360e44e4f8bdd2d1e8c8364b6 : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a54da86a31f5446bb438fb90293d4ffd : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9aa5b6b6cfad4eb98e7eb05c32bb25cc : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b14a8a7e9d14269b1168059f66bd8ce : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc39183fe292411eb775502e2ae70809 : SUCCESS in 21m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/acea41f061e8475dbac51b2d64122a1b : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a4f466da7ade4717bd8e379f355ccc45 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f0d0481c48c740f6b46ad22fb9373daf : SUCCESS in 18m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75b2e1cfffec43dd9584ee201aca774a : SUCCESS in 26m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5e429ef0562a468b8ddc6a596d35c204 : SUCCESS in 42m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d2e385810b7745938f2076c3dc4839a8 : SUCCESS in 51m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11b5a2112390473ab29c6faa3e4d9bc7 : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60c1cc06c3c74843bf387b88cad2f1f0 : SUCCESS in 53m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e60a22faac9447ca878d2af67faf3bf : SUCCESS in 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692966725051443cbd7c67f2ad5e28c6 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49d3adca1c854e919f0ac9398966d0e3 : FAILURE in 7m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96d689cbf337495f9250754be6c622ee : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8e6b7ec9988d1968b79fa12e565d77136c0b730a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:43:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1659278cff6553450061f8e9563286f79e40ba5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:32:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ff22889140419fb439f551dd15e77a : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/7bec0d24a79743f3bd4e078beeee64cf : SUCCESS in 56m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6609bdb8174146818a6584f896534124 : SUCCESS in 51m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4bb69b1ce6ab48bf825a012a48a1a7ed : SUCCESS in 9m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bdc7a7453e23423492e45cc0a998dbb8 : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6a547d88327441f821a8ca3c7e6d65d : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2392b604e22c4f9d8f7bc6401654c712 : SUCCESS in 16m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/76e855ad546e4038addefe472e71cf7f : SUCCESS in 14m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a2534b2bf854daca2ff5b2bd8111aec : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fdf7df64c4d4d479ddf9423099664a6 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cd0b7f33e78f44a09fcea58edac8156a : FAILURE in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9ed9d50f1ef4c87aec87b56ac4a0f73 : SUCCESS in 18m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2b8a59b6f0b141e3b4336c537cf2b07b : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/645cd20424c24b09be378d5d7548ec25 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/49cb2562dc524345aeed7052af1f4381 : FAILURE in 18m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9939490bdb824c4bac81a8c2fa1749cd : SUCCESS in 51m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/582b753e56db47409dcb9bac3671555a : SUCCESS in 54m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a54ef806efbe4c2fbbaefac218ad6fab : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14ec9a79e8f14bd0967bdc79e40156d0 : SUCCESS in 1h 00m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2589b8d76e67468b90b66eba465dabbc : SUCCESS in 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd5904e0d2364b9b9db08dea77730ae5 : SUCCESS in 6m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b188aab5cdda404eb399d3104edc3e71 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7e0c757ffaf539679af7c125ebab7e9a07e0e139","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-15 01:05:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bd2e6ec04dac9f0a2f39af389a9235c2f25afc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:42:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea2fca05cf04621a7c00bce36db35c6 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/24319e7de29442628f394276093f5f26 : SUCCESS in 55m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9788451187ef4cf587deee8b3c553c3c : SUCCESS in 50m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/877403b06b9145e083dbaafc3e311f64 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5ea38dfb5804fdebf9f9e11e06af27c : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71e758fc0cc044e891ddbd9286cd89dd : SUCCESS in 13m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/085bd12f0ef34955b0fa0b993377b2bf : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/76070f80844e4db795e567aee1c2e10c : SUCCESS in 14m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0066f4e1925f42cf8b5f48eeaefa4faa : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14a19e6bde604e9e8c54b2b69191596c : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a580034f17a4b47b09a2b468dd6adc2 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a47076e6a9f643619eba67c792fc51b9 : SUCCESS in 18m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/85353180523d40b2a3cad65a0ec5a50e : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88393acf542d457f873349009d537b8e : SUCCESS in 21m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ffa8a295553148cab4a07875a6b17fd4 : SUCCESS in 20m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8a0007a225c437199ff3f355d7327c2 : SUCCESS in 22m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d49728ed7bbc4834b4a793de05c488f8 : SUCCESS in 47m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/275a67307710459085d13db6196e31aa : SUCCESS in 59m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d709f1d061c54cbaa886a840c1290f87 : SUCCESS in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74f8b0db487c4881ad6ffa19bc317a62 : SUCCESS in 43m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce16629d912406ab08816251704e334 : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aaa6ee52ea3499a965d7c0f7eb5cf77 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55850a87ff4242408a0365d4ee924a58 : FAILURE in 9m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/730375e0de4648008ef493d504e72b98 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e14f1a1db63eeb3f87cbb60131367d4a79a9fc43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:42:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b7f187782c6c9079ef3713cef7a16cbee569ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:49:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d713aca76f4f7cb0855c3cfe13eebb : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/3a8166920af34d9eb3273071324b45d3 : SUCCESS in 57m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9fad57cf4a9468b8a7ec759ad7fbb82 : SUCCESS in 54m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/895cce899dfc4e229a4967080481473b : SUCCESS in 10m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/90bcc5d65d364539954a7603d745cabe : SUCCESS in 10m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a76b8659d7e54d96995399cc549d8739 : SUCCESS in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88925fc5eedb49bc83edea89479f6b18 : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1698c3297f3c42e8b6eec97ecb4041b5 : SUCCESS in 23m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7aac74b23013471d9401d5cee8d63a41 : SUCCESS in 15m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7766da9152ba4ca089bfee4df673102d : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b08a6070b5845408cd058168b659484 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/310b571104d84c649a61e709323138a3 : SUCCESS in 18m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1521328642fd4d14b0a006bcd01d0d8b : SUCCESS in 27m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4aab0d6599414b0091e8beba0983c4b4 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd0ace63ea1c4c1da21b9f3745804026 : SUCCESS in 22m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad1db50f9cac41f3968457342998f71f : SUCCESS in 44m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0de60457574b43228a0bc63b5e5651e8 : SUCCESS in 44m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd225883dc0a45eb8847b7984106760b : SUCCESS in 40m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c05ee8220074df9a4c18ce1bd77df5f : SUCCESS in 46m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fc1778b4d6341f9bf9010b9d698b12b : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a76a00773e0417cb60253a0f216717b : SUCCESS in 9m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7514ad1cf5fc410ea48daca93a1c4aa7 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":3},{"id":"87bf4b733b1886d7acfec613182a922f4508de8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:49:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c7618f19348c143a7fe4d52d4923f7f518290126","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:54:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3888c779b7d54fe296fd8da578174b93 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8e3d3d9e0e5a4d7ab65531887485671a : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f06c5585305e4114b658b994d9b894f0 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a5fa3cfcafbc6a441139b6e9bccc06bf38042d64","revisions":{"c993b8090513424d50ce59d14d2dc01100f90cfe":{"kind":"REWORK","_number":1,"created":"2020-10-21 15:25:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/759075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/759075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/759075/1"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-21 15:25:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-21 15:25:48.000000000","tz":-420},"subject":"py3: Fix status comparison","message":"py3: Fix status comparison\n\nChange-Id: I6a2ec3739ac92affb0f3dac248f26a1fc4ac9493\nCloses-Bug: #1900770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c993b8090513424d50ce59d14d2dc01100f90cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c993b8090513424d50ce59d14d2dc01100f90cfe"}]},"branch":"refs/heads/master"},"c73606c7315bc200fdeeac04aba8c25f07296710":{"kind":"REWORK","_number":2,"created":"2020-12-17 22:27:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/759075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/759075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/759075/2"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-21 15:25:48.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-12-17 22:27:22.000000000","tz":0},"subject":"py3: Fix status comparison in ECGetResponseCollection","message":"py3: Fix status comparison in ECGetResponseCollection\n\nThe initial dummy bad bucket was previously stored using None as a\nkey. In some circumstances this None key might be compared with\ninteger response status, which is unsupported in py3. The value of the\nkey is insignificant as long as it is distinct from any response\nstatus code, so this patch changes it to 0.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I6a2ec3739ac92affb0f3dac248f26a1fc4ac9493\nCloses-Bug: #1900770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c73606c7315bc200fdeeac04aba8c25f07296710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c73606c7315bc200fdeeac04aba8c25f07296710"}]},"branch":"refs/heads/master"},"a5fa3cfcafbc6a441139b6e9bccc06bf38042d64":{"kind":"REWORK","_number":3,"created":"2021-01-09 00:57:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/759075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/759075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/759075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/759075/3"}}},"commit":{"parents":[{"commit":"127bf9707c383ac737fce9f243bcc8f6655910df","subject":"Merge \"New proxy logging field for wire status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/127bf9707c383ac737fce9f243bcc8f6655910df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-21 15:25:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-09 00:54:53.000000000","tz":-480},"subject":"py3: Fix status comparison in ECGetResponseCollection","message":"py3: Fix status comparison in ECGetResponseCollection\n\nThe initial dummy bad bucket was previously stored using None as a\nkey. In some circumstances this None key might be compared with\ninteger response status, which is unsupported in py3.\n\nMove away from having any sort of default value and instead return a\ndefault dummy bucket if we don\u0027t have any responses or only have 404\nresponses.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I6a2ec3739ac92affb0f3dac248f26a1fc4ac9493\nCloses-Bug: #1900770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fa3cfcafbc6a441139b6e9bccc06bf38042d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fa3cfcafbc6a441139b6e9bccc06bf38042d64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
