)]}'
{"id":"openstack%2Fswift~759183","triplet_id":"openstack%2Fswift~master~Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3","subject":"memcache: Refuse to error limit the last available server","status":"ABANDONED","created":"2020-10-22 05:05:04.000000000","updated":"2020-11-09 20:26:30.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dfcbd470bf09c7027bf4cce3d9ef70cf3c06623","_number":759183,"virtual_id_number":759183,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-22 18:12:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-23 19:06:06.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2020-10-26 06:41:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2020-10-23 12:59:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 18:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 12:59:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2020-10-26 06:41:55.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"fa45c5ecdb8c657b2a096d8ff8c0da4fcb15caed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-22 05:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2769445edb5b950020b38521abba72fe52293fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 06:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d30f1bb6c64a0296fd11a032ed8d53 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/d938e933b2844910becd0271c14b31b7 : SUCCESS in 1h 02m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92eba84b3bed4f69b175b5ff077c64ba : SUCCESS in 49m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7950ae092c049ec9f44a5f79dd38c84 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d0b18792510941829b5629d7745d08e3 : SUCCESS in 11m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2c9d96b82eda4d9192bf761ec5d383b2 : SUCCESS in 43m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d92be6d644804e9b832f104320caab46 : SUCCESS in 30m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/210016e54412491293ea4d670c01c212 : SUCCESS in 30m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af18e4f115534e5b8b65e18567f4de1c : SUCCESS in 29m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eaecf83fa27c444389e5cedf078c5eee : SUCCESS in 1h 00m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dd3780b0bae430dbe0648e9472b272e : SUCCESS in 57m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe48561a32d0493a9cfde3f170de6b23 : SUCCESS in 1h 01m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6adc3a80e9d7407f8caf50de7e8d3112 : SUCCESS in 59m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/082b591be5724aa1959bd9985454cc45 : TIMED_OUT in 1h 02m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/86f88ed11055445e9d30563cf527bdd6 : TIMED_OUT in 1h 02m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/befdb2d8da9645e383b6aa1ce08c544a : TIMED_OUT in 1h 02m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ccd6e5c793a047bbb8cd003e7bf839bd : SUCCESS in 45m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30e8a1b327214ad9a58478bb1f376015 : TIMED_OUT in 1h 04m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9bc0071ae9924ed3b1c7bbeda371fa0d : TIMED_OUT in 1h 04m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d780c31c409f4103bee54a7da0890b4f : TIMED_OUT in 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133aed4525c046a2af0e6314e2cddc2f : SUCCESS in 9m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a735e4ae8564cb2a1d2a3d77c29b1ae : SUCCESS in 38m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd29bd29ca1747e896decf3695d8b7a8 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bb2516e0f7ebdc3155f48403ef6a776e6b23d357","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-22 15:04:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86d022d86f5d9ea687a7322fc718f74717d97e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 18:12:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52dcf5d07e234b6c84cf78edd11abc20 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/716ee82ce96d41a5b5df514af77f4a22 : SUCCESS in 1h 07m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b6ef4daa8a94120bb54de747d62ff73 : SUCCESS in 41m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5841d4a5a1b347fcaf9ebb06832cd72f : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/523658169f994b218301ea7a3d3eb28f : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6e7810ef80d40108a6aaa6ce3b0ada2 : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b617568f6ef746cda1164132950eeb32 : SUCCESS in 25m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/143ba430567d493a96a103aa92411d41 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/35e6848b731743f7a5f04a6a601f1f6e : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/373117c5dcbe4d82bbc0c3750c92d67a : SUCCESS in 26m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88d1f1c1c8eb4f57aff1814a6b837f5b : SUCCESS in 28m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab885da459b64232b9787eaae064085d : SUCCESS in 27m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/be51b33d1c4d4ea89e729cee9bfc0a01 : SUCCESS in 24m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bbb23bfcfc1847d3a3c67f8eb971804e : SUCCESS in 22m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/37778d31ac26451291d6f6cdc46c6f73 : SUCCESS in 24m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dad36ba1c12841258190a84062c3b52f : SUCCESS in 24m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/283f7a26dc0b4623bf05245993ee3938 : SUCCESS in 49m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b13dc60d2c3147ccb25d876910bf06e4 : SUCCESS in 42m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ede6f8a031741e0b51ebc9e6c5fa3dd : SUCCESS in 48m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d37727850f084eeda9243615e87947e8 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd81a0908c2949659568305977ebbb5a : SUCCESS in 7m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f4b2b79aefe4c8baa7d733970b433a1 : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/950a75fa67584b6bac9ceacfa101049f : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bac7954d1f4b4dd2f9c4e8a080284f5c6561abd4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-23 12:59:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)\n\nThis looks like it does what it claims to do.\n\nI wonder about an alternative strategy for keeping one server always available (see comments).\n\nI also wonder if this is what we\u0027d (always) want to happen - do we want to always (potentially) timeout on memcached? And should this new behaviour be a configurable option?","accounts_in_message":[],"_revision_number":2},{"id":"5b07dcfc9e8f092bbfcf2491e74e87d8debf655f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-23 19:06:06.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks again for all the follow-ups! I think this is gonna look great once all\u0027s said and done.","accounts_in_message":[],"_revision_number":2},{"id":"a6167bee39a459a10ebb27a40d437bba3abea7e9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-10-26 06:41:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7dfcbd470bf09c7027bf4cce3d9ef70cf3c06623","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-09 20:26:30.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/761029/ sees sufficient for my needs.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3fde7677f7a0e11b367adb19d91c737290e98852","revisions":{"96e4a6189c1c91774dc6a438677096e12fe27ca6":{"kind":"REWORK","_number":1,"created":"2020-10-22 05:05:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/759183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/759183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/759183/1"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-22 05:03:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-22 05:03:36.000000000","tz":-420},"subject":"memcache: Refuse to error limit the last available server","message":"memcache: Refuse to error limit the last available server\n\nThe old behavior could be especially problematic if you\u0027re using a\nmemcache proxy like https://github.com/facebook/mcrouter to reduce the\nnumber of memcache connections you\u0027re making; each proxy may well only\nhave *one* memcache server configured!\n\nChange-Id: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e4a6189c1c91774dc6a438677096e12fe27ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e4a6189c1c91774dc6a438677096e12fe27ca6"}]},"branch":"refs/heads/master"},"3fde7677f7a0e11b367adb19d91c737290e98852":{"kind":"REWORK","_number":2,"created":"2020-10-22 15:04:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/759183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/759183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/759183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/759183/2"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-22 05:03:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-22 15:04:11.000000000","tz":-420},"subject":"memcache: Refuse to error limit the last available server","message":"memcache: Refuse to error limit the last available server\n\nThe old behavior could be especially problematic if you\u0027re using a\nmemcache proxy like https://github.com/facebook/mcrouter to reduce the\nnumber of memcache connections you\u0027re making; each proxy may well only\nhave *one* memcache server configured!\n\nChange-Id: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fde7677f7a0e11b367adb19d91c737290e98852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fde7677f7a0e11b367adb19d91c737290e98852"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
