)]}'
{"id":"openstack%2Fswift~759438","triplet_id":"openstack%2Fswift~master~If6c91d8169602b60e9a6b9dd9e2d93a4e246ad41","project":"openstack/swift","branch":"master","topic":"p-memcache-error-limit-strategy","hashtags":[],"change_id":"If6c91d8169602b60e9a6b9dd9e2d93a4e246ad41","subject":"Add unit test for memcache error limiting with single server","status":"ABANDONED","created":"2020-10-23 14:45:48.000000000","updated":"2020-11-06 10:41:19.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a569fb280a743f7d7ec5eb1bede3a0c43391523","_number":759438,"virtual_id_number":759438,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-26 13:16:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 18:36:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-26 13:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82f6ea86f04226b5a06d8a6730ce59f46a66021a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-23 14:45:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0802c687cd7f429ed227d5811b6e2d3b99134a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:22:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b550adf41424c5b8b3d91ffe425852c : SUCCESS in 9m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/140dd0fb168a4451981a56a020438a53 : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d96ef34d8ed43ac810ce2bf4bb42236 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58a5ba4744af44a1898a1e87506a4a7d : SUCCESS in 13m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5a0a2e8c26d4c278ec9fc07d6c8ed41 : SUCCESS in 13m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e23d5f86abab4f7bbdec06d175f24e7f : SUCCESS in 15m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/26a21f5e1f8e407bbd1e94eeec185051 : SUCCESS in 14m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0150c6bb02854f79aa20d0bd6a3d4366 : SUCCESS in 22m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0edc000b25d4bd688e8851762edc0d9 : SUCCESS in 21m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f87e946e83f42febf19b9eb9a174cab : SUCCESS in 23m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f1c8f9956be9411b929bf5354437af14 : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a6a4167e7b2b4cfe988dc763d3f86279 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fbc434c0064d4f938e4f8fa7cfda69f3 : SUCCESS in 23m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57b59b5d4f6744d6af5f3ffe62b3fa6b : SUCCESS in 47m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7f69e834b574c9f85e256c7ac80e153 : SUCCESS in 51m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab888a9df5c456d97ad63e5597fb3d5 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151f765da8ad41ee890b4df875788455 : SUCCESS in 7m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c30cac44bff45cc9f126258cea6bffc : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"aa7dadb707b09f2100cb54409ad732195d4cc426","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-23 18:36:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nYeah, this ought to get squashed in, thanks! Really, it\u0027s the only case I actually care about working properly :P","accounts_in_message":[],"_revision_number":1},{"id":"28b6ce5cc4bbda76916051239fd52af1f45c7063","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-26 10:59:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4770d782a01af9dd88d7dd3278e2fdd001dcb14","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-26 12:28:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80a4210b9aa9076ec0b6fc8ba509e05f6f213073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:16:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14685c104224ec79ac62ef70726ad71 : SUCCESS in 8m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/665caa02fe5b4549846e2d39aba379e1 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87c62f62b5d049b9a5b993ec25ab5e47 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8470ceebb6a145efb7a1b16be1f99bd8 : FAILURE in 3m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9a16fdd934545e48f9e0f76e04a217b : SUCCESS in 29m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32a5672694194d4b84d27d9443f7adc2 : SUCCESS in 13m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/de1d5ed2058c4cbda55b5f2c12d6b6d1 : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0feef98b92394df5aa6007594ffa3b52 : FAILURE in 3m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2293b8b4fc44003a15e9751b628efb2 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3cfdb2a307b4e2888043fc94b5feaeb : FAILURE in 3m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3ef8c95de3514e38bbfee022aa5abf3d : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/17e85fda00e8406da2147e8bce9e764a : SUCCESS in 16m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ac9a42b35ac547a6a7e81b602b17ceeb : SUCCESS in 17m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75d2eb1aba084cf29ffb67b06a3898ec : FAILURE in 15m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13a63bf3044244f19d60c72e02e6d86c : SUCCESS in 39m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c76b5085b68843759161eefd931e8ca0 : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895c66b8cd1d4d928afe13cd120cae2e : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a4ea5676de14176b5e30ed376d101e5 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":2},{"id":"58684ab6ec2fbc271a1abef5a4506768a20f287c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-06 10:41:19.000000000","message":"Abandoned\n\nsuperseded by https://review.opendev.org/#/c/761029/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"247108a4b87d94d7caaed213d2a22c5d0f120a8d","revisions":{"217d2e6b53d25bc7add055539f542eb4211d13e2":{"kind":"REWORK","_number":1,"created":"2020-10-23 14:45:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/759438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/759438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/759438/1"}}},"commit":{"parents":[{"commit":"3fde7677f7a0e11b367adb19d91c737290e98852","subject":"memcache: Refuse to error limit the last available server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fde7677f7a0e11b367adb19d91c737290e98852"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 13:08:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 13:37:17.000000000","tz":60},"subject":"Add unit test for memcache error limiting with single server","message":"Add unit test for memcache error limiting with single server\n\nAdds a unit test to verify that when MemcacheRing is configured with a\nsingle server, that server is not error limited.\n\nChange-Id: If6c91d8169602b60e9a6b9dd9e2d93a4e246ad41\nRelated-Change: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/217d2e6b53d25bc7add055539f542eb4211d13e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/217d2e6b53d25bc7add055539f542eb4211d13e2"}]},"branch":"refs/heads/master"},"247108a4b87d94d7caaed213d2a22c5d0f120a8d":{"kind":"REWORK","_number":2,"created":"2020-10-26 12:28:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/759438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/759438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/759438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/759438/2"}}},"commit":{"parents":[{"commit":"3fde7677f7a0e11b367adb19d91c737290e98852","subject":"memcache: Refuse to error limit the last available server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fde7677f7a0e11b367adb19d91c737290e98852"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 13:08:16.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-26 10:58:30.000000000","tz":0},"subject":"Add unit test for memcache error limiting with single server","message":"Add unit test for memcache error limiting with single server\n\nAdds a unit test to verify that when MemcacheRing is configured with a\nsingle server, that server is not error limited.\n\nChange-Id: If6c91d8169602b60e9a6b9dd9e2d93a4e246ad41\nRelated-Change: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/247108a4b87d94d7caaed213d2a22c5d0f120a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/247108a4b87d94d7caaed213d2a22c5d0f120a8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
