)]}'
{"id":"openstack%2Fswift~759493","triplet_id":"openstack%2Fswift~master~I2e10076b7e0741e2c45395439bd2e6d2f8a6025b","project":"openstack/swift","branch":"master","topic":"p-memcache-error-limit-strategy","hashtags":[],"change_id":"I2e10076b7e0741e2c45395439bd2e6d2f8a6025b","subject":"memcache: make minimum number of servers configurable","status":"ABANDONED","created":"2020-10-23 17:32:45.000000000","updated":"2020-11-06 10:40:43.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bef21f1ae8aee785c1db00c876130d74b3e3ad13","_number":759493,"virtual_id_number":759493,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-26 13:47:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 18:45:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-10-26 13:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ab7c8fd656d285ccffe31727ce825d1dcca5f7a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-23 17:32:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89619c3583abbc6ee2331d5f610516e87ab332ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-23 18:45:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nSeems like a reasonable and intuitive way to configure it. Let\u0027s shop around the default value at the PTG next week; I\u0027m torn about whether to preserve the old behavior by default or have it default to one to at least *try* to keep us from overwhelming db servers if memcached goes sideways.","accounts_in_message":[],"_revision_number":1},{"id":"957e8880980423407619edd311b74005480c98e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 19:52:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcb0234ac8341288b0f95e3692d3028 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f39fb5998a2d4377994f8a6430f581c7 : FAILURE in 54m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/180e269cd97d42d49b4a6aca31fd3016 : SUCCESS in 52m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c83d21798c664df9929bb17a66dfd5a8 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4a936f1a77174138a0e7287ea6e14d03 : SUCCESS in 11m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b330737544ff4cffad8d8efdcaac06dc : SUCCESS in 14m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0537d56188a5478e9b484fb3f13eba57 : SUCCESS in 15m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06714edd1f23460bbfba5b88d7bb7784 : SUCCESS in 15m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fd86b39dc9e74566890c55f45d29b023 : SUCCESS in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dea4620ae5d54be7b7a71c26708f4750 : SUCCESS in 23m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3cc9d8f5d0d41b6b13c7f1c86a09b1e : SUCCESS in 24m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9cf8edda75894b55b05130911f0c4fbb : SUCCESS in 19m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6157b2385f7b4a7b914a6fe47a40f068 : SUCCESS in 26m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3500e39459a44cbcb5969a27db85fb05 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/64b6978ac1a540adba200430eddb194e : SUCCESS in 27m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d2fd4ec04994bf284bc502e19b246e1 : SUCCESS in 25m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4290bc01d26b4f6aa58831f97fbad599 : SUCCESS in 43m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/420f9106ae924d25b8936304cf458bbe : SUCCESS in 46m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dadc3d98cea45b8b0205afb28d069d1 : SUCCESS in 54m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5905bba8586f4008b0202b5db0730c02 : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9747650ec9374cf5ae87397955ae5bd8 : FAILURE in 7m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cead77cbea0e4c0f9d95af2f140a7857 : SUCCESS in 25m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15f2f7a2e4404326883326e3a6f10daa : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2974c4d48b238717390761aebeff06c32a804271","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-26 12:28:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90f7a00ecaa8546edfe23039b03e4250cd26e403","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-26 12:30:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b8041d7a4532e5033d983921bed7abcb0e462179","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-10-26 12:33:35.000000000","message":"Patch Set 1:\n\nI\u0027m not sure how useful, or even desirable, it is to set minimum_servers \u003e 1. That can result in one failing server always being tried before moving on to a good fallback server. Or, worse, a failing server is tried, fails, is error limited and another failing server is reinstated (prematurely) only to fail, before a good server is used (ass illustrated by new test in follow on https://review.opendev.org/#/c/759502)","accounts_in_message":[],"_revision_number":1},{"id":"5a27cfe6c2cda37d2dabfd90bf9d78f421bd21f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:47:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1cb08a4a0194bf9a0900db09d61a3c6 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/55aab31623984135b5e990e17552c26b : FAILURE in 57m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6bc8998eb8c340789b63fa13731cd848 : SUCCESS in 47m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/576813d6a5c14bd7b3532b0b0a00abd7 : TIMED_OUT in 30m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7441789aed5418d9c4864a03aa15822 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42a1e0dc05e64e5e98f9955df5880bcb : SUCCESS in 12m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7be2e31c2a6c4d7d9d21007d72e5c75f : SUCCESS in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d81f451f74cd45a382e17cf0ded4b04c : SUCCESS in 39m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d17fa091ac69448dad8822a469ab6e56 : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e90a010591ef4375827421501255188b : SUCCESS in 31m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6e3361f05794cc180d7351e0cf50c7e : SUCCESS in 30m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6226e99fffac40d9acb239678b39358a : SUCCESS in 16m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fa2c8c262041498fae2badb1e085d1d6 : SUCCESS in 37m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5630cdfd49ad467e932dc60ab11646b4 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e3fee430bf340b6bb50490929187314 : SUCCESS in 18m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b46625c8290452098456bdb5413c7ff : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf3728b9bd6f43a4bd717a16d367bd01 : SUCCESS in 45m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09345113eb174c3091ef0c2b7830b148 : SUCCESS in 42m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c357c64c5dc54714a533eab6a4ef036b : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/690f3301242448bf96e47a34cdb1b54b : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b045ed1f01f4eec88f4b7a992b60e25 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/477c27d40f154c7586562a9b882ce17c : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e680d3e84c664ff9b7a32df258381291 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6f0451df4f30149a9cf449fa99c2f9687ce3e31d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-06 10:40:43.000000000","message":"Abandoned\n\nsuperseded by https://review.opendev.org/#/c/761029/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e10b4e2457061ffb3235dae978ae9182d045bf55","revisions":{"686fde31b9e779c8f8a0e2ad232725a0317060b4":{"kind":"REWORK","_number":1,"created":"2020-10-23 17:32:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/759493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/759493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/759493/1"}}},"commit":{"parents":[{"commit":"217d2e6b53d25bc7add055539f542eb4211d13e2","subject":"Add unit test for memcache error limiting with single server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/217d2e6b53d25bc7add055539f542eb4211d13e2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 14:44:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 17:31:32.000000000","tz":60},"subject":"memcache: make minimum number of servers configurable","message":"memcache: make minimum number of servers configurable\n\nAdd a memcache config option \u0027memcache_minimum_servers\u0027 which\ndetermines the minimum number of servers that will always be available\nfor connection attempts regardless of error rates. For a given value\nN, memcached will not error limit a server if doing so would result in\nless than N available servers remaining.\n\nThe default value is 0, which results in legacy behaviour i.e. all\nservers might be simultaneously error limited. A value less than zero\nwill be adjusted to zero. A value greater than the number of\nconfigured memcache servers will be adjusted to equal the number of\nmemcache servers.\n\nRelated-Change: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\nChange-Id: I2e10076b7e0741e2c45395439bd2e6d2f8a6025b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/686fde31b9e779c8f8a0e2ad232725a0317060b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/686fde31b9e779c8f8a0e2ad232725a0317060b4"}]},"branch":"refs/heads/master"},"e10b4e2457061ffb3235dae978ae9182d045bf55":{"kind":"REWORK","_number":2,"created":"2020-10-26 12:28:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/759493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/759493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/759493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/759493/2"}}},"commit":{"parents":[{"commit":"247108a4b87d94d7caaed213d2a22c5d0f120a8d","subject":"Add unit test for memcache error limiting with single server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/247108a4b87d94d7caaed213d2a22c5d0f120a8d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-23 14:44:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-10-26 11:11:52.000000000","tz":0},"subject":"memcache: make minimum number of servers configurable","message":"memcache: make minimum number of servers configurable\n\nAdd a memcache config option \u0027memcache_minimum_servers\u0027 which\ndetermines the minimum number of servers that will always be available\nfor connection attempts regardless of error rates. For a given value\nN, memcached will not error limit a server if doing so would result in\nless than N available servers remaining.\n\nThe default value is 0, which results in legacy behaviour i.e. all\nservers might be simultaneously error limited. A value less than zero\nwill be adjusted to zero. A value greater than the number of\nconfigured memcache servers will be adjusted to equal the number of\nmemcache servers.\n\nRelated-Change: Ibeaf6bdcdfb32addb277858f7095c9f0035fd5f3\nChange-Id: I2e10076b7e0741e2c45395439bd2e6d2f8a6025b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10b4e2457061ffb3235dae978ae9182d045bf55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10b4e2457061ffb3235dae978ae9182d045bf55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
