)]}'
{"id":"openstack%2Fswift~759535","triplet_id":"openstack%2Fswift~master~Ic2e9eff5cf37e3197cb22d035c5b125efe512619","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic2e9eff5cf37e3197cb22d035c5b125efe512619","subject":"swift-recon: Use replication ip/port (if set)","status":"NEW","created":"2020-10-23 23:57:04.000000000","updated":"2020-11-04 23:41:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb19e7a8a8c526c2b3b2d6435abd3bbc9c73b745","_number":759535,"virtual_id_number":759535,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-04 23:41:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-11-04 22:26:47.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 23:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"230e55e8669fd386e4b4a9a19f3ae924698c58ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-23 23:57:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30d78ddb4e9541008abaf28fcce006282dc11031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 01:02:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78454f08c3cb4043b98c03eff9275f1a : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a3b7ea41230842bcbf9b3993d5c9600c : FAILURE in 1h 03m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af8631eda5cb40d786453c0c68a66b25 : SUCCESS in 40m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3dfb296686a74f76b8b1af353952aebb : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8fe35d9cc3d44a04be56f24b150b357d : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97052974bca44bb4947c2f5d4e456adc : SUCCESS in 13m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd7d495ec3045938fbad0bca5d68190 : SUCCESS in 12m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d78c4365b194608805354e1ab05a01f : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8501811ab8440a88b85ce5569cd788 : SUCCESS in 12m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42375e7374634060b7d82fc9488482c7 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d4374edbb66246c9a4d3e02dd30fa3dc : SUCCESS in 32m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2e2e4e8d0b146bca323bf27c2b9914d : SUCCESS in 17m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d36d4446893f457ea55a5d0969549b0c : SUCCESS in 31m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/31ecf99987044c33ac614a44e915a41d : SUCCESS in 18m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/691be84803d94b59a2af11dc9cc9230f : SUCCESS in 16m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1751b0a097c549b494d80039c6f29abf : SUCCESS in 22m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30b533bb9e8941f4b051ff75bb24b442 : FAILURE in 45m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d18bfdfaa964b86a41cf9969a3e8f9e : SUCCESS in 41m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fd6883e359e4d3992be3bfa0f6112c5 : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d642e25dc6bf4b37b4d0a85cc3aff41f : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bc1e07d3d24d39b98a516d02dc59e6 : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d491538e448b451aa9a015ad60dfdddc : SUCCESS in 23m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59e30695868a45e2997f0f0666a00840 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b39ee98d14d04b67acf4b3b06d3b427e838ae1a7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-28 17:26:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc2c199fcdcd54ea2c0fa1b43360af59ac4d3d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 09:30:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e441a0c41574a6096ecc4615ed41ded : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fe00d7183e404b6b90628add24c40911 : SUCCESS in 57m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5bcaa8e465f845de96e08eff56f93bd7 : SUCCESS in 41m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56887b01f8ce42328e2e6bc0093693f7 : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c41b112436a04a71b979ace864118f70 : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef759e5aeda349c3b5c0912e38b15a2b : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76a51fdaa57a497d9eac25052dcbbfc3 : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20d44531dbd348f5b5d03a721b8b2b79 : SUCCESS in 12m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4954f3cc59a94d759a98d027b28ffdeb : SUCCESS in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e679fc189c3f418d83af4703e1e3e3d5 : FAILURE in 59m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e3332d709014da19949a7573f97666e : SUCCESS in 23m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa46ec9e8f32484c886061314339e35b : SUCCESS in 24m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d36768bc4977416592bba6d2a60362da : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/68a35ed68f374210a88b4b63b29e8771 : SUCCESS in 26m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8c45709e507d4bf19424ea15e6223ae5 : FAILURE in 25m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc0422bdd7214bdaa13c2e89c9af9297 : SUCCESS in 26m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/73e987b54ee44bfdb53436eca63889d2 : SUCCESS in 48m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a58a4142dc4b43ba95dfea1f8c19961b : SUCCESS in 42m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0eb649b5bd144229809c316ccd15e69 : SUCCESS in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf341d8f9938478a83289b8c286bbd8b : SUCCESS in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17e2ba465564fd9b28494aa98733e41 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a79a37c8b8be40a9ba624d7a37b4faf8 : SUCCESS in 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92ddedb0210045a0b33944d1ff282526 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2dcdc2aa034e0bf9d7598993e376095edd4cb86c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-03 18:17:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09915f1e69920e503141d02c229e02d733579533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 05:13:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38476c2c3de4833924f6ca2fc908f2e : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5949207ad3c647f7bd6c8e1212e373eb : SUCCESS in 1h 03m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b489c2bfa1442ff8a1c73e32ac6fd76 : SUCCESS in 51m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1bd39624393c471791873d3e9c9d8ca2 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f38a0299025a46de9ecf0027c3b23a61 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f2db3e6e5c2c4264af6b966c5492ccc8 : SUCCESS in 13m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4308b17492c54afea878d30552801124 : SUCCESS in 12m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b10fededee124713a48131d59781eb9a : SUCCESS in 12m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5292d1ab74584168b08cb4af7e30addf : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8b5b31487bc4f53a2ceb8c5b59c5470 : SUCCESS in 25m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c427129608974834b30a389130fb0346 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3111a09a08d7410eb5a867b132c15dc8 : SUCCESS in 25m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b27591ed1355419097267d4c3c006717 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1a303bf0f4b04dde8d4bbe769186b0d8 : SUCCESS in 19m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ec97a324dae2447c90d051d2f82e986c : SUCCESS in 17m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5aa75f51760d4115a653216df7efcd9e : TIMED_OUT in 1h 01m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae6fc98c21094dd4b3d5c5f018a79d2a : TIMED_OUT in 1h 31m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53b0c336b31244bf8fb42bead369c7ea : SUCCESS in 36m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f01d1559e9de493d8a3eb6dcd24a5ef8 : SUCCESS in 47m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6720aca37a1490b878df3c15298b256 : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa0b605684c4728adbc6de731be5212 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c0056df831a47849850711cda423baa : SUCCESS in 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f385729c3b59465ab1bfa2d0220ded63 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"369cffa11450702233aa91d749aed7588caefdca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-04 16:02:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b02be3b892465bd128624cff85b95ec8381f9b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 21:51:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98714317f6fa4441981390f02534d4b8 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b90913a5cee344a4b5dbe36eac71f80a : SUCCESS in 59m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aeccd88a6c82425a8a15754fefc96ca4 : SUCCESS in 51m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5507772f8c794a63aeb948b622a6011b : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f4cceb8500cd4ae187e79ac58a18a2f2 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b84cf8318090428092d942870216634d : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59f53991363a48d7acbd8a918b7aa62a : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1561ce930d304033a9104c8477f8bdc5 : SUCCESS in 12m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/17b5ecb5755e4abb8bdd987ee5531629 : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/446531621d954d809ddeffd31cb0f1d1 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f91b871b1cb46c5b3d7782a8ac968aa : SUCCESS in 23m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea734796c48741e2a9f6f0de99ec8c38 : SUCCESS in 16m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/db39a15d863b4065a75b779e2af4fc63 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e3593e73e9d543f29341fdc9b44b0050 : SUCCESS in 20m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e75fa7ef697451ea499034f4fbe9948 : SUCCESS in 18m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94c2f95c6a47493486d57f5b1261410f : SUCCESS in 28m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e75943a868140f2948aaefa94dbe6d8 : FAILURE in 46m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3820af13bbe47309050b8d410dabac9 : SUCCESS in 49m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fcb032838666456490131f6cb104a255 : SUCCESS in 44m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc2f010c0a714ac09f95c830e3afe528 : SUCCESS in 24m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e354deeaf047e1be47b745479ada4b : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4973fdc93b464777bd783422c028f4a7 : SUCCESS in 23m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6dbc43a843466195c05b8956dcb044 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3fc3c8601b7545cc9859fd0fcbb8fd6312bb9306","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-04 22:26:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb19e7a8a8c526c2b3b2d6435abd3bbc9c73b745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:41:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee4695c92c241ab91197fcc4342942c : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/2ef7c4fd513742e99bc90d8d1972f269 : SUCCESS in 1h 00m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8c6b937390640e8bf3ba88b4ca8ac00 : SUCCESS in 48m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29fe818fa5c44856b090ca68c4d85a3a : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fcf10b6875354ffb96a655a63070603f : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e498470274544b179ab93d91dc7c8b14 : SUCCESS in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c0fed60465e14a8aa3ae55df9172fd6e : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8bc70e82563b42e0b744ca27d28a8afb : SUCCESS in 13m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae4b8c28af004fad91e191b4600b97a5 : SUCCESS in 12m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94a5413ccb8247919c5d11b2d4d71129 : SUCCESS in 23m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1f0dd4fbb664a39aec6beaa24e3cf3d : SUCCESS in 21m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbc958f537da4e819adc27ae4c268f72 : SUCCESS in 22m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e14e448efd8749a29fb5218600a6ab86 : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/295c664c854b46e6badc5c229fab5908 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dfb84dc18f3746ed81b949ed2ad74961 : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4db637fbddc24deda43cb76f6aa135c9 : SUCCESS in 22m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b2c5a595fac9418bbe2cbfc14dd7679b : SUCCESS in 41m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cb13eaee2a74ff7a6b7562facfc0a94 : SUCCESS in 39m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eed3e6fab6b1400ea0efbe48aec46078 : SUCCESS in 39m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef63140d83b48f1b01f2826554aa3d7 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5665500eb23f4bca8e9e58885e7f6927 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/baecd187857a44489c649060b6a12311 : SUCCESS in 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47327688aefc491b9408846c6eb34f0c : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07db5ea4c62916537b52058b0dae08a2c166c2b3","revisions":{"07db5ea4c62916537b52058b0dae08a2c166c2b3":{"kind":"REWORK","_number":1,"created":"2020-10-23 23:57:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/759535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/759535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/759535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/759535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/759535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/759535/1"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-23 23:56:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-23 23:56:55.000000000","tz":-420},"subject":"swift-recon: Use replication ip/port (if set)","message":"swift-recon: Use replication ip/port (if set)\n\nChange-Id: Ic2e9eff5cf37e3197cb22d035c5b125efe512619\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07db5ea4c62916537b52058b0dae08a2c166c2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07db5ea4c62916537b52058b0dae08a2c166c2b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
