)]}'
{"id":"openstack%2Fswift~759790","triplet_id":"openstack%2Fswift~master~Ic777b435723f663b408c778f661811439b486071","project":"openstack/swift","branch":"master","topic":"remove_pkg","hashtags":[],"change_id":"Ic777b435723f663b408c778f661811439b486071","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-10-27 03:37:56.000000000","updated":"2020-11-06 23:43:34.000000000","submitted":"2020-11-06 23:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759790-1604706124061-ddfe7261","meta_rev_id":"c8401fdaec196e43bcd99a8e60f14927c5747cce","_number":759790,"virtual_id_number":759790,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 23:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-06 10:42:07.000000000","_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2020-10-28 17:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-10-28 17:24:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 17:24:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-11-06 23:42:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5c999b4bd82912faa3d57248baa336b05b14bd3","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-27 03:37:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dca2374f900d8879c9a1887f86656cbe9575b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 04:45:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bc5af13ede4cc2b8dd184b680140a0 : SUCCESS in 7m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d7aae2d5fd94698895505c6292093dd : SUCCESS in 2m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/531af735cace4721bd3271ec0abf5bdb : FAILURE in 44m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82dbdbd5b3fd4996b6c7ec7f11215527 : SUCCESS in 46m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fdda188e943e4a93818a7a2ba2b0f51f : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9a27e798eeb245cc8308b785361174f7 : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32d9e9c4e506484ab891700e0d810ae6 : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc740843bc5d4e7fb746002288547007 : SUCCESS in 13m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b27ac96d41444289dfcb9d71c95c2b2 : SUCCESS in 12m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/483c1524fbe2499fa4e976a669a8f770 : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb22bb2b62374ec59ddb9296d9c098e4 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce4586ac6d0c4305a0f6bcb202bf1b9d : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/353320b102234a32a24d45df0f28f29e : SUCCESS in 17m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5a97a8792805404995f96d66c2a7b256 : SUCCESS in 30m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/25f36289c5414abc954d9101720d07c8 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c2cff7eeaca44f1b8d409ccd77fce9b : SUCCESS in 17m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81fc279477fd430fb1a339af023d4f78 : SUCCESS in 21m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/837449e7a37a43ef8b8ea70d6356e40f : FAILURE in 43m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef90c6fd13364533b1674cc094a71119 : SUCCESS in 39m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ddf865589934a5e9faa5d8530dc926b : FAILURE in 14m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74a68307bb94deda4d0c11b00e5bf08 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a34f5c215414168a56f4479ccc1bfaa : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3f80e2a94f440f7893e0513eb624935 : SUCCESS in 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d13b8043a26468283dcff880ff65816 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9b644cba61795a9f59ebaa30c3e46e5d41985629","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-28 17:24:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"d5cb9497ab460bfc67b70dd045661d4b990a5906","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-29 00:54:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4aec181b6ac1e387f137f38acdf534d1cca6250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 09:03:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/842533d54f224886b3bb3e052e442404 : SUCCESS in 8m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51f9e9e76e9e4ae2b7ac1082010b8c73 : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2790dc807f7148eb8ab68746b3ee5a0a : SUCCESS in 1h 09m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f014ffd7cfbb4330ac59ba01aaec2c9f : SUCCESS in 49m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac571b3987154e36a126a5b94499d8fb : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/af0de2f852cf4792aa214fe28ad4e5e5 : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63c5eabac9a74b358ce9e4896f72c5d4 : SUCCESS in 12m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e2db6bf6c9b4936a7c38e644bfb2b8c : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5dc89ab774b24e5f8c4d60a029dd6aee : SUCCESS in 42m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/744a4c131e074c42a8e9b481c35bad24 : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/853c835da08d48abb8c9aefd4d27968f : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c365285f67c44bef97cfd97ae1383790 : SUCCESS in 48m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6540651297dc4fc4866761e7d0bc8bee : SUCCESS in 31m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/edd00214ff804deba2a02eb3e7ae295a : SUCCESS in 42m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/16c9c7c95b00465899fc4465f7a29117 : SUCCESS in 42m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e9717068ac6c4266bc89b65dcfb55023 : SUCCESS in 26m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aeaceb3b891c45d4932d9300d1f47482 : SUCCESS in 25m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db8994db437f41f1a466beb023d22a78 : SUCCESS in 45m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/539d1ff7ae404358bc896497f4ae1140 : SUCCESS in 42m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1c9c0dc1b1b4339aa543035f23ee6f7 : SUCCESS in 34m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c978924f08a44f6af10072f67f62d75 : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af4e36e394b4d329194999fdb7c7163 : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a9d00a3bd0d4960bc33e53ed07696bb : FAILURE in 42m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6f8f60849c04cd7a9c2c268c9333f34 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7a10628960392516514fbd789fd80091198be7a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-29 22:41:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6017b41be44f7bafbe777490566fe93342574bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a5260420b244b58fd6e59c9cae904c : SUCCESS in 9m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb4cb3092a5a47e2a41cd3ddc051ca7a : SUCCESS in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/44b49c3b8eae4715b923e966c12decdc : SUCCESS in 1h 04m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08d73dc2d1ca43499d00f4d72f9af1ca : SUCCESS in 57m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03f3b043efad4eb2a71efb2eeb098e8c : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6a543adcecad4dc6b763ae3ab8bd1506 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ede55cacba654ecaabce5b8099a4b96d : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c0380557424866b10fb54ab2244445 : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acaba8b3bf364a98b79f585d5950ca32 : SUCCESS in 13m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87179e38636448b69a7dbe882299298a : SUCCESS in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b95bc21a77e245b19cf53903086bc7c1 : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ac0b5f7ec2f4f4a90cf0673694c08f5 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad18f99713d34a64bfaa9e817bb81bde : SUCCESS in 20m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e3d1bb0df6f24cae9b36f86ede968425 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/93ba5833e793439b92d2571d8dbe7a2b : SUCCESS in 22m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a146354b088b40aba4ea9acaa9eb1a7d : SUCCESS in 18m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f728b07c418a4a46b64269a111010db4 : SUCCESS in 20m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e7f733231aec42069252c80db036a7f3 : SUCCESS in 45m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba449a8226794dfb89609df2bd4a39da : SUCCESS in 44m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5b509ae33bb4c81ae7531a6ab52abf2 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d391448a1b5a42f3be05af416efe16df : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fdeb31ce76471c9c154c6a29226ad0 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c49d82051684ba680dd72ec67121b8c : SUCCESS in 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5649c1e1e62b4d539757b4a2731c1129 : SUCCESS in 2h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"37872e127d4d0bc4a48f2938ebc4721b202f87b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ff1957fd8573859a6db3988192b0a643620699d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 09:09:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d58e1f5dbd4024a15240880baf6852 : SUCCESS in 11m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cf3e183cf0d4dcf83e967a3b7db95c6 : SUCCESS in 3m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d74df01a57954033b74b483702ace6b3 : SUCCESS in 1h 05m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2057ff7e5115459a9df1c2697168112d : FAILURE in 4m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fa0f32317a1c446cabe3248aca431f67 : SUCCESS in 11m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1dd2ad4c17464e41954d30485a9743c3 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb8fd2a4769e4227963a969bc1a6d3a4 : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40d4d86bf2514619a02ce80e320bb475 : SUCCESS in 37m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07741538753b4c3c880eaaf0f983b297 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78f73d52e528429a8ee6cea2deae34c7 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f74d842946204e29a67e39478fca455e : SUCCESS in 24m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b1da8302b844e2fb4ab7bca15a4a256 : SUCCESS in 27m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f18c9c9fdfac48f3b1aba799da18afd0 : SUCCESS in 18m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/276e67405e3e4e3d8c299cbfff155bac : SUCCESS in 21m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/10ae612966334633924dac022a60a0ab : SUCCESS in 40m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a27e2290e75446c7b8fbb40815639840 : SUCCESS in 21m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7072febbf1a6469da8f586295ebeeff1 : SUCCESS in 45m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f841282ff7c48ef958518a8ac956f7b : SUCCESS in 44m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20f02a43e8ae4f48aa2293b1521af644 : TIMED_OUT in 1h 03m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/628f900d524b48d381ae5be150c97cf8 : RETRY_LIMIT in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3fe150f500464793dcc5225e1ef528 : SUCCESS in 6m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5649666f7bf46ed861fe7c49ab0d97c : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7d71715ef32a84107d6cc188b65ae97569410880","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-30 18:47:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4dd7a7f476aa56ea6c839ab849cf2ffde19f29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 01:36:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c793ee58f2542ee894b43c70e4d3a0f : SUCCESS in 7m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/228ee4b2105f4d649f80f82fa08baa0f : SUCCESS in 4m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d5ec67a5b0794862b6b0d5ff5555f821 : SUCCESS in 1h 00m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80b13cd964fa4c3ca1c2eead8cb26b0b : SUCCESS in 49m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fdf2c067ccb47fd8cd9e4950060bdd1 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b788f5de1d44cf199d108e90541bab5 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44f0c20ecee645d4bbf8a2ffa6a4fadd : SUCCESS in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0139b0bb537e461aaaa9db1de91a213a : SUCCESS in 14m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/778ed9387eb44ad6997d34a2c23cbc91 : SUCCESS in 14m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/16cfcec93f30435fb2240d20bc9acbfe : SUCCESS in 13m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2470e22705d84114a2454098e63a6d38 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbb3dca60f32479eb2413f00e8c84477 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf1daf76965b404b8119472f9543af2d : SUCCESS in 23m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/56c9fc8dac8b42d7862bbaa143c6fa3f : SUCCESS in 22m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/009b8209d479407ea9f3394963924dc3 : SUCCESS in 23m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3f25f1f267cf4319923c18a4d7d01a6b : SUCCESS in 27m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98dced8e84354bcf974f664028a6ec25 : SUCCESS in 25m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95217238022e4f89a1dfd099f238dba9 : FAILURE in 46m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ae6dd16b2d446069c583414005a0d15 : SUCCESS in 45m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0754234d12e4f51a08ab3b92032bc66 : TIMED_OUT in 1h 02m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a0d67a947a541029c069e6ed9dcca9c : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d7c0262ba947cb99a43a6c444fd697 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2dd06a48c70946a6bb8ca3da3739279c : TIMED_OUT in 1h 01m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f75e297ef1b94f0da716cd8472ff8272 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bce7ed6a9f1830680355c60219d317f502e5fecb","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-02 01:10:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec90d5fe96290e4fba4bcc3076c42f509b24022e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 02:20:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af9d08d9be3240d0917a18bf44827c1b : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d0dc786898f43e1826fe39dffe53db1 : SUCCESS in 2m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8011fe7906624db0b934b288516d6c23 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dc77afa70f72462990e527be999dcf11 : SUCCESS in 48m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/407f769a872845c7b997342bcdb5973a : SUCCESS in 8m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/975a2bce400b483a9575199f04838051 : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dcc8d61319b42c7addc424d9a088732 : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c382911aea74eec9cf0337b4974523d : SUCCESS in 13m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37f7f16a5ec340ffa50dfb7e44e10931 : SUCCESS in 13m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/47878fb5fbc04e5d874dc0b5f6e652e5 : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/448bc678fe874fe99bbec278613a0db1 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5bb26e560be46a89499ba7324ac4868 : SUCCESS in 30m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d80a54196c8c49f992ef4063a5d1f8fa : SUCCESS in 31m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e089f46ef9cf42478c05e3f0c3335a80 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/115bf89bfed04abf98cce97e1b9b07f4 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/53ac2431857542caa62dd5556c2fbc89 : SUCCESS in 46m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4bbff450add148c996c3ca27c33f4daa : SUCCESS in 19m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b417766d7d74b16a1fc640b5a66d046 : FAILURE in 53m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7826f651baaf40ff88cf4fd5770dd12f : SUCCESS in 39m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae70bad3ec374279ab5f45f53c38b8ef : SUCCESS in 46m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b1bde62fe13422b95e9e2a7d96ce447 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94437431428446db9c027da60efab71a : SUCCESS in 7m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f0585ee63eb463b8864d1395237987c : SUCCESS in 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e82a0171a654a0e856fd80127ff01c7 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7ef26750fe2c02a8cb52e286efec2bcdcb2bb89a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-02 03:10:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ec760738533ca6475dd11187ac316d74e0ec793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 04:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64fa14c6627f4222988ee054621c558f : SUCCESS in 10m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bcddc559d884a7da5ab77cf38b8fe91 : SUCCESS in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/102f248e185d4cfd88f68730dfecdb9d : SUCCESS in 1h 07m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2323d06d61646e8a05c04718f84d796 : SUCCESS in 49m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/748996a9bd464c04bf69be3c3d29b013 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5cb673a6aa544f64babcb7bb30c1cd47 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/534ecc0a983e4f0ebb244934005e16df : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f033d01e6a914f9baf828a15b38c54ce : SUCCESS in 13m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/86d481498c8c4435ada7de65719773fe : SUCCESS in 12m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ce8029d78d4a7aafbedcf2463aeed7 : SUCCESS in 11m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f74928dfb3034d78a1b3764d3d5398c1 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb4b18482889418a9d5962bd0f8d6f2b : SUCCESS in 16m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0da9799987cb42aa85446fd4b8798248 : SUCCESS in 36m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/221c5a4b2acc45c8bfb0364882e32e41 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9263c3d549e547ae9c1394c77187f9d5 : SUCCESS in 31m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/96a8f8bd0ebe4591b15a18ec568b02d0 : SUCCESS in 42m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02ee60d0f60f42abbf63a6837296523f : SUCCESS in 23m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c76e90edb8f84e36ab7176a0803b0b84 : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c0c2c7108a14840b39064ea34a46aa7 : SUCCESS in 41m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f60c7b10cb8240a9b503d690af3be7e3 : SUCCESS in 44m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f7a4119e21048da823d74c6a95ca8e8 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4791f62cc3b4d0d9065e65166a42899 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/373e3583b841420fafc43c9f16df0195 : SUCCESS in 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a3605e413b4acb94b72cd85c652a47 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"96c3004feadbb24e7227c4be39db7ca7177c5956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 04:19:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d9746db478fa5281969543599dd1ac28215ddda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 05:25:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da37fabe76e4d05a2f76026aa7fbfac : SUCCESS in 7m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c5ac2256d82497b8ecc4b921a9ebe39 : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/59c561505d1847c5bcb2ffa6c5e8349b : SUCCESS in 51m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c9cb70db1a6140ae94af361ecffc715f : SUCCESS in 41m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7fef8340f1734076b93d115cb43565c8 : SUCCESS in 8m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/440205c433514047a93b72150a655581 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d05763fc4638442b9edda8f5e04ee877 : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9231368ba6ee4b9690442dd076ea0780 : SUCCESS in 12m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba6a0d020df746ac8b3839caaf6ef1b4 : SUCCESS in 14m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a82de84c9f4d6abe6eb7df2465564e : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c98b7672eb2549d99c6cd43e1030ac7b : SUCCESS in 21m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3bafa44da4c7414593fbfa9f7aef9583 : SUCCESS in 25m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af040eab9bc247ffb1fda42c46020e4d : SUCCESS in 17m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dbcd85b4c6a04096a606b6cb0aee985d : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4aff1e5b8a304f6dbbfd3e8412d4bd3f : FAILURE in 24m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/513c611e43f9499280cabce4cdc6e78e : SUCCESS in 18m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f651ff9f3e224d84ad432dcb23174cfe : SUCCESS in 46m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36fcf4cdac5d42ef9c4b04178cd3c581 : SUCCESS in 48m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a20bacd72954ec9b2d64cd7d86b0e9a : SUCCESS in 41m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e5608b2ec44bbcab9faafd1eaae745 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a428aa54ba7a4df79b7aa7a4960116b3 : SUCCESS in 6m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/143e543280b341d58842190c34064472 : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2f6709bb42a60b9e900183608158493d851225ec","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-02 05:53:03.000000000","message":"Patch Set 1:\n\n\u003e Build failed (gate pipeline).  For information on how to proceed,\n \u003e see\n \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da37fabe76e4d05a2f76026aa7fbfac\n \u003e : SUCCESS in 7m 27s\n \u003e - requirements-check https://zuul.opendev.org/t/openstack/build/1c5ac2256d82497b8ecc4b921a9ebe39\n \u003e : SUCCESS in 2m 23s\n \u003e - grenade https://zuul.opendev.org/t/openstack/build/59c561505d1847c5bcb2ffa6c5e8349b\n \u003e : SUCCESS in 51m 39s\n \u003e - tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c9cb70db1a6140ae94af361ecffc715f\n \u003e : SUCCESS in 41m 12s\n \u003e - swift-upload-image https://zuul.opendev.org/t/openstack/build/7fef8340f1734076b93d115cb43565c8\n \u003e : SUCCESS in 8m 07s (non-voting)\n \u003e - swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/440205c433514047a93b72150a655581\n \u003e : SUCCESS in 8m 53s (non-voting)\n \u003e - swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d05763fc4638442b9edda8f5e04ee877\n \u003e : SUCCESS in 14m 15s\n \u003e - swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9231368ba6ee4b9690442dd076ea0780\n \u003e : SUCCESS in 12m 47s\n \u003e - swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba6a0d020df746ac8b3839caaf6ef1b4\n \u003e : SUCCESS in 14m 02s\n \u003e - swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a82de84c9f4d6abe6eb7df2465564e\n \u003e : SUCCESS in 12m 38s\n \u003e - swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c98b7672eb2549d99c6cd43e1030ac7b\n \u003e : SUCCESS in 21m 51s\n \u003e - swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3bafa44da4c7414593fbfa9f7aef9583\n \u003e : SUCCESS in 25m 07s\n \u003e - swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af040eab9bc247ffb1fda42c46020e4d\n \u003e : SUCCESS in 17m 16s\n \u003e - swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dbcd85b4c6a04096a606b6cb0aee985d\n \u003e : SUCCESS in 18m 09s\n \u003e - swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4aff1e5b8a304f6dbbfd3e8412d4bd3f\n \u003e : FAILURE in 24m 26s\n \u003e - swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/513c611e43f9499280cabce4cdc6e78e\n \u003e : SUCCESS in 18m 20s\n \u003e - swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f651ff9f3e224d84ad432dcb23174cfe\n \u003e : SUCCESS in 46m 47s\n \u003e - swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36fcf4cdac5d42ef9c4b04178cd3c581\n \u003e : SUCCESS in 48m 15s\n \u003e - swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a20bacd72954ec9b2d64cd7d86b0e9a\n \u003e : SUCCESS in 41m 30s\n \u003e - swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e5608b2ec44bbcab9faafd1eaae745\n \u003e : SUCCESS in 12m 39s\n \u003e - openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a428aa54ba7a4df79b7aa7a4960116b3\n \u003e : SUCCESS in 6m 18s\n \u003e - tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/143e543280b341d58842190c34064472\n \u003e : SUCCESS in 1h 01m 16s\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58e05c506dfa6274b79ef22673796021b4b9dcac","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-02 08:39:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"509c6e0111d1cd1543ea7c604b41d04d5cbf605b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:30:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e41406891bc444d9c0cf045fad0299d : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f018736542148268d6794be350ea9ad : SUCCESS in 2m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ff94532623e04aebaea8884673f701d4 : SUCCESS in 1h 07m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9fbbab87cac641dabad937f1ad1f8a03 : SUCCESS in 48m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97111912a4094428b402e36262588d34 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34e0a2587c234ce1b02942bc33cad79d : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e51a146d84924afdb21b5d0f4038e722 : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7d3073b637741cd87a0276fc6bbe709 : SUCCESS in 30m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dce9475b74ea496d95d37a9c923f56b2 : SUCCESS in 12m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff422cb8447543d79978b561f0c4df8d : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c0c0eb0cdfa44efb1eeb93d43ca3d6b : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/37724d74d1734ec3896b36675363aa3a : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22d015c7205744fe8e95092130cb8e6a : SUCCESS in 21m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/61533b85b3454d04bec35d3313668d99 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f031b6870518448d9486ac955bb6ab31 : FAILURE in 19m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/07f24b63e146412db5b0917749c29ba9 : SUCCESS in 21m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/14fd71a3d6b94a8aa9128d74dcc3a4d7 : SUCCESS in 21m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/31a929e402184932b7b04bc3eed7bde3 : FAILURE in 49m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d5d03b3832143fcb86b5afe50ee7866 : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fe44978b14046ce9124b324f391824b : SUCCESS in 40m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e616f05cf724a59883bf3fdcf6065ac : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d47564520e492996907324d9fdb5f0 : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5476f7eb0d424c47b1a1805bcbfee6e1 : FAILURE in 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ad2be5f375f419387500d217ad47ccc : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"482ab358eeed8aeadf7b5dbae043dd911c683601","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-02 16:02:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40dfe3f532b17e041ef6d6b8ff77e05600e6ef6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 23:55:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53331cb8e51f433a8977722acfb559ca : SUCCESS in 12m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4c0ff0152b244d98446144607eedd9f : SUCCESS in 4m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/94f5f2d6b8214627bcd3a856a6f37b27 : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48fd13d670034a0ebb660c578c625028 : SUCCESS in 41m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/415cf3f0cb6243308791a4d70210e6fd : SUCCESS in 11m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/548a6ad22fff446290d09f1f646e797a : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0617ebba8c345e69f0e61176577967d : SUCCESS in 13m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b98111ad3745c5b18ffb19446846f5 : SUCCESS in 13m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13c2859b6007476b865e0306f5c4a48e : SUCCESS in 13m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2e86afeb631451fbe30b48ad4f541fb : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2fa2572180a44cf68ba03b39d9470f03 : SUCCESS in 24m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b51a528299694a6e8ac8ee1b91f23d06 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15c17c56442d40e29d5700d210464c4d : SUCCESS in 31m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3930906b999a4f4cbed1febb00de4533 : FAILURE in 30m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/134383cac9f34ae5847915177ba81e97 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a532c1ba8891433abd1c23696736c0ec : SUCCESS in 23m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6603e4f57f8948f5aff5619775b8a016 : SUCCESS in 22m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a17c3f09b3a4d3d9bbc528497f4ada9 : SUCCESS in 44m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05df7219fd8d4866a6597b34835b1cd5 : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85a0dac8030047f3912d8daf4a927aa9 : SUCCESS in 39m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de509d515f224ef5bcca198ed5c96dc7 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b2698f90214021b4f7361c63047f9e : SUCCESS in 8m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/326e61b0a5b14eeb81141ebc345082b2 : SUCCESS in 23m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50e63da6cfcd487eae52c29747451d91 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e701833ce7caff729cda219f5e242dd13e3ac848","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-03 00:59:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"194ec4bb2999a1377c3753462a0ac96ddce86b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 06:15:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80460c0272d4c648dde37eba9028672 : SUCCESS in 12m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f52717f943044bf87017da25dab5fa5 : SUCCESS in 3m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/dc9ccdc631d9437ead90d9c14bb3cd0b : SUCCESS in 1h 00m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ede3046eee54904b00c4aedaf8a6e4b : SUCCESS in 1h 04m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea2b2908cab54837a92203911b263469 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24461abbd6a84dca80f61a9e9496c8a5 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce544990aa684fd0bb872a3bba199c5d : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/359a8e5d38094a8ebaf99764a9e85f84 : SUCCESS in 14m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06cb578cd80e4b1a9d53bf1dc3e84f74 : SUCCESS in 12m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ca0b8acbf395489893321fedf2fd97dc : SUCCESS in 12m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1365f94fb1e44d3b33cdf829a33198e : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71aeab7ef93946259da2365a2b2f446c : FAILURE in 23m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb84e222c3fb455981714955537e1d9f : SUCCESS in 16m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d11309dded334b15869e203ef4750033 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d5ec3e4bff924a8b8c8c99467bd132ca : SUCCESS in 20m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8474e16f3d204c01bcc9c830a89dcf6c : SUCCESS in 16m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15dd2b596eda418fa8e0caae9a79ea3f : SUCCESS in 19m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/387c8586b7a9415f93743a1d5e30df1a : SUCCESS in 43m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7b8d2dac89649c08100f80310887f65 : SUCCESS in 40m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c675c370fc9e4ac8ac636fc116266192 : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ffd3aeaa80c441cbb64ad07ef8a42f1 : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1646c5be066048369885896a8b0a340c : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9c8086428d9459286801b8cd87e5de8 : SUCCESS in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4069a44adf1d4af7b656ce9b37d62601 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"70651de562044ddaa2fcbe6b40ce599c6f9d39c9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-03 15:48:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2fcf5f7fbd8f3b8b83198e40d36c1e4d35f46b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 02:54:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d111df6e87144a7793e49d6dcbfaffdc : SUCCESS in 10m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/07ed1aef35384c51a9e553fa17392fe5 : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ad5eb83829784e249c39e0c7f53afd9c : SUCCESS in 54m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d30eda5274b74c85b602498ffb31cd4c : SUCCESS in 51m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/642ad6026f6b40cf9c5b3cc2442ce75e : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/23737affca2e4d4eb2e74ff9d4a16c7d : SUCCESS in 10m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3305bddc7d9243c980d82b69ea09ee75 : SUCCESS in 11m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ec4d253a5b734d1a961b14753bfdc7e6 : SUCCESS in 12m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf2fb33adb441a6a06dbb87ff792e03 : SUCCESS in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/62c2cad3068c46ae9ae6519a525c9922 : SUCCESS in 12m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/957b737f91fa453baa83c9ab791177b3 : SUCCESS in 30m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3050ba00bf9d49c8943efddb1988dc75 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/790abe6ead644045b959615799a4e36d : SUCCESS in 17m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1842001d790848fb9d979f7ac492aa51 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c32e9359fa4a40b5943c2109d160ebd8 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1caaf6fb558949ffa74e69694534129c : SUCCESS in 17m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd76cb9d3c044028b0d84a6de2d6137a : SUCCESS in 20m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/deb0d87de9a04002b31ab45f2ae45cf2 : FAILURE in 47m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ab298bf42c64a59ac4dc83a2eefe813 : SUCCESS in 42m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f059a22ccf545f9957249b9078167f2 : SUCCESS in 39m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48e797a5306242ffac3e2964a3ee31a8 : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb43b8671e344a7b0c2b9ff7caa8236 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa6085a762674aaaac1bcf919208f4e1 : SUCCESS in 26m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32547b64625342afb945aad00da6c596 : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"18601c4a7e80a343f141a27f41abbd20dc5f6986","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-04 04:56:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35d07cc471717634ed4376eb20dc59713cfb36b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 08:15:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f14507d24094b629712e40ef4a22eed : SUCCESS in 10m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1d3ffb38a994d389fcf2f46c8f923f8 : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/540c2b4e3e544541bc051d7d2f4a4306 : SUCCESS in 1h 07m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3d62ca3b4a604e8596244cc06cf67030 : SUCCESS in 39m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf517e3b0bf94ede8a8f720687e52ab8 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63c9772c05db41f5b63aa33936b4a6a4 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/61ed1681605e47b7bda7aec3f5a8b331 : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76891708f2a141a9bc97d9fe7dcbd9a0 : SUCCESS in 13m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f225151560d74072928a7b9205c022f7 : SUCCESS in 13m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4a495b77c0004457980d3a5475f15109 : SUCCESS in 11m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a396a8fc08c0400cb89275cffaa34df1 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/317bc612b3474555b21ec0767a5ec5ba : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ccefa66ac51249dc9ab5fcfc58056278 : FAILURE in 23m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/819f00fd5ee8457a9fd82d946f9ce8e6 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e39494cb35f647ce8ce76e24c2324924 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d6d176281fe64caeb8f2167545c554a5 : SUCCESS in 18m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba5aa7d6d76d46759fd14c83a91c2227 : SUCCESS in 21m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9132b3314334f6693ddfa8060ff797e : FAILURE in 44m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd861c4669f34185b8a628a70a611e79 : SUCCESS in 33m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58ea02f5c9cd4d448fc0759603244482 : SUCCESS in 42m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3adff4c15c4e4001a4fa114df2745752 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c299e54a90443669ac3d009341152ec : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d0ffa3b65e841c28ca08a442b6ff422 : SUCCESS in 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1465d028af494505994c7f33b3fc1617 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":1},{"id":"29a642315a11a09b31c5906208d35d17a0b52a86","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-04 16:01:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81c64fe37fa4206580467e30508228715bccaac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 21:35:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029c67f4d5984a12b2b07d251477f28b : SUCCESS in 12m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/262a794fac4d4bf4988f156f0d84a18f : SUCCESS in 2m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ca0906b4c65a4f55aeb8515308fb9f50 : SUCCESS in 56m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20151f5f163146079c8d337350f9ec3f : FAILURE in 54m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26e8c263857a4bb09049dc636357fbd6 : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17ddecabe0a74c57be684d0568a50846 : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3a41969394449f280e254de9d86d109 : SUCCESS in 14m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19de77264d7143e6b85f84ffe5104d54 : SUCCESS in 12m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/90ace43b82914eb3a2853e27845a2002 : SUCCESS in 12m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b01da3f7ba2e4615899c86f09d24af02 : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/647af070dc5f4b6ca841ef184565987d : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73bdf0dadf01497fabc3bfd7a07c4b3f : SUCCESS in 25m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/198f846b776d4c8d851d39a1944c1300 : SUCCESS in 17m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/679108f463d347079978bd334179c2b7 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/92693f41c02c49699847a145fe4d5ed0 : SUCCESS in 19m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/47563610fd97419b821d5770bf3c4d5a : SUCCESS in 25m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9edafc1a82804fe797cf2a3bb6e1826e : SUCCESS in 25m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d32a50d73214d5081b5c40dc77e0282 : SUCCESS in 48m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d18678dda46e4f16932eada26a47a1e2 : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2b008b9a94d4c378c638745422f961e : SUCCESS in 43m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b343d2e056c415c87de4d72390fe5b0 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/827eccbc04c84c3699de365896f51706 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ee73ff5a1034328bf68831ed9dfbbaa : SUCCESS in 21m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de6bdf138b474e538ccfeeb59130253a : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7a6abb230ca9691b8742d5d4d82593f45f240eba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-04 22:27:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d51bdb7940cc62328eace01d8866542cce23933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:38:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5162c580295945e49d91440745b0bfc4 : SUCCESS in 12m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5708ae277bb4b6199e7d614a4376120 : SUCCESS in 3m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/620abe908c8b401086104a798a7006fa : SUCCESS in 1h 03m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/87cc404a2aca4856bf4308b99eadd970 : SUCCESS in 47m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca53f1e2805c4c6f9ba7af467100f591 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e345bc7618db4d75a6f896bf9b8fa836 : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3be168a890a4e6b9f7053e2497b4fa8 : SUCCESS in 12m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c64d570910c427cadac6e00763c33e7 : SUCCESS in 12m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35b73c4509e3498792a85134132436e6 : SUCCESS in 12m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c667007c55640d0b5718a4f0389c00d : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d239a5e82ac4d3da265faef203afcd4 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5214377eacf442d3ae2db4dba6705d2c : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eacd243320524de2a7c272104876bd69 : SUCCESS in 25m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9a65dfa189ee4820949dbddc32edb74d : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/151c744c8b8d4ad1b6940f5d2bdc41a1 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2f1b980f4a3f4a5cae68d6b4582ea51e : SUCCESS in 18m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f06402c457645bebe2d08b6915c9aab : SUCCESS in 20m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a74b966791b64244afaa4f87d2aeb5fa : SUCCESS in 46m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d96d3f0c272e4667b6e3bb22e6e57ea4 : SUCCESS in 40m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd2fc31afe63474180524321295ca27d : SUCCESS in 40m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e5c5a89147249afa4e41a0658ff7ee5 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d443055dbc4f2ea9809531cdc959e4 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48e6fc2b78f447888060b775a3c4c930 : SUCCESS in 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/168b9c3d995c48f0a7e8634ee2c34e30 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e743382d2acd6543d74c439331972a11cf12b842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:38:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"196a1d15dc6260ad8b52d2ab7245e54bf9a967a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 02:33:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9511e99654430a9d8fbcba11a3138a : SUCCESS in 13m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f05657498a842b0bce1731d3a0cebd1 : SUCCESS in 3m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ddde368f8b7b444ebaed68315d1ce518 : SUCCESS in 1h 09m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d536637a2a34d4d95c09929e540394c : SUCCESS in 57m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a0ade500e83b4f809dd22a13cbf0d105 : SUCCESS in 8m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1315f86b684e454082b034783fdd5611 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fceaeaf9147f4a398f44b5d4ea359f4a : SUCCESS in 11m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3691e86c9f4f4112a1526241810a48f5 : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/856688e9b7494caaae0e495b0906442a : TIMED_OUT in 1h 00m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4797f563bd994e6eb7c9352177193c5b : SUCCESS in 11m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bc6933af8bb463bb04ebfbd99995883 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8670040bc9a34b4191ba3b29c09f41f1 : SUCCESS in 24m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/406ea3a394ed417d88b0f5017e9952fb : SUCCESS in 24m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1284a086ef334e80b4399b508e3a78c1 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c848257f85de4d6a96b5b40f406eb2ce : SUCCESS in 26m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e594e126ec194277b9453de9318a0270 : SUCCESS in 25m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e7871d138d146febbc485ed8dc9626f : FAILURE in 43m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb7779beaf674825a7641e2cef3fa28d : TIMED_OUT in 1h 02m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d713a5448f3a4c71b354bb4b6afb357c : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a171156c96f431d9525c5f892592e07 : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8641ee167e4e4ac1b82180e1b53cdcd8 : SUCCESS in 5m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d7c86c014ba42209d31f7bdb5be2ea5 : SUCCESS in 1h 40m 06s","accounts_in_message":[],"_revision_number":1},{"id":"98fc000087398d50f78d2d5a537c714f71ffe04e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-05 05:10:01.000000000","message":"Patch Set 1:\n\nrecheck\n\nSorry -- this is getting embarrassing for me :-/","accounts_in_message":[],"_revision_number":1},{"id":"3b1e74a7e0e7e9c618bbe5469b9ca2e0dcd15702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 06:12:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a885ab2926c4b7aad32a5eddf1746e4 : SUCCESS in 12m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad0036c3a1ab49d6ab9cb496344d6972 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a67e648a0b5541f399417a0b1eed9244 : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d4a6d9f1e13d4d628dc21ca514bb4653 : FAILURE in 4m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f1b777c80fb4108b0948d5081d0b716 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b1ec0f845034c87aae3f4eb87e0794a : SUCCESS in 12m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7aad23273b724d6b830ae1ae8d1353e9 : SUCCESS in 12m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/476d5fc65bb94a2697933048e9518c14 : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4edb2753be164a99a935f1d6cf784f65 : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b07e6ea83034373bf3045f7c8fed17c : SUCCESS in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56a32c2426ea4ea3b745aac916407b85 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97e26f03a27545a99f2b696881e892a1 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e45bc0feb64485793329db197373dc3 : SUCCESS in 17m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2958c8337fc74a96b40d665747de691e : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a505cc46d56649e0b0491d7a80aee1e8 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b50cab7487da40f1bec37a399f72ac0f : SUCCESS in 17m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d2f0a15431644fda38fe2c114644738 : SUCCESS in 22m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2bea7d47a19d45ecbe3f4f9e54e90a27 : SUCCESS in 56m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e958395226d44a39e32910d06fbaf22 : SUCCESS in 32m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9af9e1a6dc06498398c882c12cdc10c7 : SUCCESS in 33m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b773c007a6274567a058105fd2813717 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7f3acb4988446eb8a53040247a7dac : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38f2bb386fc64ca8880ff2271d7a3217 : SUCCESS in 23m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26a5f63c6afc47ef952aa70540998111 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d3742f315e500bf1feae88f03457216f76f3dac8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-05 17:11:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb6476c6fa4234d9846bdeec4e0489c99ccb826c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 19:43:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c8ec47e8cf4bdcb54d317d68a10dfc : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e06daf2278a74331ad98b2d36a8889a6 : SUCCESS in 3m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4f668906651e45bd860679c5d8c925e3 : SUCCESS in 56m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39a274120a1d444cb9e60b1c5ede65ef : FAILURE in 43m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a562e73fc69145fdbb6bbc98a4cc0f2d : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e3cd96fcda9f466aadd97703f2089922 : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03d3fe9faef749bcb785ad20b339b961 : SUCCESS in 11m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5939593fb1184d63a8bf6c13c6300c39 : SUCCESS in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13c9fa04183d4db585cc0370b365c307 : SUCCESS in 12m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/838b5af6bdb9497c91c2e0607020bc03 : SUCCESS in 12m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d174fcd00bf401ea2ad3d3269f9849f : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7def9acdfbfc4be9860ea274b59a529d : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d0cf566f9fd4a12b725065919dc7350 : SUCCESS in 16m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9caeffc0dbdd4e85aa4a78990f248de6 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/faa01154d397445b88a48f4f2a62a6db : SUCCESS in 17m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f43d7d176b02452391472b7fddaa1de3 : SUCCESS in 21m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28e46900410a4df6b5c0baa3c54bb259 : SUCCESS in 18m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c0c81b8dd1f48659b2eee323c453c2c : SUCCESS in 43m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d1d961c1b9d45ae8d649dc5d3bb02ac : SUCCESS in 49m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0362d96c0124415da5833d53a17e9505 : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acbdd62b760b426295054e01e453c8ae : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3671e48c88354b5c8172a4559d99c245 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc7c97509382408e9b9da3bc67962d22 : SUCCESS in 19m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fed796be07414154a9fe6d564621a689 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":1},{"id":"add86f262b1ea054d095c4e7941b7bc2021b9354","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-05 19:58:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"630a679a3632091b7e7b0cbb976e0859a543d123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 21:42:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717ba70f3bca423ba51ddaf14d242b17 : SUCCESS in 16m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3909574387f144b3969047446693394c : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ab64e25defa34fe886347291c5393dab : SUCCESS in 54m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a167481b992439885d8d9a81cb9b474 : SUCCESS in 53m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0782e7db24cd409497632d69dd80222d : SUCCESS in 8m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17ee6cf8d9bc4d20ae15a9136e37111c : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c72bd1d510d4bd680f830e76f32ba3e : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/043812239b774452b9179485b40cdc5e : SUCCESS in 15m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/834f4ef44c7c414d8a4f92051b4b99ef : SUCCESS in 12m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4cc6f8605e6d493fb0fe4414f02665ec : FAILURE in 14m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f9b3ff8f2ac4e10a3016435fccf61b7 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2eb0c84b713474c8711c81382484863 : SUCCESS in 24m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87cc2dd762534e24af2d65747a3bba77 : SUCCESS in 20m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/82cd6418cf1c4b3f8b94caa94816ebe3 : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ae0ff1005d414546ac2ca3d14d7e25e0 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9895db4bc0a048878562a30b9d298132 : SUCCESS in 18m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a2d7caf91b14164878d9f0736a888a0 : SUCCESS in 22m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e6756c28990f4719950acf03725c7736 : SUCCESS in 46m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4d550a53f46432e91134345c252090f : SUCCESS in 44m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0f02e51f69f4560965c3c42f84a2079 : SUCCESS in 44m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bc686880f3a46ec91bbacc567ad7124 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8610c38b3a476b9ac3bff5f10f9a87 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb02167dd77b41a389873d76eb9ba527 : FAILURE in 26m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f234b114a1cb46a1a6aec089dee9bc79 : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b5a1eee1974186f85eefef1e71b27a38bd786ec6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-05 23:34:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28abff675c982a329323a4ca0f9bb0fe842b0e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 00:53:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf94381d617142818bd4edf5cfffdd57 : SUCCESS in 11m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/86cb5fbc30114ab8b5ab01f86b02449f : SUCCESS in 3m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/636c016c09d548bf9f25e216f46be571 : SUCCESS in 56m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/538eab7f98fe4e15a497f5355d75ad7f : SUCCESS in 45m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fba8b73f285d4f88928f039ced489820 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b4090b505d49441081e73553525b68bf : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/727c3c16bc7c431ea145b6c87092a30f : SUCCESS in 10m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/688fccc2ab084517a1d3608fbb60b47e : SUCCESS in 11m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bce0f1d7bc6e4e7c891a9cc8707048aa : SUCCESS in 12m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac9f82d8936140c89970856ef439ce7c : SUCCESS in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ead58d17c5a488c8f01205765b06c2e : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c628eac8d2de458a876f394b69f83de6 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f4e3f4f25b74c3aa1d1f477d53792f3 : SUCCESS in 16m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8451268c47374bce94703a19817d2304 : SUCCESS in 25m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d096483cf2a4a0abba518a6ea734bc6 : SUCCESS in 28m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/491c0469c502449dbd8186690b86ced6 : SUCCESS in 18m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f94c1a97357d4bcb9cfb01664517425c : SUCCESS in 22m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8cb1a3db1f2d41b5a4032ed7f886470c : FAILURE in 47m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf8b773ebd3245189b5b01175e651984 : SUCCESS in 43m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8cf34dfc1c6468dac5a326fec69712d : SUCCESS in 43m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8530c09941974a44b47a626fbcd86830 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174be082578146549ce5cb7156d1c4fe : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df42a1235b1545b6837a5e78076685ad : SUCCESS in 23m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49776037c9ec4120b7369b080eaa1fac : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"76a280f2a1e398027d9293fb262c5d7d1d87b03b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-06 01:45:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2dcb1a0ccd44dcbd7131f6ea98847a883f4b292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 03:48:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710d7e00c1f8480195ab299e3777900c : SUCCESS in 11m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e151ce2f3ee24d82a3229c5b3256e08a : SUCCESS in 3m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ec7dc2e9e5e448c5898077cd5225d8c6 : SUCCESS in 43m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20282ab9a7bb48bbb210c165f556ddfe : SUCCESS in 38m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd3dfd75ca1348afbb40b05f1d5e6f9b : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b32fed181ff43e498f301d4ace626fe : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/465f9dc6881141f4a4e4b0bfa9855992 : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fdb89ad74ac04e73bde11acbd879031d : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/56d6f50c8ebe451ea4c55222d367007d : SUCCESS in 11m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d3fd00230da4306a1a547e457fe2f0a : SUCCESS in 11m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0cfbd24069164fb3acf7a237e26c8c11 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b1913e1e6bc4a908ee12e8f32584872 : SUCCESS in 27m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88cb3f78081f455ab4c6e8e84112c443 : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0c6a7a839ecb447ab7a5352cbeb63c7d : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d7a6301cd0db4774879a744a53510402 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cc8138c526494e6399670eaf2e255203 : FAILURE in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a479b54a8926411b932f08c08627556b : SUCCESS in 21m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59ddf9bc2e9c48f68ddc739e516abf60 : FAILURE in 41m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/220fb6bb4b6f48c99c6c01883036fe97 : SUCCESS in 41m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3523271b11c4c899b1ee841ec829ff1 : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4b9cc10dddb4834b0bc4c4bae3ab5d3 : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0be78380a14079a9f21dffe40709af : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae160f92261a4468abe1f6dd7fe74c9a : SUCCESS in 21m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb430b66b73543f18055380aff1976af : SUCCESS in 1h 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5b894fca214ddfb0c5b61ffb41e6e25aefe12669","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-06 05:20:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"232dc83239c73adf6ee28993d7502bed8b9dceff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 06:31:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3096bc5ec0f04c0aaafcb4ac649ca441 : SUCCESS in 11m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7650896285a84957bb9e338c66140a8d : SUCCESS in 4m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0a2ba2ac919247dbaa046c60cd861a8d : SUCCESS in 56m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a11e18775c854ad2a8e610f0c6306e08 : SUCCESS in 50m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab47de031e5b40ff85dd7cb0eb99f82d : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2768ee39f0934a37a068ee9770da00b5 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee9be66618a042948ae7a2266a4ab648 : SUCCESS in 14m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6669b164a1e640108bb9922ce7104fb9 : SUCCESS in 19m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a05c79c1ac464c9fa9efac091641809c : SUCCESS in 18m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cab7acf230004be58ede4c850fc7fd62 : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cd377f0ee2f43cfa60f9de40060b7cc : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f81e4d1c1d7144f890fc5ea57103e360 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56f0ae3c3f0745a5bc81ccea5339377b : SUCCESS in 16m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e6609606c0784051acdce46b35eaf42a : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cdde16e74da646d0a453e4d0326de28c : SUCCESS in 16m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/690deb3322db4a938b61554c1d605dda : SUCCESS in 19m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ea0f2b86cec4365a0173d596df32dda : SUCCESS in 19m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec83e9802c97420eac5f647d680e4fb2 : FAILURE in 48m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22f025182938406abfeafa1feb8333a6 : SUCCESS in 45m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85a2431223b549b1abd47a0523a3a4e7 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e01356046344bbc9fca1a98824417dc : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b73db1f7ef3f45f6964bf5f6083e9740 : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/22919fc5987d40f39c08326184d2886e : SUCCESS in 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee796636d0f94f4db71d5f231badd346 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fefd805c6549e85e58da2cda506ad699f17bb835","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-06 10:42:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b60434bcde50f6fc529d5a5b7fad949e30c8b9c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 12:19:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fcd25f59e7425ca27b570879a96d1b : SUCCESS in 12m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f7a84445370463ab24d35cd60e15901 : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/fde94b643b894a99bcf2bf787ed8b2ed : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/059a788b00b74bd09e0769e8a35198bf : SUCCESS in 43m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/09d48556725c432f8dd287ea2044d6dd : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ca7e024ea65499d8b59e23ba7f0aac7 : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/461b2286bb0941e98cc1f20b11eec0a4 : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8e5f7fd9db447d89479ad2705eabc3 : SUCCESS in 15m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c08bf90e6224fb8a4a7186cdba14e51 : SUCCESS in 24m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/14f04054292f415bb5c03be4b5a07e6a : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8202586a065d477bb2a7ff45f4852a8f : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/249e6a0e57724546b974e9dbe2e3c83c : SUCCESS in 24m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f6aa7a096fff4e3da5bcb1679f11a763 : SUCCESS in 22m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b53685e05e334d3587aad8d71a0f5ada : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/314662e74f684ec7a40ff40c16f8c4ec : SUCCESS in 18m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/06870e2053b44b4ca9a198da699a0930 : SUCCESS in 22m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd733a7c0145464785fc7930fbf0cd17 : SUCCESS in 23m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b497866b23714204b3ef9ce8f0b1bf3c : SUCCESS in 44m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7a9fe3bd14e49f3ac81ef5cce2e5e3d : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/161ba863e9bd4711ba4edd4af3b07bc3 : TIMED_OUT in 1h 09m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abb22835795344b39b1ee9eb85709f0d : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe0083db4ca45cd97e4387b9e31ebbf : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f95016301cab4a029c87ad35729aeee8 : SUCCESS in 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6701fb8f492f48c6adf0865d7c1270b4 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a216f5197457b824308fb25925bf041cbed3c000","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-06 17:05:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70f553fe0c331f624e9898b7f1875e80f164de17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 18:35:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1735535226497da458dddf679fc1cf : SUCCESS in 12m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc9f021ecf434c73a9e2e1badcae3ae9 : SUCCESS in 2m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8c7bf38079f34ce182de9341888f3561 : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/167028b4cd814f22b8c4bd69b2e6a198 : SUCCESS in 54m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5229928e40324286b031ae0d32575cac : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebfa8f786ee44ed1822e198deb868de5 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5c194506f4e46a0ac8a50598dc777dd : SUCCESS in 12m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e48a0dd69f0244e4986a7714b992568d : SUCCESS in 14m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aad5f75012b149c4b783caddeaef6805 : SUCCESS in 13m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5bbef930e0624a1ea31d522c2227f77b : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/874a10d5f2fb484c9685c4dabc89fb86 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f12b1776df94a4da2e8179b5b9d4f65 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72b1966ed3194504910be46eafd9ada3 : TIMED_OUT in 1h 00m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fcc2ff7db8c346aeb31e0f5b8e9d5f22 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/853b6542fce141af9700845848eebcea : SUCCESS in 24m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cf182e2ffdf14ac0ba97b354065bf9ab : SUCCESS in 17m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b46b57d7427c4bdd8a778c91fc63be37 : SUCCESS in 19m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/241f4e9619884914b3b7ee2dc00ceac7 : SUCCESS in 43m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47262d753f374052b7697baf5463cf6a : SUCCESS in 44m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20b83000e8ce4408a5435ceedde26cfa : SUCCESS in 49m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a95f01a08ea4e1da2a0666e9e44e797 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fefd1c39e02240298dcc96140852bca3 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee0f4938cfe343ba84ce6655bf5a9a13 : SUCCESS in 22m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/586a7bfd1ad6466791ba7d8505ecebc8 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7facd5a9d06e089d3cd0d21bc852a50988919d13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-06 21:23:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b698053f6add9668f1bfb2cf66bba8caba01127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 22:33:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1abe5aa71449ec9a91f6b3946b591d : SUCCESS in 14m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9317fd41b044ab09157c3ca83d47fe7 : SUCCESS in 2m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/acf5b9e127244e788450f28df6890864 : SUCCESS in 1h 03m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c622ada6e26444c8b466b5b25e75538 : SUCCESS in 43m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0b887c2010044238a7f9d82ffd90cc5 : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c7cbbf9f5904efea2337428616583c6 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28f2870d64bf4c968ab2e64f30dea072 : SUCCESS in 11m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80bee4bfe5024c27be7db10089e29873 : SUCCESS in 11m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40da8a5d97ee4f54932fddd3e4670c3d : SUCCESS in 11m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c65de499e431460693cfaf4b3bf32ab9 : SUCCESS in 12m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4580ec957c694d09bb5482043e740525 : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b3002e64ae1e4f3698367b741b04aace : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8aa9d47cf94e4bb3bfa83e1dcedfbf69 : SUCCESS in 24m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/36171021bcd24287b8cf0e08e6337ea9 : SUCCESS in 29m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f1e1b72151144a89b68f940932b8bfbf : SUCCESS in 17m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/07647ca4980b4369bfa1283f1f9ae0dc : SUCCESS in 18m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3855cc33d2af487ebcb043bee9224a22 : SUCCESS in 53m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cbd05816367345339fdc56dcee840c7d : SUCCESS in 43m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa25e2c774294c2cbf4909a024f865a1 : SUCCESS in 37m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77ee1fbfe654464187f5b10565d2eda2 : SUCCESS in 35m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f187bd28f9124fcc94579d1c447c2f69 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db807eed1d044ea3b718cb3584a11118 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b02917b320e4491aaa7014eaaeb8002e : SUCCESS in 21m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c2b70de67b2429aa5dc0a5618dc7b55 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"76ed11af933d2eab0ac123a95f101b6600ba0d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 22:33:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0dbd84a1aba6c81565dacdf6509b1f724077770f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:42:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5da6402cac04c2d873ef8c9e08ae0cd : SUCCESS in 11m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2b151d87f9f4b6085e2db48fec8cb9b : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1173468f400e4ea8b8a71065a0edce8b : SUCCESS in 1h 01m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a35330f0a3a04c6689c1ae32d2eafdf0 : SUCCESS in 51m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a1362a5bd4c34eb3912d270770270605 : SUCCESS in 9m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6a924a4fa2354809b03775a6e0270015 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a48feda5338a4680a9c81600c7d5a050 : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a92fb28a35245cd9557435297a090d1 : SUCCESS in 13m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2e5219f8572461d92d7ea1b14e1b648 : SUCCESS in 13m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e4e155ef7fd04abe854f5d3dd4ce377e : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/16559795484141958aa0fd4767439ce5 : SUCCESS in 22m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b126954b0d6846998dfbf2da8b84551c : SUCCESS in 24m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a3f62e78bb941869f4e9d09d7cd5b7d : SUCCESS in 25m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9f404638f8164da7bbd014bd0e341c24 : SUCCESS in 24m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/657cad1a44db4519a3dc1715420e0807 : SUCCESS in 24m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ce99ce028d8442a2933dfa85535f71a1 : SUCCESS in 25m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e03d7bbdbacb4a1fa5f21594ce5a013b : SUCCESS in 42m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8bec49d1bc34abe9340e8d529a84b5c : SUCCESS in 1h 00m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9ce667f5dc6496d9aedc372061991d5 : SUCCESS in 40m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cec55e7901a842cc81ceb0b7f99f7acf : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3074ea1bfe26434ea2ef693a798c53be : SUCCESS in 5m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e48205ce7e9a4ebfa506f0c9916475bc : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a925e7fc56788c495f6c6716b38c173d628c78a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:42:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c7b25b733e2a109264fe513d9758c6411529051","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:43:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd813ccb3b3483b9af8d5dba81a5767 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1e7abfc9bc6e4d77ae45bada10e2e387 : SUCCESS in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d4f4b83de0ba40f8b8767667081b4590 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f97d8e920405ed9c21c9789c62ab60a615603ce","revisions":{"0f97d8e920405ed9c21c9789c62ab60a615603ce":{"kind":"REWORK","_number":1,"created":"2020-10-27 03:37:56.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/90/759790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/759790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/759790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/759790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/759790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/759790/1"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-27 03:35:53.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-27 03:35:53.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nthey are not needed during installation.\n\nChange-Id: Ic777b435723f663b408c778f661811439b486071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f97d8e920405ed9c21c9789c62ab60a615603ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f97d8e920405ed9c21c9789c62ab60a615603ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
