)]}'
{"id":"openstack%2Fswift~759898","triplet_id":"openstack%2Fswift~master~Ic1d699c077ac2aa582d799b3c9ebc61bc4338049","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic1d699c077ac2aa582d799b3c9ebc61bc4338049","subject":"Have PrefixLoggerAdapter prefix exceptions, as well","status":"MERGED","created":"2020-10-27 16:09:50.000000000","updated":"2021-01-08 01:50:53.000000000","submitted":"2021-01-08 01:48:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759898","meta_rev_id":"31cd802dde47857b49decb279c346aacb26ef0bb","_number":759898,"virtual_id_number":759898,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 01:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-01-07 04:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-07 18:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 00:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 04:00:46.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"c3237f38dde7a1073eaeaaa6dd818fb744423a41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-27 16:09:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02aaf734ec122cdbf54a9c0c51ed64000da2f2c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 00:21:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ab357cee024077af1cc2ccde4a440b : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/680dda6a27544cd68c9eecfdc539d6ca : FAILURE in 46m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c63b4e5688545f5b24be65ff839e438 : SUCCESS in 1h 32m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/268bc732fedd41fab5b62e0d428a63c0 : SUCCESS in 9m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0f06c2ac1f844adac3c48434133c1ba : SUCCESS in 9m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83ee143d89c44ff1be45608098052dcf : FAILURE in 15m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7526ec8db87b429e89a21aa93f1dedb0 : FAILURE in 24m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fe04e97ecd34fcca9c0b39b8829bc92 : FAILURE in 14m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/240ce9c909d94753997ad45990804685 : TIMED_OUT in 1h 01m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da8b891ddc6b4506a6deaf7eb535c19e : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f2c40906e53f4b97b993c7e57477dffc : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bbe732dd4f75406fbee0a55e0146b864 : SUCCESS in 18m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c40adf65a3c94470841ec575c1e7c6cf : SUCCESS in 21m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dd5558cb1b354fe5a539ecaf0ca124f6 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/19d9c8f545da449a9362ec09b5f5a73f : SUCCESS in 20m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/add177e6b25441d182e5f1211e072ff0 : SUCCESS in 18m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a478729a0b5840929ce6a98f09392c1b : SUCCESS in 45m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f24f2f42da1e4bd0b2496e56c36533b0 : SUCCESS in 39m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b78cafaa4644846b795dc04632cf988 : SUCCESS in 39m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4dc409ce3b24cc8a8caf104ad8dbf01 : FAILURE in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5263d19934d43c5abcf3ac07caacbf1 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e9810b24ed14400a5c6b9840ee90011 : SUCCESS in 20m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/899ee1c6e1e843eb9d84c811542658d2 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5b17aa0ca0fc64c46180597732e9009af8a41f87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 17:07:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c8d72b998a128bc27898c70ce01681d000227b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:08:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71cc03a7dedf42d19725f83ba3150239 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/6bfecbd0bd6149fe913c3c0444597e4d : SUCCESS in 1h 07m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c85c9da8c604c8c98f997ab778ee795 : SUCCESS in 42m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b5182093a1a42299e67efa57029ec38 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0775d60bd34432cb984451866b61c27 : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04027b00f76e4b2889c15b403ac70359 : FAILURE in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e410a7fb870e4ff29374ac6123194a64 : FAILURE in 24m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c12e03d2727447b3a1f9902e9278b153 : FAILURE in 13m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1bd6fca389e34555a06875f818245e8b : FAILURE in 13m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9e4fd2428d049af899a1de20296a04f : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/844e64378a1a4f61a3a8719d74be4116 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b64b0a5e2c54db7bd57d44adb3693af : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9b9d597a65f94fa682bda2c3e8d3b142 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b35cf4b04f94486f8e2a201ebda39794 : SUCCESS in 19m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7c1644b6015942c7a5732ae103be90b5 : SUCCESS in 20m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c54b1ea5265b4dc3b33b3af17ef47c08 : POST_FAILURE in 5m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d39e69f7bc5d495e85342267d946e428 : FAILURE in 45m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52caada7485b461e9ad93734cadbd4dd : TIMED_OUT in 1h 02m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9032b940a9fc4a34b412bf0e66f83c57 : SUCCESS in 44m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2799d70eb52473e8306b44c04a4505a : FAILURE in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66411fd4afd546319c60ce87bb58d7c9 : SUCCESS in 7m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e6e14038915460fa4cf05b57c04c3e5 : FAILURE in 11m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fa31180d1fd486f9c5d4abda4abedd4 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fc377d97094120d5b5ac75f654cca80e1ccf8e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-31 05:40:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65418013444d334fc82bc30e2b97ff26431fa1b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 06:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a74cc04adc546cfba151acf5575e79b : SUCCESS in 7m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d870f9e6cd30481698e692c0e9958461 : SUCCESS in 50m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b721c937ddd54981bd6f6ced306ca0fb : SUCCESS in 40m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/110aed3e6b2b469385454ade130a2cf6 : SUCCESS in 7m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ab26e72181648a0ac79561af9fba546 : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b2b127877606496e81922bf9aca1a16d : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c971d9d18c2c491bb0decc8fc11ca696 : SUCCESS in 11m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68454ed4ff374dff96dfc4f825f7718f : SUCCESS in 12m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f046e2d1534f4265afdf1a0f6661e885 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c6993aa24814fcdb0edb081bfdfbc34 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7400c2aea8a48709fa2d7660fb5529d : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05d35871458e462ead179ab8e20b3dc5 : SUCCESS in 16m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/027f9319184341a1aa2a9617eb2f0838 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d798e5fd45b84180a142ecaf7bc2425b : SUCCESS in 16m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ffe7728a302a434699daefa0c22e7bb1 : SUCCESS in 19m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c5ec1615a8a4c868fc49d137cde3568 : SUCCESS in 23m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a8454e763f9e44b5aa7753c41b252d7a : SUCCESS in 50m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6d8d446ee3364b4f9703c8ed83875676 : SUCCESS in 42m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50a111a5482245d3817b03aa99d0eefa : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0385c9c131c14e15b41ba251834ddcd8 : SUCCESS in 33m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf289c7abcee4a04a0c6e52cd363b27f : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddad707bbf284672b9743e25ba9f8c90 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f97a66c3e1eb457d91a34b86a3d01c90 : FAILURE in 7m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c790a5ab8ff44399bfab614d2d6c2ba2 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ea2539b2ac068ee3849b7d58c641349e0d5a9c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-06 00:06:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"898df92d74ebb5d5f06e6f5810b5e707c0809d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 01:13:43.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd862e0e33f473eb2522ff2f21ab3a3 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/eb31fac1b85d4ceca94d18517e569e2f : SUCCESS in 52m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48d9439cbb624556bce0461bf0624de1 : SUCCESS in 48m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07ad61156e4d412688b35541eccbc6b6 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cabc5d8ba00a496cafa9978bea3701a3 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24d5ff89b9e54e808d03ead49f0f06ad : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/733b0e95683e465abc1755e78bf74585 : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1120d280e66a433cb7a7e530cff34186 : SUCCESS in 58m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f220c03aa4b4e0584f30b5fe2593df6 : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d0ec612ca884471bf23df72027ed0ec : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0befbc6654b148ecb893ec6fef237af3 : SUCCESS in 19m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7573bbf4e32c48fbab8843a99321cb00 : SUCCESS in 17m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46de3cc0b62c42419c4aeef86616c166 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/771f1b433b6e49debb496c527c00b69f : SUCCESS in 17m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/210dbc6ffad54910825d72cfd8ab26e5 : SUCCESS in 18m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd0d50ccba044f399dd089b8458ac422 : SUCCESS in 20m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e2859460e9044055b042405386bdc718 : SUCCESS in 44m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/98639691c885443f8e87076cdd017576 : SUCCESS in 44m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4b2a4e8e8e14a04bbc41a408a6471de : SUCCESS in 39m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/090ad60ba59643b594f6e31146e8598d : SUCCESS in 37m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/439f002e0c7d473fa04cfa8ff7161bd3 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea71ddffa78457e88d2562fd92b8b16 : FAILURE in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a583940c2e4749a1a25548464aa9e36a : FAILURE in 9m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eee5f744fe3348e49376c2f027a54a07 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b1be468040b85475092973058a681ab415cfa404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-06 05:51:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"740522258a66504c2083c14496b45802fa6580a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 07:03:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9b841aa157483a9ff4888afab5e176 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/21b2d104c2af4dc394c8c5f418f81d81 : SUCCESS in 1h 07m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39ffc866b21f42348afa5dec032f7d96 : SUCCESS in 48m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4aa03b9da038410895d5dd1adfe3864b : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cce37304b9ee44a884d0303474cc0338 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb87d45ea9b14bde8db393ac5e06c595 : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f577fc32c114ef4b2713b504a2e5d8a : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aab84b18caef48409d26c10435fe1a65 : SUCCESS in 13m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c07f7a4beab64f0686aecd49ef0d4071 : SUCCESS in 12m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c722a58a618347228690cfbe10485858 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b47844473ba4aa582a000dcba52c91e : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f7cc0e33e3348bebc6ec2b45b68c4af : SUCCESS in 21m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e5b16c7ca81f4fb9974ef26873791af6 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/befc842c6cd641e98adb753364a23553 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e80909e49d1541ef9b7b915bfe2de70e : SUCCESS in 21m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b49eda617f84572bff89ef2fdde327e : SUCCESS in 18m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1f721d44afa4b0e8af6f641a775c6aa : SUCCESS in 45m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/35e05d71272a48e48db6db0e4ebf037c : SUCCESS in 46m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/27d20aa14ca249fa96df3177d4c457ac : SUCCESS in 35m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3dce33cfb8c44989c04b81792b423b2 : SUCCESS in 38m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/855e1afe06d6445891b610f3744c5229 : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a4784cde0148829249c27ef0eed0de : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a54358f0fed54fb0919f7d43b7391b6b : FAILURE in 8m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7f954d819454b32ae7234b3ffcbde76 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9f9bbb6d8c9cc85fbf70cc63d50b48406bb277b0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-07 04:00:46.000000000","message":"Patch Set 4: Code-Review+2\n\nNice, this will be useful in the watchers or rather anything using the prefix logger adaptor.","accounts_in_message":[],"_revision_number":4},{"id":"41c744539e083214cfe03d1997a3e5e2504e4173","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-07 18:49:13.000000000","message":"Patch Set 4: Workflow+1\n\nYeah, I think watchers were where I noticed this first -- I was scratching my head about why the prefix wasn\u0027t showing up for my NameError on a typo or something ;-)\n\nI\u0027ve also noticed it in multiprocess replicators, though, where you get a message like `Error syncing with node: ...` that includes info on the destination but nothing about the source.","accounts_in_message":[],"_revision_number":4},{"id":"05d69eaa3c1e8bf410e1cc0252dd08b98b83cfe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 18:49:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bf994badadd15ce3395ff4db9b1a9a2014df3e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 01:48:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08654d3116014f5f8af6107ed84582f2 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/26f31b48e07f42c893fa7c78ffca10bb : SUCCESS in 1h 00m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38d5822cdfa14f1d8103689e0ba0fa8a : SUCCESS in 37m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4488bc5de2b64e8fafbeeb1d83500d35 : SUCCESS in 9m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c874f1b7cdc64159b63422fd897b941d : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10fcc2ea879d4027b9db078e4db02623 : SUCCESS in 16m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a72bf518499446792ecffbcb0f61b88 : SUCCESS in 12m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/87f31446688846d49083a59a4c3fba5d : SUCCESS in 22m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/63165a7aa5eb457993613bb1ae718d34 : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d00076b9e534e7fa51d3e016a60fa5e : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3bf07f1896cc4e7f88944c9b26cff7dd : SUCCESS in 21m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/827a8f3b26834682a09df36049e00b12 : SUCCESS in 18m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1224bc0667d14b0796b4059604cba930 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/99cb277852ec40df834e329fd11f0762 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e9b63b46f85743c7a2433a23dcd0986b : SUCCESS in 19m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f15410e59b14c349b126db1ede70c6e : SUCCESS in 53m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/772d263047084ed185292aba6fbca6e8 : SUCCESS in 45m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af33b53f4d8f4cb0b9b0de0b5d786c2d : SUCCESS in 44m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ebf5979d6e64fc3a9118988f952a79a : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f122a68f884844c496d9456f879009e4 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0b1dde85004bd4a0c11623519eab62 : SUCCESS in 6m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1cc549168b24887b292b4fa704c9954 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":4},{"id":"0107609d4addc06937b28fe2bd2eb76c473dacc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 01:48:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"31cd802dde47857b49decb279c346aacb26ef0bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 01:50:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2100ce9b516d40af98b6c4dbc85e1d42 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f1a62f0082334eb696708f4dad454e28 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7d0f8614579c4f199652b0df13c45d58 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cc69e5b08cd7bd42715eaff25120612e93c8a59a","revisions":{"c6bbe01a7a5e60068a693d5e49d6c4dcaf2cba2b":{"kind":"REWORK","_number":1,"created":"2020-10-27 16:09:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/759898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/759898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/759898/1"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-27 16:09:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-27 16:09:43.000000000","tz":-420},"subject":"Have PrefixLoggerAdapter prefix exceptions, as well","message":"Have PrefixLoggerAdapter prefix exceptions, as well\n\nIDK why this had a carve-out before.\n\nChange-Id: Ic1d699c077ac2aa582d799b3c9ebc61bc4338049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6bbe01a7a5e60068a693d5e49d6c4dcaf2cba2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6bbe01a7a5e60068a693d5e49d6c4dcaf2cba2b"}]},"branch":"refs/heads/master"},"37698768f9e96ddedafa4bca032f358d2b75d6ed":{"kind":"REWORK","_number":2,"created":"2020-12-31 05:40:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/759898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/759898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/759898/2"}}},"commit":{"parents":[{"commit":"6b7eecab5de70b454fa836515183432e1a97359c","subject":"Get unit tests passing on py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b7eecab5de70b454fa836515183432e1a97359c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-27 16:09:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-31 05:38:00.000000000","tz":-480},"subject":"Have PrefixLoggerAdapter prefix exceptions, as well","message":"Have PrefixLoggerAdapter prefix exceptions, as well\n\nLooks like this was overlooked while we were getting support for\nTimeouts (etc.) without tracebacks?\n\nChange-Id: Ic1d699c077ac2aa582d799b3c9ebc61bc4338049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37698768f9e96ddedafa4bca032f358d2b75d6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37698768f9e96ddedafa4bca032f358d2b75d6ed"}]},"branch":"refs/heads/master"},"d0fa0abaf0d478ba9a90647110bf32dc1880ce30":{"kind":"REWORK","_number":3,"created":"2021-01-06 00:06:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/759898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/759898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/759898/3"}}},"commit":{"parents":[{"commit":"92b15b0436b2246e658af94bdc312bdffbebfc74","subject":"Merge \"Let developers/operators add watchers to object audit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b15b0436b2246e658af94bdc312bdffbebfc74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-27 16:09:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-06 00:05:56.000000000","tz":-480},"subject":"Have PrefixLoggerAdapter prefix exceptions, as well","message":"Have PrefixLoggerAdapter prefix exceptions, as well\n\nLooks like this was overlooked while we were getting support for\nTimeouts (etc.) without tracebacks?\n\nChange-Id: Ic1d699c077ac2aa582d799b3c9ebc61bc4338049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0fa0abaf0d478ba9a90647110bf32dc1880ce30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0fa0abaf0d478ba9a90647110bf32dc1880ce30"}]},"branch":"refs/heads/master"},"cc69e5b08cd7bd42715eaff25120612e93c8a59a":{"kind":"REWORK","_number":4,"created":"2021-01-06 05:51:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/759898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/759898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/759898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/759898/4"}}},"commit":{"parents":[{"commit":"92b15b0436b2246e658af94bdc312bdffbebfc74","subject":"Merge \"Let developers/operators add watchers to object audit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b15b0436b2246e658af94bdc312bdffbebfc74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-27 16:09:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-06 05:51:05.000000000","tz":-480},"subject":"Have PrefixLoggerAdapter prefix exceptions, as well","message":"Have PrefixLoggerAdapter prefix exceptions, as well\n\nLooks like this was overlooked while we were getting support for\nTimeouts (etc.) without tracebacks?\n\nChange-Id: Ic1d699c077ac2aa582d799b3c9ebc61bc4338049\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc69e5b08cd7bd42715eaff25120612e93c8a59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc69e5b08cd7bd42715eaff25120612e93c8a59a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
