)]}'
{"id":"openstack%2Fswift~761659","triplet_id":"openstack%2Fswift~master~I5fc696625d69d1ee9218ee2a508a1b9be6cf9685","project":"openstack/swift","branch":"master","topic":"p-cache-shard-listings","hashtags":[],"change_id":"I5fc696625d69d1ee9218ee2a508a1b9be6cf9685","subject":"Use cached shard ranges for container GETs","status":"MERGED","created":"2020-11-05 19:44:30.000000000","updated":"2021-01-08 20:52:58.000000000","submitted":"2021-01-08 20:50:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":148,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"761659-p-cache-shard-listings","meta_rev_id":"30e091bb3caa08ba5985be7156e6bc511587c67f","_number":761659,"virtual_id_number":761659,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 20:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-01-07 03:30:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-01-08 17:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-01-08 17:12:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 19:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 23:28:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-12-13 23:14:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-01-07 03:30:58.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"9bd28d13a8b30d3a554f01867e07fd15eadb752f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-05 19:44:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8001855073bf5a934d5335577979d2774198b28","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-05 19:45:00.000000000","message":"Patch Set 1: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"33a40107b5c3d8b6b9ff48c4823f7d79627f8f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 21:57:26.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50fd8ab58d547369418b628510e5be1 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/da6c7c2567c74e3babe481328d4f9d9e : SUCCESS in 1h 04m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43bd174b27894bf29531da68a4a7faf3 : SUCCESS in 45m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/547c1332c1fd49aea3977eed24a97836 : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a4f7a22cbb234da79c7447ba5571079a : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51a9cf1b50344867b7d88ca8e7c410f4 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff6744f7b1344c209141d5af3e603ab3 : SUCCESS in 11m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2e7a8a37931494aa59a8fe47feacd97 : TIMED_OUT in 1h 01m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/17ee15815c484c3ba8352d67055560a9 : SUCCESS in 13m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cedbe3d0b40e4773949d37c0593c403d : SUCCESS in 21m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae26a05661c14b7dbb12319b33e7f01f : SUCCESS in 24m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/edd72fa0869e4c40afc231b924e87222 : SUCCESS in 24m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c636dcbdeba14458a79b36b886dd7577 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/505ddac1429842f39360fd64211f2096 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/044bc48ded584cc1bba79d6c7e6b3731 : SUCCESS in 20m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40819df8a157446f92ae706ab4633b6d : SUCCESS in 23m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6722a55c5d6c4cb7b79dc2a3c1428865 : SUCCESS in 48m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3aa0f3cd49e4840a88cec0005594548 : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2e3eec0236a4fdf83738ab09cacb0cc : SUCCESS in 39m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8391354fddff45e089cf3c508c05752f : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f893b40a00e04c5aa776f9b9ffc9afc2 : FAILURE in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89ee61b6326b44a48750214ac65f91c4 : SUCCESS in 21m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1677373c399e408297fb1dadbf542882 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"204c00c51d8ec89445b41bba2f16d34f17ab9ed7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-11 21:28:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac52a07c77fa55e8f75a5d89cb9b40caf1608b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 23:03:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6757f89589d14eb4874be4224272ff62 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/934569f0b6494070831f95debd8b763c : SUCCESS in 54m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b87f7e369c54aa2bf56cf444f19f543 : SUCCESS in 46m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4867eacea4b148d29d7063bc2ac05cba : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0cf7522d30044a2865b55784daaf3b7 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d31658b7762a456f858baaa5f826209c : FAILURE in 12m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/961dd95f2d804995b6281d18e9696f6c : FAILURE in 11m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4ff82f8bd0240668bbc08e0027f0faf : FAILURE in 11m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b646203a0a64fb3b3489fe14fdce7d8 : FAILURE in 12m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a933c68509024f12a649754b7d009b50 : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/863f387371de4d56aa06c68d4bd89e32 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c798ec9206844c0bd1c6c89610cbb3e : SUCCESS in 17m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c130e2736b224bd88b843d24791a1815 : SUCCESS in 27m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b52e339dfdbc4f26ba9b18be0c3795f2 : SUCCESS in 20m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/646ae13c796b44f9bb09f2e7412e09e1 : SUCCESS in 19m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3ab44950b5a44f9a5c9b1a3b713c6d6 : SUCCESS in 24m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7183f0b480c947ccad3222150c1f1537 : FAILURE in 53m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb1744ee4010483bafbfde06048de9e9 : SUCCESS in 42m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ff8898e038442caa3ccd9980f193292 : SUCCESS in 41m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb4b18a1ab8c463d827d8b49c9158dbe : FAILURE in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af75bdf85894a4390af910bc4318a0a : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1881e709769f48c6ad10a3db5402c749 : SUCCESS in 23m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61303c42ac27405186c445925962bf40 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7610d6fae4423923f2d506c7e1a1e450d239d35c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-12 15:38:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c02cce05a08cb185f7a48690fc0d3d7adb141d38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-12 15:39:19.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"cce7efedd9e1c9828c1dd511f2b0ca3b33eb29dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 21:17:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0360b93451fb4f14a05de74b2ea01eb6 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ec67dc84409c4b33b1a7854043acb9fe : SUCCESS in 59m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53cb766ea2f049cbbe62b99833d0f3cb : SUCCESS in 39m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce965e0f566542eb8f7da9b152a04f3c : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0c0f3f315db49baa5a2746cf037e914 : SUCCESS in 10m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ea4655f8077431eb67c691144c6dd13 : SUCCESS in 13m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb898835be314831a2a449bd47b844d8 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0319d6aba08443e885ec8b9168e2989 : SUCCESS in 13m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f0593fa0c2854ecfa62dffa7cb012149 : SUCCESS in 12m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fcb79a1d13348a7be7f13066961825e : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/821bb86f7e0f4f7e8aebffa214a933ea : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b450d69e45d42c1a9b9cea2043fade2 : SUCCESS in 21m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/730c5d4af95d413397f54a2b3c992920 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aeff73266dc24555989611356d48dd61 : SUCCESS in 24m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a85d165acea04f4395091d51ca1af519 : SUCCESS in 20m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db889d7400074f618be1313ef031f580 : SUCCESS in 20m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/72da422ca19945cd83a3f9b31b452e89 : FAILURE in 49m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ba4a5b4c9e74ba3b9fb4efbbf2bc4af : SUCCESS in 39m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62bc1b8f890a4844a25527f8a2041d5e : SUCCESS in 47m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbda574eb9ac4e3f96a8dfbc100d82a1 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2f75018f3b45e5aab052dac8f46e73 : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/486d1bcfe96b488d8d229161168bf948 : SUCCESS in 22m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff64ac0d4fd64b7698dd5a6123636ee5 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":3},{"id":"cfbb1592c863a8e758202f83d27d36ca7894c1d1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-16 18:00:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"62b51d65c9eebee2c5c28732caf7c03a9e10f784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:48:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63a3f0431264c919b0c06381c7aebd6 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/bfba371eea9b4ba0ac70b05e1603631e : SUCCESS in 51m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96e653ab0e774095849f05bf9baa0e0d : SUCCESS in 44m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20f6ec6d424c4a5393142ea4dfb05d1a : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/082eb362ed7b496aa1d7365881a3086c : SUCCESS in 13m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/710661888a6140798f5808269eaaedf5 : SUCCESS in 12m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1dd70ef2ca1448f7b08ee55ed2121d6a : SUCCESS in 18m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0ada9d52a1b483191563dd3580bd89a : TIMED_OUT in 1h 01m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b1991c9baed742268153fedbed25bb67 : SUCCESS in 11m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be20c3229fff45b6995a84911ad71a1e : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d69c572b6bf4fa0bbbdbfd2fe9270f9 : SUCCESS in 21m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb2fd3e888e7420d8e8b7b04fa0166c3 : SUCCESS in 20m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e54ee7915677428780f177992730a88f : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b79c2118bd054b568bde0210a1502c34 : SUCCESS in 20m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/360ad3a0e2ff4ac988360ed320d0950a : SUCCESS in 19m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad0add900a8243fe8f156866d1779b04 : SUCCESS in 24m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/399693c5e4c643a8913f9c563339f55a : FAILURE in 1h 04m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89af214ec49e4df0882e48556116661a : SUCCESS in 45m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb533b60c4f94eca8250b2d4eeed67a9 : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2bff860866a4f3fbf8815195174c605 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0a8cc867db40098a78f082dae07f54 : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4105058319647d2bd3c33b3084aa7af : SUCCESS in 23m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/773e7764875f456082009000b9e7c779 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":4},{"id":"4a02c0be75bd5c2e7d2fb69addc7054eeb94d261","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-17 23:28:09.000000000","message":"Patch Set 4: Code-Review+1\n\n(24 comments)\n\nI think I about have the big parts loaded in my head, for a \u003e1K new LOC I think I can already tell it\u0027s going to be great.\n\nI had some issues with new test(s):\n\nhttps://review.opendev.org/763102\n\nI haven\u0027t been able to find anything wrong with the approach or any bugs.  I was mostly playing with client request params and keeping an eye on memcache [1] while uploading new objects and sharding new shards.\n\nI\u0027d probably ship this.  Nice work!\n\n1. https://gist.github.com/clayg/535b2f88384153e5ef3d92775aeacc1e","accounts_in_message":[],"_revision_number":4},{"id":"c72e81f0c0873abbe4eaf5b524bf5cdc9f6f9ef0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-19 17:12:01.000000000","message":"Patch Set 4:\n\n(14 comments)\n\nworking on next version","accounts_in_message":[],"_revision_number":4},{"id":"718f12018d7e31327e9d9e66575481157c04bbb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-23 17:36:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27230eb2ef7bb043340bbaf7692e43b6419af36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 21:44:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e56ebda1a84106b6e520b6c48c270a : SUCCESS in 18m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/6ef5f0297a6843c8b71d145cfb212ee5 : SUCCESS in 1h 47m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d78c66ba6a1475c8299a570f2d1dd4c : FAILURE in 47m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60f68dc28eb94738a88fce831671c18e : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/50246070ee344790ac4fadb341f13006 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9dfdbc8f3c0a490bb0c6b4b43697903c : FAILURE in 4m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b15a74e3e24b49c497341978b926a806 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/52e74c941fb945dea750b74b757ca1cf : SUCCESS in 12m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/82c50db4817941098a49f185438f19d1 : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa1a6c6f29dd473ead61908650c6ff64 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3cde4b2fdf914060b41040d2ad1bc6ef : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1eef7dde16594ba6b7a44411b0c93fcb : SUCCESS in 19m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2673e7177f214bf3a5dcf2d2f854bd45 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/725e65914c3246d7b96c3ba7e51bd485 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/902259050e64412d9d71a8f0e798b901 : SUCCESS in 19m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/612f41630272496394cf08754357bd35 : FAILURE in 22m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/326ad94ee7ac47569bf9d7c9ff3eff0c : SUCCESS in 57m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01d81c9b781640edb4d8e4d6e447edfd : SUCCESS in 39m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d455a535fa84c45b2c0baa1247baa90 : SUCCESS in 42m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff65916f0e5f48fd8a042e88872092cc : TIMED_OUT in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff566df126e74898943d25d34cc4d1df : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7385c8d8e550444db977069c275ade68 : SUCCESS in 22m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c947af1139b42b1b6f9ea30995b439e : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":5},{"id":"22154d50eabb125c157e9264d937433d9adf6e86","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-25 20:27:14.000000000","message":"Patch Set 5: Code-Review+2\n\n(21 comments)\n\nwe\u0027re gunna ship this and everyone will be jelous until we +A\n\namazing work!","accounts_in_message":[],"_revision_number":5},{"id":"44b8d607aa3f00c745595c3adc7b5e0306c9ae62","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-04 18:25:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e61db1835c64f55b26f3e672f95e01af5ba4ea3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 04:56:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28166b63e59a49c69374a3ac42bd60db : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9361c24ec85649fc9d0275181fa205b1 : SUCCESS in 1h 09m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06088cda68e64d608324c1fddd8aa96d : SUCCESS in 46m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a93d4de344ba492e88d8d933c1d002c4 : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/84f5df53c67c42e890a312a952f57f65 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59e5c1152a42424dabced35989c0ea7a : SUCCESS in 14m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b67d6c567e4942beea6325c9c8e9e5 : SUCCESS in 14m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5e7191f7e4a547f788ca9a8552d8258a : SUCCESS in 13m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf6ff3e8cb94451ca5145f3db7b419fa : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29eef2712afe4efa8fc2ebfc2f00fd59 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa83817bc61e42ce8a6f3cf5f440273a : FAILURE in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e573c9194deb438c8116bda16af30220 : SUCCESS in 17m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/746e4740784d48a29f28a87b6801c7ab : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/005eb6761fff455680a2e457330ce725 : SUCCESS in 39m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9d282c393d8c43468992ec47d03aaeae : SUCCESS in 19m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5a594fc4dfb43e785a98d4e3a3fed26 : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc36bd43eac94a709f9c85bfd07cecde : SUCCESS in 1h 09m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f709f29eb87c476788f68ee67210f5cf : SUCCESS in 50m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bba957e525f44199a76236adc0c018c1 : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d903b40917d24e19aaae0ef102812ae3 : FAILURE in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780db475ca074d4c887ff3d73494be80 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95d7434124164b75968273102581bf51 : SUCCESS in 23m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1984b7784dd44cd80141248b71638cc : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":5},{"id":"6e8d41962ce758b293045a60b8869f7b7d437fa6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-13 23:14:13.000000000","message":"Patch Set 5: Code-Review+1\n\n(27 comments)\n\nrecheck\n\nI need to just post this -- I\u0027d meant to finish my review on Friday, but got waylaid. Seems OK, though, and certainly showing improvements in production!","accounts_in_message":[],"_revision_number":5},{"id":"bc7ef464d2ed3eaac4f5951f0a3f25cd5501afba","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-15 20:24:10.000000000","message":"Patch Set 5:\n\n(47 comments)\n\nThanks @Clay and @Tim for the reviews.\n\nThere\u0027s one issue Tim brought up that I propose leaving for follow-up. If the shard ranges fetched from the backend have gaps, then perhaps we should not cache them. We should use them to make an object listing just like on master now, but maybe we should try the backend again on the next listing request.","accounts_in_message":[],"_revision_number":5},{"id":"03ccc300065d28dd451f69c2f79721d388a0673d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-15 20:55:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f4d205cdd4d56b7ca4a9a3fd6d5dd274190be21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 17:32:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f39190b5a604e7e9e05462686e2a266 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/af497f46165249feaf36459733956297 : SUCCESS in 1h 04m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/99153bc6ef1549cea7bdf0d92b4fd849 : SUCCESS in 43m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1422239d71a24b9a83cfafe631eb6f10 : SUCCESS in 11m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c697fe404f024e56b8435a3b6511cf4b : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/283593fee2614077be5970717975e396 : SUCCESS in 12m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2a4c976b85a4545aed2063a8acf1529 : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/33af021dc3c3495aaafa0134893f6d6b : SUCCESS in 18m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d50f9f9807b54abd82a902c90131e759 : SUCCESS in 18m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59b4b3e7ff894c9aa5969c36c798c06a : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2cdef9f2dce34397967a0be84bbb7504 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8756c32e3ff043f0825071ee290b4ea2 : SUCCESS in 18m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/652f27a7f2524dc59755f64d6943b6f5 : FAILURE in 17m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4ae1624871274a949ffa938d1ed319a6 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7545ec6eaea14bc4b70b9bb31d430fd2 : SUCCESS in 19m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3dad4b621c824a9ea0055f2b8612735a : POST_FAILURE in 4m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/58dfcd6635d74e80a8ea494582d37654 : TIMED_OUT in 2h 00m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f16121042d1c470c8d0e94b4c18a71cd : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90520db0484d4743961ab31441ced7f4 : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01700def8b4c4d419bbb1650a4d1d6ed : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9578896177074b1690de3de055266f90 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4276281fe32c4906aa49c1f27ee6bcfc : FAILURE in 7m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5539eb78bd8f42cda63513846e55afd5 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":6},{"id":"26646b45061e9ad787497d684d63166df36362b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-31 19:28:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7d5eec2cdb023989c6db7454dce2b6edfa3d3f88","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-31 19:28:30.000000000","message":"Patch Set 7:\n\nrebased to fix merge conflict","accounts_in_message":[],"_revision_number":7},{"id":"9c6017322cd6d61a4e1095f3eddc162e5499e997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 21:32:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b977aa36e0b54ba2b276d30d5b0d86fc : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/5e513799929b47d2b0ceb2cc4a2f8e04 : SUCCESS in 47m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb282a31ba4d47708f98640445238ad8 : FAILURE in 32m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ab864cc35cf43de881ef5b9b7ea733b : SUCCESS in 7m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e93b405208ea4f478da067bf02d2351b : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9e51435d54647d99da81c92ceb40713 : SUCCESS in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91c9760e8126490b8252f44eed97e0e7 : SUCCESS in 21m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ea3ffa44bab4624a4ced6282c3b769b : SUCCESS in 12m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b6a61eeafe3348efb01e06dcd016ea7b : SUCCESS in 21m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f1b29a90e7f45ea9b64918d2e75c370 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/229c5ad5ee3744f68ce21d4e1e83572d : SUCCESS in 15m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d8e5a3aede74b3f8cafcd7ced5ed701 : SUCCESS in 16m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1ed86a21bbc49449162bc6b5f714c29 : SUCCESS in 28m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4b045511d75d4736b8ddda8782f97b51 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c7a0b19d599d4ca7bf05d584a462d36d : SUCCESS in 25m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb89d581b00d4fdf93ee82e157ade9a5 : SUCCESS in 21m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/14cc08a8427741fda088b6ba94c3abf1 : TIMED_OUT in 2h 01m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8955ee35bac24174a99df2cbeebc2589 : FAILURE in 39m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/479336247f7b4e3daae22c19d24bd68b : SUCCESS in 42m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a333e2598eaf4a5ead0cb47bbd33f500 : SUCCESS in 38m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d73596b8c44ccfbfe4156a4a3fffb7 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03cae2ba18cf4e89b148b32a1c4b4ab5 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c5ba06891ae4aadad7199ab1245d1bb : FAILURE in 7m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4836887b32a942d0994434ec53397d5b : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":7},{"id":"13a322633a8a9e3e0b757308a791e39db3774161","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-01 09:40:28.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8c257d34cbb3f666d9872722a68e56a852287378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-01 11:45:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feea272b076a4014be86d2bd3d85d2f0 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/2200e5dae59d43dba363ce709b5b4308 : SUCCESS in 50m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53a6944718944f6e9983883c0debe359 : SUCCESS in 39m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcc7633aa9e54392bb1b304854ef81dd : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd0e15179b594e2991116240ed5d5e5e : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8576669e3fb04bc997881cb2a8463d73 : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55b9796f72774337b9ec18575b7194f8 : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee3b856237534b699436f50d74063556 : SUCCESS in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ce9c058b9e5d4edf95da1f74e58deefc : SUCCESS in 11m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db0d9ea3436740879717593e80b6c1dd : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c0aeadd48c004b808780f6a26cb4900f : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9876bdeb7434e58aba6cedf69a4b07e : SUCCESS in 20m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6ff640700e7c41d0ae1737e36e2465ae : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/63324138d3fe47f3b794cc4d025ba6bc : SUCCESS in 16m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/118daf0356b24f7396d08b61241acf7e : SUCCESS in 21m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b478d1c422f74ee6b8fca314727b67b8 : SUCCESS in 22m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb385be36cfc435db086fb601e384b86 : TIMED_OUT in 2h 01m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fb35a20beb214255a7a4355933dd7853 : SUCCESS in 43m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f9d38a5e8654d1199e618c2f8ce4208 : SUCCESS in 38m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed401e12b4a248819366d54fb2bd2f00 : SUCCESS in 36m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2d3ff41f824f7a992c83750920b2fa : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58947b0d6fa045628c7e9e774ad2bcc9 : SUCCESS in 7m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bffb9eebe7884156856fd1501f9358e8 : FAILURE in 8m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d302c65460847c7b9a6740a606b9f76 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":7},{"id":"83363157dcc4a2903c1436e650ebc606210697f6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-02 17:32:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"86b639e374d6bc8ca25ef810452cb9bbf804d662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 19:38:12.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df08f7fd2e0c4281a875c6dbdff120a4 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/480ad63827d242f7a61f060feca12f80 : SUCCESS in 1h 00m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cee11b07ca264f5fb4a24a9814d72781 : SUCCESS in 47m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0806bdbd9c2c4797b846e17bf2ecaaa8 : SUCCESS in 7m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d97029608f6b4bec8164c44fbdd0bc37 : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/147803fce79f44ab9af6373f659c3b43 : SUCCESS in 11m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d9958ab7567496cba88f175473fadc5 : SUCCESS in 12m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9c142626c2d14ab48c2745b43ffc86fb : SUCCESS in 12m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d72f871e38a44e5480ffb4345c8057e8 : SUCCESS in 13m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c042bc7704f74740bc7c20afb9e101b2 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c46029a9cefe45de816ba9ca0a136477 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88ad426f3e6746249c1f9a062f95bb81 : SUCCESS in 16m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e9dd48674bbc4e54a485f49d5800f8f0 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fcbc143efc504bee802126bc8576d434 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7109cf4c2ef14b1b82e17ddbf8bfb08b : SUCCESS in 19m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38f7b5a11c5c49f09f871ad26e869831 : SUCCESS in 19m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c1e622ab95e41e1a9ebe724d2c20b33 : TIMED_OUT in 2h 00m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9a307f8ff10a496193b95c6412a73041 : FAILURE in 40m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7e2b13a2e5b41b5873fd444360362d7 : SUCCESS in 30m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de94549b75a443a29b02b0e763871725 : SUCCESS in 29m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/277623142eb64e4485ff754e566d2da6 : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2d7f96cd9c47ca8e9fd864406b1c37 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41d3d4978a2d4b6cbaa9f35350b25ecc : FAILURE in 6m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e94d9cbb63946d7a058895544b95fd0 : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":7},{"id":"899a869d7625c4c86584bbfa4038582c0344a5d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-05 11:14:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5ce83b2a2ddd9f510b329794ff123ce470177661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:19:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc42dfe3ed2475a99fb66ec69b2a582 : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/202357a69bfb4fab9efa6f2e4027bc12 : SUCCESS in 58m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29496bb22282473994d6b93f1819f3de : SUCCESS in 45m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d972e8e6d354100a807ca5c705423b1 : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/073b7339a1564c0eb2df8030893c3723 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bf51e508c404a679f23ae69855ad8ed : SUCCESS in 12m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ffa613f3f7d42abb1f7dbe14cc22ab6 : SUCCESS in 13m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74727a5e4f0d4b0a9e991b40772c79c1 : SUCCESS in 13m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3e70b0ad499249be81a04e9d95eff553 : SUCCESS in 14m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae56cca16ac442e6b3e4be9710a06105 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6924601171f4346a3feec31682631e8 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9172bc7480144193aff3890af8737c0e : SUCCESS in 17m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3e3c9129d3844eec9aeb5f25306aa202 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/46a0d723cf79475b9066f91ce4215d82 : SUCCESS in 22m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/40411d5061e9423aa324aeda336b0ef0 : SUCCESS in 19m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18fc567d6ee7473d80685bb351476c8e : SUCCESS in 25m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d9ecfa6fdd1461885a14bdd7488dcac : SUCCESS in 44m 53s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ae8dfde153384771bfd06cb564c9152a : SUCCESS in 44m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/988fb26833344d45a40232d9f14ac2f7 : SUCCESS in 37m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/643d6e5c012c4d4b83ce80027d9c67fa : SUCCESS in 47m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41013d8f4e1e4e8888c1882791b342a8 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab692ba8e1848b7bad001458f821c8b : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d028afb7a969421c9135a292dacb604e : FAILURE in 8m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27f7060459914e3a83a25d6b42abe9a7 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":8},{"id":"ba20634680c146edee437142e85fe02adbbf30a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-06 16:29:13.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"f0d56302b8b4176520a0ffef70fa2da6fca996d2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-06 16:32:20.000000000","message":"Patch Set 9:\n\nrebased, no other changes","accounts_in_message":[],"_revision_number":9},{"id":"0d28cbdb10c3d94aa635b3b41240bf4dc3aa88f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 17:37:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca579e8d6464f9bbfeae10f4924ade8 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/9187c449c355434485f730116ae26c32 : SUCCESS in 52m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5e79a511c8a4e3bb3869fdacd4a5a8b : SUCCESS in 38m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b19b6a2c09ec4313bf8857663ef47b78 : SUCCESS in 12m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20f4e779556548b78f71d0085be9c6bc : SUCCESS in 12m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1dcb60522fd4ef6baa830dfc765b590 : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33284968c12e4e6a8a19df945a8ac137 : SUCCESS in 15m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7032cb5e4753487eabd15a88fdbdcabb : SUCCESS in 38m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e4f0e97a61d24117998a0863464ae2d7 : SUCCESS in 22m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c429513012694afda6ff056c67db1d1c : SUCCESS in 20m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1080db86312d411bb7cef0472e842903 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b753cc0747604e4fb00d79a442eb9d60 : SUCCESS in 17m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d4b2c4f4960340398f2c5acbd02adbae : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/223c42bfd33b4313925d0f820840e9fc : SUCCESS in 18m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9d0e60d5858347029bf99ef155a1a80c : SUCCESS in 18m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c68542ebad5c4f958879d92602214d1d : FAILURE in 20m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35a6fdd931c14288ac0e17a57aa54a73 : SUCCESS in 53m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/21306010cedf4e528e6bd30e044c74f4 : SUCCESS in 46m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7230e38ef97642ecb0e23dda1eec59c8 : SUCCESS in 47m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60dbb92724c045c29ed84ce1c79ed777 : SUCCESS in 44m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80a0ee89f43a4de0b994b082c46a8f53 : SUCCESS in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b091bbfcaf3458185b3c01dbb25fda8 : SUCCESS in 7m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f8ade5978bc4f898abab3b3621549be : FAILURE in 9m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2cce200d18444b9bbe2e91a585970e5 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":9},{"id":"4c8b4596f446d0d1e30e58c0512064a8a70dad6d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-07 03:30:58.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)\n\nLooks really good. Great work.\n\nJust running the probe tests a few times before I upgrade my score :)","accounts_in_message":[],"_revision_number":9},{"id":"cf1db11b8a4f5e0daf871821d5d3d3c62f7f45ea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 09:15:48.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3c739c5da54601ef1eae4a219403bcc16162f0c7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-08 17:12:22.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\neveryone seemed onboard in the meeting this week!  we\u0027ve been shipping this and LOVE it!","accounts_in_message":[],"_revision_number":9},{"id":"61d17964c0614d10fdb8c96b617a70847f11532b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:12:43.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1a6440ee5fc7133bfaf26352a9326815b4851701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:50:44.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d38f79f9ff54b5384dbddae7c2d9b6c : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/88b16dd85918464587835d334f2f38ac : SUCCESS in 1h 01m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a58009531ecb42b0afe9b4a2ce30e398 : SUCCESS in 46m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e7f7638491684c33b65fbb408280030f : SUCCESS in 9m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d40b2d40d47d4217b4cb5193b0d8716b : SUCCESS in 10m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79e3d742ab15481f857474f5fb398add : SUCCESS in 18m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3708e7e5e9fc4d40938574430dc323cf : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ec9bbcbd02e44e9855fd47c010a21b7 : SUCCESS in 15m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d422df88a14b490fb340512a1f1cc87f : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3f8b631b9ca482494526695a9b7f451 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04e1edc84ffe43e3b0cb326a177ba205 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fdebb90af5904b0b8fa6464c7f44d30f : SUCCESS in 16m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/90838db4fe454d068ce8b9ccd66fccd0 : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2f992f63fb3d4c5aaf4aa7bcba35f2e8 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b70c4defe1c34b8b9b3e05abfd74f2eb : SUCCESS in 18m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/264230924e6b4b9e92c9e78295fbe66c : SUCCESS in 46m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/539607aae6e54084b607ee73718a01ef : SUCCESS in 45m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a33dd3709204475b9b6802b9cfc721a : SUCCESS in 38m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21ac7b093bb444149b776e32bea8ffc3 : SUCCESS in 41m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fa626d4cac2484bbab5c82429482b75 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5697fcb4b0c94ce187588fcb98e2c7e8 : SUCCESS in 6m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55503b3096704298ad35a7770ba6ea9c : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":9},{"id":"d1cb707c32112eabb184e12445690018ece3bac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:50:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"30e091bb3caa08ba5985be7156e6bc511587c67f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:52:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d56eabce4241738dca426b1557cff5 : SUCCESS in 1m 29s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5c8f6352fcfc4ffc8aa5f06e0c7a22c2 : SUCCESS in 54s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d558dc8c4d774652bfa70433337d0ddd : SUCCESS in 1m 14s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"077ba77ea6f735c98709b60e1fc9aa9f0f90676a","revisions":{"6d8bf951082e5472349e1509de04cc59d8273b62":{"kind":"REWORK","_number":1,"created":"2020-11-05 19:44:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/1"}}},"commit":{"parents":[{"commit":"14bd6b916da65189ba99c391cef79731d05e4151","subject":"Merge \"tests: Replace FakeLogger with debug_logger in test_proxy_logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14bd6b916da65189ba99c391cef79731d05e4151"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-05 19:44:09.000000000","tz":0},"subject":"WIP - cache listing shard ranges","message":"WIP - cache listing shard ranges\n\nWith this change, when the proxy knows from cached container metadata\nthat the container is sharding or sharded then it will cache shard\nranges fetched from the backend during a container GET. On subsequent\ncontainer GETs the proxy will use the cached shard ranges to gather\nobject listings from shard containers.\n\nThe proxy will only cache listing shard ranges, and will only use\ncached ranges, when an X-Backend-Accept-Cached-Results header with a\ntrue value is present in the container GET request. Currently,\ncontainer GET responses that use cached shard ranges may not have a\ncomplete set of headers as described by the API; a middleware setting\nX-Backend-Accept-Cached-Results should therefore not expect a complete\nset of response headers. Get responses that use cached shard ranges\nwill have an X-Backend-Cached-Results headers with a true value.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d8bf951082e5472349e1509de04cc59d8273b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d8bf951082e5472349e1509de04cc59d8273b62"}]},"branch":"refs/heads/master"},"e23d336505428866a3496fc9293faff96f196550":{"kind":"REWORK","_number":2,"created":"2020-11-11 21:28:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/2"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-11 21:27:48.000000000","tz":0},"subject":"WIP - Use cached shard ranges for container GETs","message":"WIP - Use cached shard ranges for container GETs\n\nThis patch makes three significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n\nWith this change, when the proxy knows from cached container metadata\nthat the container is sharding or sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nTODO:\n- remove need for X-Accept-Cached-Results header\n  now that response headers are API compliant\n- make code path to use cached shard ranges conditional\n  on shard cache expiry time \u003e0\n- unit test HEAD from cache\n- unit test GET when server does not respect\n  X-Backend-Override-Shard-Name-Filter (e.g. during upgrade)\n- functional test repeated HEADs to check if cached\n- unit test headers_from_container_info\n- unit test get_cache_key\n- more probe test changes\n- tidy up logging\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e23d336505428866a3496fc9293faff96f196550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e23d336505428866a3496fc9293faff96f196550"}]},"branch":"refs/heads/master"},"aa14ea3be90943361a8da5a51edf9f21b38da155":{"kind":"REWORK","_number":3,"created":"2020-11-12 15:38:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/3"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-12 15:33:15.000000000","tz":0},"subject":"WIP - Use cached shard ranges for container GETs","message":"WIP - Use cached shard ranges for container GETs\n\nThis patch makes three significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n\nWith this change, when the proxy knows from cached container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nTODO:\n- move the HEAD-from-cache feature to a follow-on patch and unit test\n  HEAD from cache; functional test repeated HEADs to check if cached\n- remove need for X-Accept-Cached-Results header now that response\n  headers are API compliant\n- make code path to use cached shard ranges conditional on shard cache\n  expiry time \u003e0\n- unit test GET when server does not respect\n  X-Backend-Override-Shard-Name-Filter (e.g. during upgrade)\n- unit test headers_from_container_info\n- unit test get_cache_key\n- more probe test changes\n- unit test container/server changes\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa14ea3be90943361a8da5a51edf9f21b38da155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa14ea3be90943361a8da5a51edf9f21b38da155"}]},"branch":"refs/heads/master"},"6850a4383ef1661082f976832cf5e5c3482c276d":{"kind":"REWORK","_number":4,"created":"2020-11-16 18:00:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/4"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-16 17:59:01.000000000","tz":0},"subject":"WIP - Use cached shard ranges for container GETs","message":"WIP - Use cached shard ranges for container GETs\n\nThis patch makes three significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nTODO:\n- more probe test changes\n- assert API compliant headers in probe tests (since these are only\n  black box tests against a sharded container)\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6850a4383ef1661082f976832cf5e5c3482c276d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6850a4383ef1661082f976832cf5e5c3482c276d"}]},"branch":"refs/heads/master"},"4609bce98f6ee1d1aab06517b6f933657624a411":{"kind":"REWORK","_number":5,"created":"2020-11-23 17:36:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/5"}}},"commit":{"parents":[{"commit":"e22cad666a7bbb16a40a85a162a829da0f23c187","subject":"Merge \"xlo: 5xx while validating first segment is a server error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e22cad666a7bbb16a40a85a162a829da0f23c187"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-23 17:35:54.000000000","tz":0},"subject":"Use cached shard ranges for container GETs","message":"Use cached shard ranges for container GETs\n\nThis patch makes four significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n  - the proxy server now enforces more container GET request validity\n    checks that were previously only enforced by the backend server,\n    e.g. checks for valid request parameter values\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nThe X-Newest header may be used to disable looking up shard ranges\nin cache.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4609bce98f6ee1d1aab06517b6f933657624a411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4609bce98f6ee1d1aab06517b6f933657624a411"}]},"branch":"refs/heads/master"},"c0a08fc3becfa12723275a77844ef80a1feb45d5":{"kind":"REWORK","_number":6,"created":"2020-12-15 20:55:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/6"}}},"commit":{"parents":[{"commit":"e22cad666a7bbb16a40a85a162a829da0f23c187","subject":"Merge \"xlo: 5xx while validating first segment is a server error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e22cad666a7bbb16a40a85a162a829da0f23c187"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-12-15 20:20:13.000000000","tz":0},"subject":"Use cached shard ranges for container GETs","message":"Use cached shard ranges for container GETs\n\nThis patch makes four significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n  - the proxy server now enforces more container GET request validity\n    checks that were previously only enforced by the backend server,\n    e.g. checks for valid request parameter values\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nThe X-Newest header may be used to disable looking up shard ranges\nin cache.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0a08fc3becfa12723275a77844ef80a1feb45d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0a08fc3becfa12723275a77844ef80a1feb45d5"}]},"branch":"refs/heads/master"},"e83a0d9e949e392ff415becff5aa7f820b2ec0a1":{"kind":"REWORK","_number":7,"created":"2020-12-31 19:28:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/7"}}},"commit":{"parents":[{"commit":"5e3302649584b2aec523e5966671e43ea959ed27","subject":"Use CloseableChain when creating iterator of SLO response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e3302649584b2aec523e5966671e43ea959ed27"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-12-31 19:27:54.000000000","tz":0},"subject":"Use cached shard ranges for container GETs","message":"Use cached shard ranges for container GETs\n\nThis patch makes four significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n  - the proxy server now enforces more container GET request validity\n    checks that were previously only enforced by the backend server,\n    e.g. checks for valid request parameter values\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nThe X-Newest header may be used to disable looking up shard ranges\nin cache.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e83a0d9e949e392ff415becff5aa7f820b2ec0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e83a0d9e949e392ff415becff5aa7f820b2ec0a1"}]},"branch":"refs/heads/master"},"40cdc7e4e339248bce899ee1f01b92fb7c6d662a":{"kind":"REWORK","_number":8,"created":"2021-01-05 11:14:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/8"}}},"commit":{"parents":[{"commit":"5a5980e9c4f640e0e99b970ab14431c48be03a60","subject":"Avoid loops when gathering container listings from shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a5980e9c4f640e0e99b970ab14431c48be03a60"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-05 11:14:05.000000000","tz":0},"subject":"Use cached shard ranges for container GETs","message":"Use cached shard ranges for container GETs\n\nThis patch makes four significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n  - the proxy server now enforces more container GET request validity\n    checks that were previously only enforced by the backend server,\n    e.g. checks for valid request parameter values\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nThe X-Newest header may be used to disable looking up shard ranges\nin cache.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40cdc7e4e339248bce899ee1f01b92fb7c6d662a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40cdc7e4e339248bce899ee1f01b92fb7c6d662a"}]},"branch":"refs/heads/master"},"077ba77ea6f735c98709b60e1fc9aa9f0f90676a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-06 16:29:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/761659/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/761659/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/761659/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/761659/9"}}},"commit":{"parents":[{"commit":"453983764727ba528222a413691a0f0fc5a8f8cf","subject":"Avoid loops when gathering container listings from shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/453983764727ba528222a413691a0f0fc5a8f8cf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-03 11:08:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-06 16:28:49.000000000","tz":0},"subject":"Use cached shard ranges for container GETs","message":"Use cached shard ranges for container GETs\n\nThis patch makes four significant changes to the handling of GET\nrequests for sharding or sharded containers:\n  - container server GET requests may now result in the entire list of\n    shard ranges being returned for the \u0027listing\u0027 state regardless of\n    any request parameter constraints.\n  - the proxy server may cache that list of shard ranges in memcache\n    and the requests environ infocache dict, and subsequently use the\n    cached shard ranges when handling GET requests for the same\n    container.\n  - the proxy now caches more container metadata so that it can\n    synthesize a complete set of container GET response headers from\n    cache.\n  - the proxy server now enforces more container GET request validity\n    checks that were previously only enforced by the backend server,\n    e.g. checks for valid request parameter values\n\nWith this change, when the proxy learns from container metadata\nthat the container is sharded then it will cache shard\nranges fetched from the backend during a container GET in memcache.\nOn subsequent container GETs the proxy will use the cached shard\nranges to gather object listings from shard containers, avoiding\nfurther GET requests to the root container until the cached shard\nranges expire from cache.\n\nCached shard ranges are most useful if they cover the entire object\nname space in the container. The proxy therefore uses a new\nX-Backend-Override-Shard-Name-Filter header to instruct the container\nserver to ignore any request parameters that would constrain the\nreturned shard range listing i.e. \u0027marker\u0027, \u0027end_marker\u0027, \u0027includes\u0027\nand \u0027reverse\u0027 parameters.  Having obtained the entire shard range\nlisting (either from the server or from cache) the proxy now applies\nthose request parameter constraints itself when constructing the\nclient response.\n\nWhen using cached shard ranges the proxy will synthesize response\nheaders from the container metadata that is also in cache. To enable\nthe full set of container GET response headers to be synthezised in\nthis way, the set of metadata that the proxy caches when handling a\nbackend container GET response is expanded to include various\ntimestamps.\n\nThe X-Newest header may be used to disable looking up shard ranges\nin cache.\n\nChange-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/077ba77ea6f735c98709b60e1fc9aa9f0f90676a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/077ba77ea6f735c98709b60e1fc9aa9f0f90676a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
