)]}'
{"id":"openstack%2Fswift~762317","triplet_id":"openstack%2Fswift~master~I8f87e6b5f5bcb9dc003bf3b3a98ea034cd93a032","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8f87e6b5f5bcb9dc003bf3b3a98ea034cd93a032","subject":"Remove six","status":"ABANDONED","created":"2020-11-11 09:12:19.000000000","updated":"2025-02-05 17:43:02.000000000","total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e53e2d049914ba15922fabac674ec48a6b5ecdb2","_number":762317,"virtual_id_number":762317,"owner":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-11 16:30:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 16:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aae510ca10190d51d4a54bc1e1180ba7049ace42","author":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"date":"2020-11-11 09:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f6c2c62d16b265f8564f0d4be2253f2809f60b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 16:30:29.000000000","message":"Patch Set 1: Verified-1\n\n(29 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cff398ef5b4d25a6a0e600d6908e7d : FAILURE in 6m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/60706fe05b994e55bc7466c1616ac504 : FAILURE in 1h 04m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47a1e757777b4b0cb1fcb87bc9071a25 : FAILURE in 34m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7309b8516853486e9601e41ca3f2bf19 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/44e701b7253a4e9a86cf2f0d6ec01b42 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a51c7b75b5684b2880a30d7dca938775 : FAILURE in 4m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a67be42c6b4beba75631cf1824306a : FAILURE in 5m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/463f74cc233048f89f96878abdcc59c6 : FAILURE in 5m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef6fabbd1004b2fb61c8f38e68b2115 : FAILURE in 4m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/848c9b7de57e4fb993fbc5d4dc9466b5 : FAILURE in 4m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/687f52d54c0b465d90fd4c1391d4e87f : FAILURE in 4m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c9ce20cb6d14079b2a681d6ab5a81b2 : FAILURE in 4m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc4b85312bfe408ba1c3dea040fc1a27 : FAILURE in 6m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/70ce9cbbb9fc4f2ba1f22233fb64de7e : FAILURE in 4m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bb4627319b444de4a2b933da22dc8c10 : FAILURE in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f207c91bf958447f9b72ed4ce9aaf9a1 : RETRY_LIMIT in 4m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94d2fd2e8f3a4377ad64345afc7cc088 : RETRY_LIMIT in 9m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f90241273efd43959e67aaa3deed0e9c : RETRY_LIMIT in 15m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0947cf2604d4a29b8fadc4671010dbc : RETRY_LIMIT in 19m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59e9377aeb5462388fbe7751a72563f : FAILURE in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6403c348b33d4998ad08f80f0be1fabd : FAILURE in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3561ab756694e1cb8773ce5de3d6fa8 : TIMED_OUT in 1h 00m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6553ed52be1345f497b98f8ee6efe579 : FAILURE in 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e53e2d049914ba15922fabac674ec48a6b5ecdb2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 17:43:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"093fcf08b094aac71ba9963ab89c400f6ee4e3f6","revisions":{"093fcf08b094aac71ba9963ab89c400f6ee4e3f6":{"kind":"REWORK","_number":1,"created":"2020-11-11 09:12:19.000000000","uploader":{"_account_id":32163,"name":"suhaiming","email":"suhaiming@inspur.com","username":"suhaiming"},"ref":"refs/changes/17/762317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/762317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/762317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/762317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/762317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/762317/1"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-11-11 09:10:34.000000000","tz":0},"committer":{"name":"suhaiming","email":"suhaiming@inspur.com","date":"2020-11-11 09:10:34.000000000","tz":0},"subject":"Remove six","message":"Remove six\n\nChange-Id: I8f87e6b5f5bcb9dc003bf3b3a98ea034cd93a032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/093fcf08b094aac71ba9963ab89c400f6ee4e3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/093fcf08b094aac71ba9963ab89c400f6ee4e3f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
