)]}'
{"id":"openstack%2Fswift~762718","triplet_id":"openstack%2Fswift~master~I99b351eee4fb86b6fcd54f7936b11f4918e1b999","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I99b351eee4fb86b6fcd54f7936b11f4918e1b999","subject":"slo: 5xx while validating first segment is a server error","status":"ABANDONED","created":"2020-11-13 22:43:16.000000000","updated":"2020-11-17 16:39:32.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e8763985e925d1873a710925f1baa2b4f910ab4","_number":762718,"virtual_id_number":762718,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-14 07:48:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-11-14 17:23:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-14 07:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-14 17:23:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"8eb5fe1a4a81ad13cdd17b69c99d9d8e8216deab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-13 22:43:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1135d2ee919a613305670eb570933c77ee15b559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 23:42:03.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede90019a09948ab93e86b74971ccfb1 : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/e352d61e81c04087b4146906ec0d5143 : SUCCESS in 56m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/99ed06cb1de34837bd85ab0c6398c908 : SUCCESS in 42m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e39d32c57ff4dadb00360ef9fd74d2a : SUCCESS in 9m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b211b3f361b74b098aac2954c56d5751 : SUCCESS in 9m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7322f66c46745f6859c1156bb1493e1 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e62fb8ced6044307865986e5dc39e541 : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d44d357c98a4005b42d542bda7c8333 : SUCCESS in 13m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30d1d990f790437594d6de8f0824b444 : SUCCESS in 13m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/437e38a58bd44acf9d1b89a7434f5e4b : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d5e32350e614b4a943c1b7ed2abc526 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1b0e1b25f96640c0a0441aab5d507b2f : SUCCESS in 17m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/86e338253fe34abdb84f5232ca78d54b : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5c3a85425dd748c29c69bd82b235d072 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3637cdd0ff7e4d71893d9f90bba57d6b : SUCCESS in 18m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d23ee30bc1e04f9c8df0b5a09738ff44 : SUCCESS in 25m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0eef6bf7a7e14d36a441771a7fe92366 : SUCCESS in 43m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/529519eb690d41ae8013a9c259277b02 : SUCCESS in 43m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd3414db124546dd8f46df4e435a4e4d : SUCCESS in 40m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f0d0b58bc66419ab5724881d98c56a4 : SUCCESS in 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a827895d5bef40b298ced037a2e7852c : FAILURE in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e83bb6cb83604c378aa62e32ac1f9eec : SUCCESS in 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/519ee2bbc36e42f28fc710a538a88aaf : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"00bea46a83a477820efaa67ea154c30c40a0fe29","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-14 06:08:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce89c5520efd6d1d753c0bdf089ca30fb8155065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 07:48:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba45943cffe4dc895129ee49baf4dc9 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/259252e136a04f228eebb36c836dd9cc : SUCCESS in 53m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ddf52772788d4c738e237ab4579c29dc : SUCCESS in 49m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/369c78f03ba8456782d6d10a5b4cc4d4 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59015b5914404528abc8051bb27f2822 : SUCCESS in 12m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/217d448b79554799b8b9bfe47891b082 : SUCCESS in 13m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9794105265f416aaa6b61d507be3d17 : SUCCESS in 15m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/90c359423a9f4d17a2979717ccab7e26 : SUCCESS in 14m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f0886e5db2b143eaa5056afc530f6ab8 : SUCCESS in 12m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7557a056805548b697776c3372cc9d93 : SUCCESS in 33m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/39a66d6924ac4d64a9af74ecc40bc6f3 : SUCCESS in 27m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb143b27cfa14463b1578e1a81e86063 : SUCCESS in 29m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e8a868bd3e63417389e206d453503a45 : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d34abb3d9964f538f8e81de3e75bdef : SUCCESS in 17m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6bbdf2f811fb43999475591642bd460b : SUCCESS in 19m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b8f799a8f43b44a1b33708790febdc5a : SUCCESS in 22m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df83d3b6bbb349709742ae413f89f621 : SUCCESS in 48m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae666a16c4a249c3b127ee7fde170d7c : SUCCESS in 32m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1a4adea8c2d4b088607558e53da5102 : SUCCESS in 39m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc75f11ec4244dd8dc7c0f8ad1843a8 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/638dbb63d0d649b8a90c468ae69f7371 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c47431c02504970bd73579cb6ec4798 : SUCCESS in 23m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/081c32fd808b4974922ca7dd46abb4f8 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":2},{"id":"614ce147569f4bc9432c2a86f6af347790775e1f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-14 17:23:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(7 comments)\n\nI kinda think a little something more surgical would be about just as good:\n\nhttps://review.opendev.org/762738\n\n... but dunno - i\u0027m sure there\u0027s downsides and upsides to both","accounts_in_message":[],"_revision_number":2},{"id":"2e8763985e925d1873a710925f1baa2b4f910ab4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 16:39:32.000000000","message":"Abandoned\n\nTaking the approach in https://review.opendev.org/#/c/762738/ instead.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e92ba19590ab6981417cdd9d8b5ddb5ae009742","revisions":{"a63f3e6444a0f7fac613e2b0cdef9ea2b699a123":{"kind":"REWORK","_number":1,"created":"2020-11-13 22:43:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/762718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/762718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/762718/1"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-13 22:43:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-13 22:43:00.000000000","tz":-480},"subject":"slo: 5xx while validating first segment is a server error","message":"slo: 5xx while validating first segment is a server error\n\nChange-Id: I99b351eee4fb86b6fcd54f7936b11f4918e1b999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63f3e6444a0f7fac613e2b0cdef9ea2b699a123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63f3e6444a0f7fac613e2b0cdef9ea2b699a123"}]},"branch":"refs/heads/master"},"6e92ba19590ab6981417cdd9d8b5ddb5ae009742":{"kind":"REWORK","_number":2,"created":"2020-11-14 06:08:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/762718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/762718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/762718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/762718/2"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-13 22:43:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-14 06:08:16.000000000","tz":-480},"subject":"slo: 5xx while validating first segment is a server error","message":"slo: 5xx while validating first segment is a server error\n\nChange-Id: I99b351eee4fb86b6fcd54f7936b11f4918e1b999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e92ba19590ab6981417cdd9d8b5ddb5ae009742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e92ba19590ab6981417cdd9d8b5ddb5ae009742"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
