)]}'
{"id":"openstack%2Fswift~762738","triplet_id":"openstack%2Fswift~master~Ice5358ff85ee2d5fe60785b73b67dea493044a2c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ice5358ff85ee2d5fe60785b73b67dea493044a2c","subject":"xlo: 5xx while validating first segment is a server error","status":"MERGED","created":"2020-11-14 17:22:20.000000000","updated":"2020-11-20 01:15:01.000000000","submitted":"2020-11-20 01:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762738-1605834806363-6db76d6a","meta_rev_id":"44ad5836f235541e8a7ab70342dfa187f590f1ab","_number":762738,"virtual_id_number":762738,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 01:13:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-17 16:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-17 16:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 16:38:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-11-20 01:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"957e96b2d87d3513b1ca527204d4793546de35c5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-14 17:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d79eb702525d8589c89453ade34ee779eb6dac19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 18:35:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf84f2d1c61b4886af966470f85efa58 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/bb1f97d848604c4ca6727645688912c4 : SUCCESS in 54m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61966a147a34475fabf154930cff3627 : SUCCESS in 41m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d40a159d17949bf98ee8254ed2ede00 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/81864c4e15394096b8b2ceb924ac9824 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24331be1cb474ea88b9d39d8f61957b1 : SUCCESS in 12m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f786083d2843a6a38e935b8adfe7d4 : FAILURE in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f88b8306f3404b15b33fe933578abb82 : FAILURE in 11m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca78156de70415b818302339715cc7a : FAILURE in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/656a4b39707c44f58eb6ae7ca79cecd0 : FAILURE in 6m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03a99874c688417e846ad79a4ea70af0 : SUCCESS in 30m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/26d8d72b64a24f5bade5464eb7d05873 : SUCCESS in 17m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f0f30518a62d4ae888ff29d9b438c3d6 : SUCCESS in 28m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/056ed18edad14eaab9e522f4f76086e9 : SUCCESS in 30m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4e17851f19de4807b8c9b66836bb3f3c : SUCCESS in 23m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46a2cc626eae4b70b787627633964a27 : SUCCESS in 43m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b478ab7d78d74f77a9aac6f00c34d341 : FAILURE in 41m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9bca9cc16294428c9b190aaf3c4b1de5 : SUCCESS in 38m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d907cf2d69f45deaa0ad4555868db2d : SUCCESS in 36m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36d36fb1f4c74320b87e3c1f44939b0e : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd63e6816c2b4537b96880af50ce74fe : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0527888edf56440c9000063dd36eb5ec : SUCCESS in 21m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bddcb74af16d4f80b96c2a63adf069f0 : SUCCESS in 1h 07m 17s","accounts_in_message":[],"_revision_number":1},{"id":"59215a4015d5a1a6fa331dbc5869e614bc7810c6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-16 14:02:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d611ee59027a77ab87a514f38c1eb22523b841df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 16:31:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc30ebcd374445bd8f0e5854ebe4b919 : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/71f65acdd51843adace83f149ba8df23 : SUCCESS in 1h 06m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fe3db86b00248f2ab0788e67a5566ab : SUCCESS in 43m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77bc14c72f86464cad833ed70f591b88 : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c8da5fb5555485fa6b3459fe66c243a : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e562f99afb6468089b5fc1863a13962 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e1e17169ab74108846edafced9354a4 : SUCCESS in 13m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef0a77539a124ebdb6a0607e73f88dc0 : SUCCESS in 13m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78be2ac6a17a49c991a9b297216547d1 : SUCCESS in 19m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2d034b515be4cf5b122049fc778f90a : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45bb082544b84b08ab73e93bb85b48f7 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4d6facd58184d278383e572ea75bd8a : SUCCESS in 22m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2adfbc0e8f1040a29341e4cc173b41a3 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7d91ff17c7404acdb4a8a391f8ed25fc : SUCCESS in 24m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ad1398e74b3043e6b981d7495552c200 : SUCCESS in 41m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea1ff611078240228c660f6fb8cf26fb : SUCCESS in 30m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/efd414b4830d4467b9496023fa9b7124 : SUCCESS in 45m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/620d6af111144339ad914714aa5efec4 : SUCCESS in 38m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1693172a0f0450b9ce10eca98fbf93a : SUCCESS in 44m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc315c0e6ff2440fa4006ad5a0b2d254 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe982123840459ea58047889f6946d0 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/385cb5f513f04bfbac651f701bd3dd77 : SUCCESS in 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99d95351d9b64958a3497a7a2d7f2a23 : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f5b6115fbef62822ea11c615e71c5d16e8808442","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-16 17:55:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7aef0fa5ba4f7eb85616f5d1cdb4f2fe611d50f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-16 18:00:36.000000000","message":"Patch Set 3: Code-Review+1\n\nYeah, this could work, too -- now I\u0027m starting to wonder about what happens when the container\u0027s down when DLO\u0027s doing listings...\n\nFWIW, the only other SegmetedIterable user I\u0027m aware of (ProxyFS\u0027s pfs_middleware) looks like it\u0027d be fine catching the HTTPException.","accounts_in_message":[],"_revision_number":3},{"id":"5283d5f98c3c771032bcbbd805db98159655e27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:40:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd02a6cbf5e841b6882203ffb05853f6 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/019700fa3e5c4eb38c1b7bcd4c174a30 : SUCCESS in 1h 03m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/488b69bb5567433e893e6dcad8dc2e6f : SUCCESS in 47m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1cb617b8ebad470db4f4032a39fa06cc : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bdba9d524658426daf3a415f9fbfaddd : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c1bb7bc06f34f3e90db4d02b13a9f8c : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b10b48207a64bc2875cb1c4bb70d9de : SUCCESS in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/53caeafc572d47c2bfe50b959734e614 : SUCCESS in 14m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ca04bfe140e459d97b1d2c28d1e6b31 : SUCCESS in 12m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6c74884adfc43e9b9c2586e12d57936 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/61bdca04e23a418fb3e7b9f809a26245 : SUCCESS in 21m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d260770a75874f57b0d42c72edc28723 : SUCCESS in 17m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d049362a6565459aaedf2187d69e1df1 : SUCCESS in 22m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9ebe142187dd42ffafea6d561ca191e4 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/58ebb21409134c0a8778cd8c364a564b : SUCCESS in 19m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb58b6bbeb334042881bd796667013a0 : TIMED_OUT in 1h 01m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/de2d245b672f4f6daada0978f41e66db : SUCCESS in 45m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61846d829aeb494ca9ea1c8f095636aa : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/247c3ea6dd394d49a2aad1c2cbd36241 : SUCCESS in 46m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eae38f4967742a8b69dedfac5d28414 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a841fe9f63cb40d4b4d9870833975c43 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ac8b18d1dbc476cb9047c43f8b065fe : SUCCESS in 22m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7db89b3c5cfc46ee9b5dd5952756c1a3 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":3},{"id":"07d892bdb32b22b9cf68cc77effb20fad84ccaa4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-16 22:31:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(5 comments)\n\nI think this is a pretty solid approach - the dlo robustification fallout is pretty nice :+1:","accounts_in_message":[],"_revision_number":3},{"id":"b5561482e94460538ec16452dc625e143e0703c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 16:30:23.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ea1bc33da96dd50bf9e01161b66456201466ec18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 16:38:41.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f0a54fb8aab40ca2b46b62b70e08079a32a3645f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 16:38:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f99e036c29368802bbf1933fceb1ecc23a286ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 20:46:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1a6959abb345fc99321b18d9814e99 : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f1206d9ff5de47659f3663dab1bd6ec6 : SUCCESS in 1h 13m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8add0e1c90be40a58ea974f75ab3d78c : SUCCESS in 46m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/951d55431dd04404b8df79ffbda71f48 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74c85e5ef37547a3813cab3b5d23cecb : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffeac098a5304d30a4865065b4c82998 : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa60119badd4728a8de13dddab21417 : SUCCESS in 19m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35b468af593c4f46b99b7b73bf480e20 : SUCCESS in 14m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/14faa8149e68459da268f8d11f7afac7 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dbc3f59e65f6460db698ee1b85b454ab : SUCCESS in 30m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c8261b4cc3d406884640743b388fa64 : SUCCESS in 21m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8eded4086ab45fcacf39625cfa2e15f : SUCCESS in 27m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/99e02b4a839845289ec4744eeb4a67f0 : SUCCESS in 31m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1abd72ac0661433097be91822ad8d255 : SUCCESS in 23m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/83c9124496dc47aaaee459e64d4d5400 : SUCCESS in 24m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38a423c8c95543caa368204fa4bf965e : SUCCESS in 20m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/974edb6dd342450680aad61473b5f0aa : FAILURE in 57m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b82b030c271246d38e1ec84d73d353f3 : SUCCESS in 56m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d86062eed6049de918ab92146ccd49c : SUCCESS in 35m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70fef387dfae49978c10699de60147a8 : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76524c266cd44838867bd1627d5c8ff9 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c0b682ea30a45ebbafa7e01d0f59a21 : SUCCESS in 23m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0fd72f73364b8a8689e04b1333035d : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":4},{"id":"deb25ba7008c13c9947f730b487251b07673cc67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 23:35:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5140f691404dd0dc90ec782975189356459ff3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 00:48:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc9ee7a0eca4acdb7e4d692b5cfa021 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/678dbc77936c43e9ba3bea0b14585984 : SUCCESS in 1h 02m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28f223ddb99d4a5d864f8845fff675d7 : SUCCESS in 50m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3deb85f3c0954056af46e0deaf98f25c : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b82d2792358e40989fe4b433320ff87a : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/951fad92f28043698b52188ec578bf64 : SUCCESS in 13m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7a9c3f9bd144929b93a23ca348ba57b : SUCCESS in 12m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1c3f095b9c204d1d9917dc71e63aed62 : TIMED_OUT in 1h 01m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ed941bbe87b44156a8ee9ee6c34ed2a9 : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bee3b380631145c2b207725df685ae9c : SUCCESS in 39m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9096a5111eee4c71bc56223855ae0ced : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5655fddf3ae64e309ba9a20b413a18ed : SUCCESS in 20m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c26a599677a54d82b7d637ed5b25beb5 : SUCCESS in 24m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/285dec68c7c34002a076b2cf4148d06f : SUCCESS in 26m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/71a64d18d8bf4cab84144705dd572c00 : SUCCESS in 27m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a21e52c091749dcb3dd55a76e6c2f29 : SUCCESS in 26m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9737d92827954badbdfc40f3754a5b6d : SUCCESS in 47m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d87207817de340ffac323670fae5699e : SUCCESS in 41m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03820141784e441999ab34d2da5f6a2c : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db988460ecf54cda9184cae15e37dbaa : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b005f676774b3394ad7b11cf391330 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/868ce77a8fce4ad4a6522669761131b6 : SUCCESS in 22m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc8beb11ef6848338bd9409e347f2da4 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d89ea5905ba81352d03b41ee67f7357cd4aa273a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 05:00:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"757ba853391b86774c1288b2b9f276a3000f5e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 06:08:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c462e0233340a1bf82d68f485236f1 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/56f3f3fa89734bafa0d794bb4ca54ef0 : SUCCESS in 41m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f3944357f304d2f8a44b084dccd83de : SUCCESS in 43m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8370b717335e4573a236cd3459bb30c2 : SUCCESS in 7m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ea212a72b88c4e11a59e3a44b957f544 : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7d1458b55eb44bc9a7b5a09a602b561 : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51ebcb4be7f547da9b017c60114db80b : SUCCESS in 12m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69701f2b8089430daed8250d944a1f57 : SUCCESS in 34m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c9da8c72ccb74b6f878f400e6d22f8df : SUCCESS in 26m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b8e013cd69324974a960157cd17adb3f : SUCCESS in 29m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/872ef3c3953e4053b2652de386317c21 : SUCCESS in 53m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef0d655ec93c45b9a75ac05837f10c8d : SUCCESS in 27m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1b6d1df2447e4aa1ac67cc596bcee940 : SUCCESS in 42m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fe12f089fc52475a954b4917fb354a0b : SUCCESS in 23m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1bb0f0fddb534518b3338daac8d7f885 : SUCCESS in 18m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b8a8addbc0b849e5bd9ae377a951105c : FAILURE in 23m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f931fe63d6c14036b6ed5f8bbc1ca173 : FAILURE in 56m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcdd845b067d4d4e994ee66f96780716 : SUCCESS in 34m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68f2a0124fc84de2bc057f120d1d0516 : SUCCESS in 54m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97e65f1f7ad6477fa8f6d85036624b59 : FAILURE in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c4643b067f4222b7c568096dbd7998 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2bb263bc52b4e0bae5458f28c644942 : FAILURE in 24m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abd7f877fc60446996ea7713357c9bf1 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c640577fc495dffe1f21708b57c2fcf5b542c3a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 16:34:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"03edc36a0d027cbd80804ca41ea80fac21ef9342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 02:37:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423bb5262b874621816e19b9a2df15f7 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/0daa988789d44aaf89e5fabb5c2945ff : SUCCESS in 1h 00m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30491c60a7244793a2d6bbfa6510285c : SUCCESS in 53m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/032bd4f93b024c3993a3f04a32d05029 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5b3b4f17823945a7b70a37a9436bd788 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6616b33752ab452db6889b5f39fc9def : SUCCESS in 13m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5418a623203456ca80502612211a859 : SUCCESS in 14m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8f0f25081384d988dd35a212ba6462e : TIMED_OUT in 1h 00m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/739d7e996a6e4a848f7c989c8b54380d : SUCCESS in 13m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/731692f1b72c45b38b77cb55421c8d17 : SUCCESS in 23m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df04e79c38bb48f1b2d3b779e5874bc5 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e87ab4c41d2435bb3f138154d3c33ea : SUCCESS in 24m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/75bdcc6f4a854ec683952673bdd888a1 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/50405519220f464ab7106c71372b7675 : FAILURE in 29m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f678b9602b5c49caa667de23ac229cfe : SUCCESS in 22m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7cea8a118efd41939cd68e7d4f56321a : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/45e0d37f75d84d12a5f601646a9feb01 : SUCCESS in 51m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3fc908422c1946c6b84d39a8970c9161 : SUCCESS in 45m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef967a21bfed4d4ea9e501ee459a8900 : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3581923803484af290c781f7d9fa9ce3 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d963af3eaaef4735a087041a6b4123d9 : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/906b61e1e8274b75b167f99dbd0ccd2f : SUCCESS in 28m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe8231d45cbe48969616642b3aafed4f : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"1bb508c089ef0d60dd973d9d207773cdba4e37f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-19 17:22:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5d5e475ec8e248de75fe4f15d6725b0c83a9d6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 23:54:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6686941e0a04514a3806f4c356e1a94 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/fedcbf68989c413db0320246eb3128c0 : SUCCESS in 1h 03m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/150b8a32139547e78fdd1529603b9750 : SUCCESS in 56m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53bc29d63bcd4f168304b8cc09ae9499 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/78b684d3a3cd4e12905d3d4a7bbac308 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01753b31b4c2458c865f1596eec3e040 : SUCCESS in 14m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95cb72ef77ec410080dfaf73511dc272 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca3ac35f95e74d8e889babd3fa333e04 : SUCCESS in 12m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0eebd4efb21849988214fa3431f2a661 : SUCCESS in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de28d43fe68f46e0bbc0ede787dbe64a : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a90a053f505438593904602bdc95b3d : SUCCESS in 25m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5126c1c8b32d49d8a0262ccb7e71e4d5 : SUCCESS in 21m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/71506a6bac004e14b17ad34508dce137 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/afce039eb7e54037854528cb5dffcce2 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ca336fcfac94434ab73a8fc150686cd5 : SUCCESS in 23m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0290ee0161a42b0b9eb4bde335c47e5 : SUCCESS in 38m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/effbbc8b813e4297a2cec908dc7ea742 : SUCCESS in 44m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4c16488d20a4d46880337b3b01c9f1f : SUCCESS in 41m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d028cd5a7b9e4770b6191df370167988 : SUCCESS in 43m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a3fe09aecea426ca6cb3083bd5421eb : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5262cfbf25d343e683e7e038b65913da : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f9a01d210cb483f9926e788fb2c4fd8 : SUCCESS in 22m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9240923dfce4bec949795581c584a43 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":4},{"id":"e2376c1fb761b3b4818dca4360c87ff692ab7642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 23:55:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a6b65e8f0b595c9ec247c2e780718b050ffd310a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:13:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9194f7c67e4d719e71c49708ca4970 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b9d734aa9e7f48379c7ac0e5422a182e : SUCCESS in 1h 10m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2487abce3a404e8a8d0ee665db3e4925 : SUCCESS in 40m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/26f6a3b2b8364090b165900a86e382eb : SUCCESS in 9m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c448e1fdf2ef4076a85179e2eded2216 : SUCCESS in 11m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb420f9c46bf45318314c6412463ea6d : SUCCESS in 13m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d65a8962feeb4f179bcc2a53a5b23b7b : SUCCESS in 12m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e2910740ea746b88b77a08a0bec5e1e : SUCCESS in 18m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/12d86ebeac234da68346ea0e28780553 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa1ff679fe434716ac6b48045620f034 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ca5d6dda3e5d4841ab3b46df4375cda3 : SUCCESS in 26m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d19f0164793c43c5aeeb13e15807547c : SUCCESS in 19m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e1315239d01e442ca3df3707b1335f91 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2cc85efb788240a1bf48da8b1b9de6bb : SUCCESS in 18m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fdd7e5b0b7434471bfd7f9f5a55f4f61 : SUCCESS in 36m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a1598d3dcb9a4350b07361d6356b8073 : SUCCESS in 49m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d917f49ddd04bafa6fca32c08b0902b : SUCCESS in 48m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0fdf85b628b44fd974f1442199cf4bd : SUCCESS in 43m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/babf6c527aff4a1f865a60a1ff19de18 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9986f450161548a4822a86aade28d7f3 : SUCCESS in 7m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/184953d1ae9d4244bb4b8ccc56e953d7 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":4},{"id":"57026a42cf1b31dcdb1548799e9efaadf4f3f1c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:13:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"44ad5836f235541e8a7ab70342dfa187f590f1ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:15:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb08238fa3564265b69540adb5d79111 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d5177f928ff74fef860deb9b7fe966a0 : SUCCESS in 1m 01s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5aa4021decd74226ab4f139f8be3c046 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d787ddff85bfcb0c86c0e4b5594995f98cdf048","revisions":{"b1fa0cea5f9abd1d45f2c0661e75f1ce1e2d4ba3":{"kind":"REWORK","_number":1,"created":"2020-11-14 17:22:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/762738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/762738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/762738/1"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-14 17:20:53.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-14 17:22:08.000000000","tz":-360},"subject":"slo: 5xx while validating first segment is a server error","message":"slo: 5xx while validating first segment is a server error\n\nRelated-Change-Id: I99b351eee4fb86b6fcd54f7936b11f4918e1b999\nChange-Id: Ice5358ff85ee2d5fe60785b73b67dea493044a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1fa0cea5f9abd1d45f2c0661e75f1ce1e2d4ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1fa0cea5f9abd1d45f2c0661e75f1ce1e2d4ba3"}]},"branch":"refs/heads/master"},"7726dde8d3afab5424ce6a776605289d52ea9d2e":{"kind":"REWORK","_number":2,"created":"2020-11-16 14:02:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/762738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/762738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/762738/2"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-14 17:20:53.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-16 14:02:40.000000000","tz":-360},"subject":"slo: 5xx while validating first segment is a server error","message":"slo: 5xx while validating first segment is a server error\n\nRelated-Change-Id: I99b351eee4fb86b6fcd54f7936b11f4918e1b999\nChange-Id: Ice5358ff85ee2d5fe60785b73b67dea493044a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7726dde8d3afab5424ce6a776605289d52ea9d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7726dde8d3afab5424ce6a776605289d52ea9d2e"}]},"branch":"refs/heads/master"},"0e934f30a45332c632d0aa0f9db19450b740970d":{"kind":"REWORK","_number":3,"created":"2020-11-16 17:55:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/762738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/762738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/762738/3"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-14 17:20:53.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-16 17:55:12.000000000","tz":-480},"subject":"slo: 5xx while validating first segment is a server error","message":"slo: 5xx while validating first segment is a server error\n\nRelated-Change-Id: I99b351eee4fb86b6fcd54f7936b11f4918e1b999\nChange-Id: Ice5358ff85ee2d5fe60785b73b67dea493044a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e934f30a45332c632d0aa0f9db19450b740970d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e934f30a45332c632d0aa0f9db19450b740970d"}]},"branch":"refs/heads/master"},"3d787ddff85bfcb0c86c0e4b5594995f98cdf048":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-17 16:38:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/762738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/762738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/762738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/762738/4"}}},"commit":{"parents":[{"commit":"90c737e3550a3fee1c7581c3050ea64f2a9255cd","subject":"Use swiftclient Connections in sharding probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c737e3550a3fee1c7581c3050ea64f2a9255cd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-14 17:20:53.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-11-17 16:38:33.000000000","tz":0},"subject":"xlo: 5xx while validating first segment is a server error","message":"xlo: 5xx while validating first segment is a server error\n\nWith DLO and SLO, we validate that we can read the first segment before\nsending data to the client; this helps catch auth errors where the user\nhas access to read the manifest but not the segments.\n\nSometimes, though, that validation fails for transient reasons; if the\nproxy couldn\u0027t get enough responses from primaries to determine whether\nthe object exists (for example), we should send back a 503 to indicate\nto the client that it should retry the request.\n\nChange-Id: Ice5358ff85ee2d5fe60785b73b67dea493044a2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d787ddff85bfcb0c86c0e4b5594995f98cdf048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d787ddff85bfcb0c86c0e4b5594995f98cdf048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
