)]}'
{"id":"openstack%2Fswift~763080","triplet_id":"openstack%2Fswift~master~I959adaabf7c16b62a51e8700a9292869c25fbe4e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-08-26 15:44:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I959adaabf7c16b62a51e8700a9292869c25fbe4e","subject":"Have internal_client accept bytes on py3","status":"ABANDONED","created":"2020-11-17 19:58:40.000000000","updated":"2024-08-26 15:44:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9fe3e61af9cb29664d245aa78dc52b591c7edac","_number":763080,"virtual_id_number":763080,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-17 23:25:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2020-11-17 21:38:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 21:38:40.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-11-17 23:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51d7465c8cd6f54a51d4826a63d0beef2dc10bf2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-17 19:58:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f33b29771e85603dfc84fd97c686594b08f29fb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-11-17 21:38:40.000000000","message":"Patch Set 1: Code-Review-1\n\nWhat in the world is this?! There\u0027s no explanation at all!","accounts_in_message":[],"_revision_number":1},{"id":"d41875ac0275a75befb25cdd31fea367fac8c885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 23:25:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3822b13d244235bf721c1d7da11275 : SUCCESS in 18m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/48a551546d0c403f8a6e2a61918e1ede : SUCCESS in 1h 10m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5847169423124eb1a7c733aea809c959 : SUCCESS in 50m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a83284194a8401f8984f7851fcfa9cb : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/85015f0b8e184ff4aff71a0ea788c797 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e70928d4b274ae198d4f84de6b8b2bf : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d9d8de00c4549fab8b9cf4e219fea6f : SUCCESS in 12m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6783a6b33a9743f3bc198b44040b574d : SUCCESS in 13m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1a336ad654d1476595a8a352e2c72ddb : SUCCESS in 12m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fca3a1912124f37b9d1f2a2bb792187 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5cdcbebd6b142b889933a546a41bc57 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7a60c1683714e56840e39ad17180e83 : SUCCESS in 24m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2266e74670534744b26d417ee2a28261 : SUCCESS in 25m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dc6a53f8d5fd4c61905f1ba262b3539b : SUCCESS in 23m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/db8ab7b1b84a4f7fa59b813f7351452b : SUCCESS in 25m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b0687d0385e4c3383c0fb3b41f46185 : SUCCESS in 24m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b402f42ddeed4d47a8e6da69103e0d95 : SUCCESS in 46m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39852f666a4e44c99c33b0bbf311e52f : SUCCESS in 42m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2f7a93576284398a0fba483726ad716 : SUCCESS in 48m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d63c45f89fb4cadbac5c0cbb23ddd7c : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab03a7189884a3bbd82a4bbeb443720 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61161eca763743d7b51a978ca8753786 : SUCCESS in 22m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf5e5b5890bd4a71b5445060467569b9 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8021fe117603ec6b43d532252ab528782bfff782","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-23 19:59:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9fe3e61af9cb29664d245aa78dc52b591c7edac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-26 15:44:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f38525b275018e9ce9d8a80b540cf44b0e538098","revisions":{"f38525b275018e9ce9d8a80b540cf44b0e538098":{"kind":"REWORK","_number":1,"created":"2020-11-17 19:58:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/763080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/763080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/763080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/763080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/763080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/763080/1"}}},"commit":{"parents":[{"commit":"6d988a451836b02cbab80f8b34832c8cce0ed729","subject":"Merge \"Dep\u0027s should be restricted by upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d988a451836b02cbab80f8b34832c8cce0ed729"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-17 19:58:29.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-17 19:58:35.000000000","tz":-480},"subject":"Have internal_client accept bytes on py3","message":"Have internal_client accept bytes on py3\n\nChange-Id: I959adaabf7c16b62a51e8700a9292869c25fbe4e\nDrive-by: Turn make_path into a static method.\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38525b275018e9ce9d8a80b540cf44b0e538098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f38525b275018e9ce9d8a80b540cf44b0e538098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
