)]}'
{"id":"openstack%2Fswift~763102","triplet_id":"openstack%2Fswift~master~I36afd3bab6dba8ac38e3dead913e527a533d2fe3","project":"openstack/swift","branch":"master","topic":"p-cache-shard-listings","hashtags":[],"change_id":"I36afd3bab6dba8ac38e3dead913e527a533d2fe3","subject":"DNM: start to breakup shard test","status":"ABANDONED","created":"2020-11-17 23:25:28.000000000","updated":"2020-11-24 15:56:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c6ec771c692bd39dbc087c2d8e6d36732904c95","_number":763102,"virtual_id_number":763102,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-18 00:09:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 00:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 15:55:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"df234fdc2b49ad8532f503d3731aa34a76985974","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-17 23:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdc0987de57d05e928a3e545c848d772e2c8e215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 00:09:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b378b13dce5c44c38e52e7ea9dbcc896 : SUCCESS in 13m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/934fb7b801134087b1e276a4ac8051c7 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26cd69266a864177b8702836de18bd21 : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2fcd997530ec45e18a3b81f4a9906153 : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3ecea8340594ac1a5f3b7898e216d75 : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e3dcf97ab7a4a41966603fbeaa586aa : SUCCESS in 22m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85a29ff4e0604918920a28b5fe128a81 : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b1166b6b19014f48874e7fb2af167b26 : SUCCESS in 34m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbbe045571004fe09d4b26443bdc9a9e : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1bf80396681249108ec0d90b91bb32ea : SUCCESS in 17m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/55473eab15e349b3844b51b6ad1c87a9 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8eca57f2616a4100af68755d15565dc5 : SUCCESS in 21m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a1db1aece2384363a7e23a0b6e717837 : SUCCESS in 19m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/620e8cd232784e689618eb1827e808d1 : SUCCESS in 37m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc159f927db149079581a26612ba5fce : SUCCESS in 40m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5403ac21f96344e592b5032d490e2801 : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2588f6a23944fa824a443e7758d7bc : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/505b965a5d2b4eaa89b079fa6c834aae : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bbddd44b99ff91a95903887b4390cd58ec814dfa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-24 15:55:39.000000000","message":"Patch Set 1:\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/761659","accounts_in_message":[],"_revision_number":1},{"id":"5c6ec771c692bd39dbc087c2d8e6d36732904c95","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-24 15:56:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed669bae987704240bc9fd84ec16def7d5b31d43","revisions":{"ed669bae987704240bc9fd84ec16def7d5b31d43":{"kind":"REWORK","_number":1,"created":"2020-11-17 23:25:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/02/763102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/763102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/763102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/763102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/763102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/763102/1"}}},"commit":{"parents":[{"commit":"6850a4383ef1661082f976832cf5e5c3482c276d","subject":"WIP - Use cached shard ranges for container GETs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6850a4383ef1661082f976832cf5e5c3482c276d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-17 23:24:40.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-17 23:25:04.000000000","tz":-360},"subject":"DNM: start to breakup shard test","message":"DNM: start to breakup shard test\n\nRelated-Change-Id: I5fc696625d69d1ee9218ee2a508a1b9be6cf9685\nChange-Id: I36afd3bab6dba8ac38e3dead913e527a533d2fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed669bae987704240bc9fd84ec16def7d5b31d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed669bae987704240bc9fd84ec16def7d5b31d43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
