)]}'
{"id":"openstack%2Fswift~763106","triplet_id":"openstack%2Fswift~master~I3d23d5ff4b3f76db15e84ed37b1cb8503eb58dd5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3d23d5ff4b3f76db15e84ed37b1cb8503eb58dd5","subject":"s3api: Clone ACLs when creating +segments container","status":"MERGED","created":"2020-11-18 00:44:08.000000000","updated":"2020-12-11 22:44:35.000000000","submitted":"2020-12-11 22:43:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763106","meta_rev_id":"8c85043ebaf62e686e9233926dd3549206abe867","_number":763106,"virtual_id_number":763106,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 22:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-12-11 17:56:40.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-11-18 21:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-12-02 21:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 21:43:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-11-20 08:32:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27b0cf71eb33d9e48d3122c517622b380efefdac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 00:44:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02202ddefdeb1615f29995d211b8e78a007d0101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 01:49:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8d79de1e7345898735965892378b02 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/ab2f76cc61a64e3cb1066618851e4efb : SUCCESS in 58m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6344e2f15c11407b8253e3ac1aca0ad5 : SUCCESS in 48m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed57cfe68f134a27bc3d3e753f0d2968 : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/647458d533e7466badfcf342ff9a0570 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5edf746b56442c8a730a543eab91d4e : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/593a73c69b614defb17c1eb72e46be40 : SUCCESS in 12m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9323d317d0b1423eb69a940a195b09ee : SUCCESS in 25m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f7f92e3a7b4947d8b91bd9f3e7756672 : SUCCESS in 24m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd0cf1e3e8794822843240d12dcbb04f : SUCCESS in 25m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e70f521da7dc434582dc746105efa7e1 : SUCCESS in 29m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc317f2f0822418a93af91e3d7e5d757 : SUCCESS in 24m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6ffc287be98144d180715af0c1f589d2 : SUCCESS in 30m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4423884a75394e18870430653d144497 : FAILURE in 43m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ca77736df81349cdace61fd5ab460e77 : SUCCESS in 32m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13f6f819d45440ceadeeed15ff66e72c : SUCCESS in 25m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1cbf9a6b5aaa4c9bb9814c365119298e : SUCCESS in 1h 02m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49eff3c10c8645d690f6d3be8a9bf27b : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c56ed02e1f0499c8447baf34fda9282 : SUCCESS in 44m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f477e43b43ce4556ab7993cc16fa68be : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b769c415225045e69e6c6f82b2b2240b : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/900f22a36f914b64b6a31ca418980456 : FAILURE in 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/867e55f3f87648ec8f4ac6c114e1b029 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d77d4e297295c9cdf47cd5b10482090223e512c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 05:00:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08ebe3e08e68e67ac813ed6b911bddc5d9093b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 06:05:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3636ef71e1764955a5b1002bf5445a08 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/8a3c4b2e56424686b6baaa5cd93b416a : SUCCESS in 1h 01m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d2c6aa85ebf487783d3019e55b53fae : SUCCESS in 53m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47c5384878cd427fa1946420403ba586 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e5e8524c4ed4ba887037a95e85f6ec0 : SUCCESS in 11m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e29b974234734595941a249db534d3c0 : SUCCESS in 18m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/827acade74d54d3e83656cb96b5a4246 : SUCCESS in 14m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e76d9dca2d63476eac0f6b04daa7c43d : SUCCESS in 14m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c26e9f46cf3b4f3d98dac9d03b0a8399 : SUCCESS in 13m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1a54ae81c7f476eb836f495f1ec3b2e : SUCCESS in 28m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64c49b0a2c4f4b28869e9e3f96a8e0c8 : SUCCESS in 20m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a24b14bdd8a5407ab58c1166e12455ec : SUCCESS in 30m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/721f2c36d3214416ba9615da1ac8d05b : SUCCESS in 36m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/89fbfd599e52463aaa0ce017ce95fd40 : SUCCESS in 25m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/953112cb1e124a23ab53822da4cc37ab : SUCCESS in 30m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/29199e8dd4a642b297de4caf28f7acb6 : SUCCESS in 25m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c90d528c7fdd438dacc58571a990a8c7 : FAILURE in 44m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/628dd13e6d7348c18702495f436b99bd : SUCCESS in 44m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/025b6df4dae740bd94ea6353658ed2b8 : SUCCESS in 48m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d97c600eb174332aa62365ac3db16a9 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a17e413babb94f7e8be37501937f2231 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e5c72a4a6e74aa4b8887d8f6b593d6b : SUCCESS in 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4797f30702ea4812becbce07b8803330 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5147a570248820705ff17bb4c8467d952b1f857e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 16:34:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06f223ce18f4311617dbf5fbc2d5123548587781","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-18 21:43:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fa638300cace5e8792141f41bd897e17c430284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 02:47:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9cf6006280455cbd33dbdb8f46a77f : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/02e8dca689b14019b5bb3e09a123ea7d : SUCCESS in 1h 04m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eed9ca7a108d4e92a4fa10cc80b1aebd : SUCCESS in 47m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cea3c48827184402be659be80d69c677 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7030a40944d248a1bb264fd5867b77f3 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78b8a85a730f424b8076168e421982f8 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35c7d0799c634fddb7d380a9b6322c81 : SUCCESS in 12m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eb05b47bd68246edaf788447b727f7fe : SUCCESS in 15m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/622d705c9e68433c98d92fc6e246a279 : SUCCESS in 11m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed80db7fb69c4542a42d0443cdb8a370 : SUCCESS in 34m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0fc1968a86b7471d8e94e334e5110b5c : SUCCESS in 35m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4d7beef52d804694b73e8a1e6512e634 : SUCCESS in 23m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3267f77e58a742c18b69309c0c937720 : SUCCESS in 31m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c9381f93531f4b33a1c771230ec3ba3a : SUCCESS in 25m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ad4fc14046834f7ea0fbfddac3bd9354 : FAILURE in 30m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef72c5ffdb2f4988b9537c9ca768cc5c : SUCCESS in 22m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7879ee7fdf06405c98dffe62b0b96634 : SUCCESS in 1h 05m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fca6849b10e54bdf920f81baa060f424 : SUCCESS in 41m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee03bce106e44887b621e8d492d3d77e : SUCCESS in 45m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5247a3cdfb54fc0a058806b66962ac6 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8f0ef4b1f9417b83948cb41376292f : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/791fb9128d3342408d77b2cfc2e4e0b2 : SUCCESS in 25m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e367f42678be40f7b26ff608daacb86d : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2de538037cb548e8a661e3f4f99725438267dd01","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-19 17:22:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d03e459b84c6345af1a2c36dcd337e59f0e8ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:07:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400dd854c5664a9ca2939abab32e5e1d : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ec9c382ef66d403d85f5b835e963c3e2 : SUCCESS in 1h 15m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2954425e92574796b7ece5d06fccac6b : SUCCESS in 51m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49a8a6f87dc34a6fafaa6d89dbde5d47 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8a944a4eafe4c31b5c3a65dde3059f8 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56139b7ae0d741a29f29b9d38e83bfb6 : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/753d324ee37a4046ba18d17e9de2f2b7 : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b828e104738843479d01b95b37756b92 : SUCCESS in 12m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a48a30a4a04b1ca7b26a8b582b4034 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/110712c2833548a3a68c6c5ececa8ad2 : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed802cc250354677b32e616ee521fdc8 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4460df65238b47e28e8f77e3bf5561ce : SUCCESS in 19m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e52971c4c0548bb91fee27072ca7916 : SUCCESS in 19m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/19f00bce307944259a0736bc56798e32 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/59b018875bf647aab38d55cc457d2601 : SUCCESS in 33m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4cb0732d4f3c46bebe7f4f4ed6c7a37d : SUCCESS in 24m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/46ab9a414d3643e29173ad682413448e : FAILURE in 47m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39f4fc98b21f433f9211f4eb8a365bbe : SUCCESS in 50m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b0832fed89d403994c4e8faebe60e57 : SUCCESS in 41m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cba497bb7c59449b80e5c416a0d743d1 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0da23ba68942d3844510c3f689a62c : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93462b685b9744aaa63aac64ccbc9687 : SUCCESS in 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2d483aa47b4bf59a7a2f6cb15938d9 : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6f951a8995f1989dcdfac1c2f65be0a3cbd8a156","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-20 00:49:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80dd76e56a16dfe6d9a39a0563acc608438e8ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 02:52:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0851cc55ec744d7fbdfff536e7fae999 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d14f5e29406b48cba88571fec0beee50 : SUCCESS in 1h 09m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62fa5437dad54da28322c45f94211f9f : SUCCESS in 54m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8aaf950c79644748e5de8600c5e504a : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f3a253f28b764e6ea48ed42c795201b3 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41fdb84c3c674880a300dd876477baaf : SUCCESS in 12m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93cbd558f3b84f2482e536b0513eff6e : SUCCESS in 13m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e5ce682a99345e587d9be3e0fedb36e : SUCCESS in 21m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60468d9935534e3ab6080689d7308409 : SUCCESS in 12m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a457d4b7c299471489cf18d43159d588 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54d8b28acd424b0e83489643c4eac418 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2034a3330a6744a798a821c79b8effcb : SUCCESS in 18m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c98f8e2786484c10bc939d693e9299c1 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/55988e2d5ab1495297d1505f6953b832 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/97d1c17247cc4b14a0fb0aaf6c833268 : SUCCESS in 19m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbdd3275268b4902bc7ca04d3aee0769 : FAILURE in 23m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfb0f1b17a834ecd862469108ee753b5 : FAILURE in 53m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a744a1130954238aac7231264d9bc10 : SUCCESS in 46m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e922f52158e49d2851a95bdcd99dcfc : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f31cf66fc81468eb27d0e5fb8c93727 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba111c329b74d7faf1e1d9d23f5a683 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ceb7d048a8464847a5e5c8d3945912f9 : TIMED_OUT in 1h 01m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ad037dc5a82421895a5d6465d95a291 : FAILURE in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fe743b24865a0a6c8cbc00397ef8dcfefd2f3e6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-20 06:00:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f759954c62304c2867d7c6a20b9135c4b629fb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 08:32:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e366d971de740e6851980734831397e : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b00a088c922e4348b55c53e3f740b8cf : SUCCESS in 1h 02m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64e20c91a2414864827fd4ab5d932922 : SUCCESS in 48m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3d85ce75d464f64bc4d71b3e7f54b4a : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a588c4ed8f0c408cbbe49b3cec8c1b57 : SUCCESS in 10m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f21f62107dd74b059b36cff5978023a0 : SUCCESS in 11m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9a680ee66343fd8003b79f0636a0dc : SUCCESS in 12m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68b65c3a15424595aab3c31e12721777 : SUCCESS in 15m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/db41d4b9e9d44fdb82efeb99b98d94cb : SUCCESS in 13m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e12d2fab8e943db876c512a5d29cc76 : SUCCESS in 24m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aede75bd17a14512ac2003ee0af852bf : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bcdaee533794471fa85800e894e95264 : SUCCESS in 19m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b65c79fd15d44c9ea8ed90c6d12b81e0 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5d715b5e05294ed39697bb6941a44b01 : SUCCESS in 31m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/033208149301447a9f185df2898cf566 : SUCCESS in 22m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6eaf278121994fb08737a7bf07e5211c : SUCCESS in 22m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d427d8f839b4f68aaf7dd1bfcdb3535 : SUCCESS in 47m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/585cc21e641e4c6694fb5f1a6616feda : SUCCESS in 41m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/615e0d322efa4222a646c2fadd75f4b6 : SUCCESS in 42m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3271979b333465a85c25b70db9c11ed : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afefa1aead9242b0979142dcccd5411f : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e303d155dd243a080a2143f77ccfda2 : SUCCESS in 1h 00m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b2e050994494cd3ab920144ddb85ebf : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6db595021614a4e3fc250632fb38cd628cbb9cd7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-12-02 21:16:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62d369685743e07874f92a8cbcfbe163a38c9bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 21:16:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cc2870c3e2c7a1bd6e154911200fbecad3b3125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:00:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f3fc5746434daa909e6353eee346f1 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/48493a2326fa492d897301da12335c8f : SUCCESS in 1h 09m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a692f87f3232423d87a2ef0d6f0c4600 : SUCCESS in 53m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/215b82da999b40338a4a7bb9e03f40f2 : SUCCESS in 9m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/161e7d0adfa543118a609328a1b99b02 : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1ecee63b0424ac19ef9d7d57c4c1210 : SUCCESS in 13m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59bccaadd269417b82b715c4345a5223 : SUCCESS in 12m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e55635d22cbe4922a0150b401739dffc : SUCCESS in 14m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/90930cdd78474696a6e4e59994461bfc : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59be9590f0134a41a4882596637e7957 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8c1011288314083990b0fe111972f68 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ecb10fa77ae4d0eb309ba24725ef39a : SUCCESS in 17m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e1e3d0ac21fb4d4dab25f29231b91ed0 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a584208e807c42e287ae98548bb84220 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4b53715f8b8740f9b078f7787f81581e : SUCCESS in 17m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf969346ecdc454ca23935e1cdc9fc2f : SUCCESS in 55m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a309c884909433dbc27b0c79dc2f71d : SUCCESS in 42m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/550969e03f0749b6a869e89c7483da55 : SUCCESS in 43m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6149c194d2f44eb98f983d459422002 : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e7dda95d9d418ebcb36f49b559f592 : SUCCESS in 6m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9625718a49a247428587264da09e7af1 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a5a42d2811fddc21f3180cff99b8a48fe4b01ae3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-05 02:16:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a70aae7668728018805ff2cd30e20a49796bc857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 06:42:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ab68d2c0b242009fbc05c4efca2a8f : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/11afd8011f5d482ba60355a1a0f27123 : SUCCESS in 1h 00m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/709f809014ed460d8df99b824b6b8dbf : SUCCESS in 1h 07m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3335daf71ea84834bfe0e72f5a816274 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36fc1aaadeae4e4ba1537088ee2bb11f : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e58fb6c24c54d109aa41a93c53db347 : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e69797164f5843aaacd81a240c829293 : SUCCESS in 11m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c4014afc4948ae870ac7641dcced7d : SUCCESS in 11m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b45fb4ff635544ce8f0781a2a17e59a0 : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9f81638a08c4afe84564a59990ae230 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/465772a0b72a4784885e7cd3609467e2 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2fd18c5982f645f0aaa3b7acfbc423f4 : SUCCESS in 19m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9891eb1587774e7e88626587c1796db1 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4f116f7739aa481d979006ba6e3c2890 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e6fca826541d4f4294ef97cc987d0669 : SUCCESS in 19m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ac7713164cd40cc95257414491e4954 : SUCCESS in 23m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/572e622f9b054a9593fb5a8d678205f7 : SUCCESS in 45m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3dd679c77d9406da6968dd3972bbcab : FAILURE in 14m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4a659246a344fa0abd27f67ffbab874 : SUCCESS in 39m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9fcd203ab1947de97b780a097b712da : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5deba906c74cd0b64b6bbd7768a65c : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5dc2be24a46c4c5e913091b479233a27 : SUCCESS in 23m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7adb35da3a9c41958ba8a204276f26b9 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4ddc665d7b651e38e27b5ef88a5318a4cfb694cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 05:39:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d8a8a498937576ac4f1d6f725c82ecd04b385f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 06:39:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e83aa6af64f41fd95fa2d3298da7caa : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/1c4a4f2a6153438a83555270583bfc0c : SUCCESS in 57m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0df38d1ad2dc45b089cf58acabc591b9 : SUCCESS in 41m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2a46904d0384d3b81dd3c02d729c49e : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5570dc46ed694183b7651d5d53057bff : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd76a5136875415285ffd9d123af09b3 : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/044ccba37b974a758d5c41950f92164f : SUCCESS in 15m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be51e188beae434fb2e80765607724b1 : SUCCESS in 11m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7b7aac0a594df8b24d33b77e9e8806 : SUCCESS in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c74d4ecb5db4345a6f80fac7393f017 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/55356db40c3748219b6bf57fe66240d4 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/922e153949bb41fc8c27ab56aa135af0 : SUCCESS in 16m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ca67a04db50147e4884e593783db17ec : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/652d518a73fc4e5d850009393a304d85 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/43e59ffb5d154a8aab1c961cb01f5c6a : SUCCESS in 21m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad5586da08a54658930c41fc747632ce : SUCCESS in 24m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/356f90f1c1574edfa752ea7e8db21982 : FAILURE in 44m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/976d4496d6f44e48b0b6969a6457de4d : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe95f211d4624a889f39712a78114d95 : SUCCESS in 38m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d762c6924fec47dea750c82108f29ddf : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59eeca5d7854438785d08f6accb89151 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9aa1fd69386c48e78aa6b653b364b2e4 : FAILURE in 6m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcc32eb5859c4ab5a00fe960c300f60a : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"83a278e3dbfe7f1b58b57e7bddf4682dbecdcbb4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 06:47:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57ef067ba2c5d6b1dc76bd57fce46561939222b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 07:58:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a0b0ebf2e0476cb3d5c37f687ecc83 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/e0a6dd9a5b324a12aa467d1446e90506 : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/818b54c93f4641d3aa92b97c36f481ad : SUCCESS in 41m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43ff95e0f88e43cabc4e50bb2d0ab6d7 : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ff27115c49434e208d9cb717901cffed : SUCCESS in 10m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/676a124521844a2d8ae9bd92801902bb : SUCCESS in 12m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6fe8e12695d4897ac504271d568180a : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/468d86fdda6b4fe6b6cca0c8bbd6ff7d : SUCCESS in 13m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c916a903a4694425a2cf361d211394a3 : SUCCESS in 12m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/265fd7af7e0f43a0a9ffdbdbe134c05d : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb4bbaca20b84397b522aee19b515054 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d42bd25ab064175aa1b3fbf8be06eb0 : SUCCESS in 19m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/850704402dc141cfae27ef528efcdcae : SUCCESS in 23m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/653cc997325642a3bc01316d213d2e11 : SUCCESS in 23m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4603a339fd5e4c5a9d2cb88b93049aa5 : SUCCESS in 21m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da66321596c2435c9923126e0fcfa317 : FAILURE in 20m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b77ad06e7b38477a880f1479c0f9abd4 : FAILURE in 47m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f13130844b5e43128d878966cb950e76 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88782e27c355471994f1a0ece73c032a : SUCCESS in 44m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c346290d4a6411d835ceaa9e3f409e5 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5306ebc0249e42fba96d984387ac9310 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01f26601b5d442edb83e873dfaad3585 : FAILURE in 7m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a0e0f158d5c41ffaabfd1da0c7eaf4a : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b11f3e771d886d38d0bea24d71f192803d3ec815","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 16:32:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14af7a600185be19f5689ea650299c3b6bd1b326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 17:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7475761ec1a94f4183a95bee3dfb7f19 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/3a32796c96b84799bf9abad40aafa2d8 : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d718e802e0d24c79ba40af0c0a5faf8f : SUCCESS in 40m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e72ba52bfda494f9a3bd7c58bc60622 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69f3f504263140049ba2068023c39527 : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0a8438bddf94522bf2622026d8dde7c : SUCCESS in 21m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3affc1fa17464acc804842e76fb82080 : SUCCESS in 14m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05996d66fb0048fe9c61307df882cdfc : SUCCESS in 13m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afdbfb9707a24045bcddbb442ba3590d : SUCCESS in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81180c0ca1ec431ca623895a9f57072e : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae47c8712867402cb5f26d36614d6c78 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b43f97f50a16455aa1e10506a46a4266 : SUCCESS in 20m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/934135b48cc541708f0bfbccd9aabfea : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ad9ce7be7da14fa5afb9d01993cad06f : SUCCESS in 20m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7f3d357f1d2f4a86a44605196e52b4a1 : SUCCESS in 19m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5cc6288860944593bd0b68e6b99f12a0 : SUCCESS in 24m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/69492bf51c6d4d1196a4bb8be12abee3 : FAILURE in 52m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c62c1503594d488a8816d1d687438230 : SUCCESS in 46m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1c5e74289f3413ca5961e0e01cb46fa : SUCCESS in 41m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be37fd0f9a66498e8fd273d1730a9feb : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7c231bcc2c42e296686ba0774425a7 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee87095e978c43a0a40280bdbee39333 : FAILURE in 8m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c842d03418a469ca740b8335ea9095b : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b3d7e59b4abfb067e4e0bd4c83f394d885a26972","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 18:02:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7d00682cf0351ba0a3f2acd40e27e2f3072a0c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:58:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433d2c67c2f94aed87f6fe642f048349 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/36862d877f164136903a2c68c816245e : SUCCESS in 1h 14m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d92e286614d344c39224df7bb0b74462 : SUCCESS in 49m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52acdda0fbb14a778130454d9ef40755 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b452a16b951e4bd4a2448280e0bc2153 : SUCCESS in 11m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa81a4adc88948df9461ef934af33c1c : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac3562b744ba453a935aa5d499f4cc33 : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a97ca5c5985f434480e83c9ff0bc80da : SUCCESS in 13m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f99aec338aa3489884339ae5ee7adacc : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a86b93b0eba44f6696248cd9006f9273 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70cfaf043ccc406ca7dac6b0b387ad37 : SUCCESS in 21m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf615e22b64b4ffca972ce6730b88929 : SUCCESS in 17m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/88b646753c8a4f5a9cd1c02bc9be639a : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/436edd2af61044b3a450b0c020720fbe : SUCCESS in 19m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4591e254a83140ef880f292b31321e71 : SUCCESS in 18m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6353bb0c463246dc8856ed5cff06fca8 : SUCCESS in 37m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92e35ff40dcd4329a3c3df04cb775052 : FAILURE in 45m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1f467141ca84500808cc3dce4c00f50 : SUCCESS in 37m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1053302b0565481395ef6c013b3ab639 : SUCCESS in 53m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39b573a915584433a16fda05cfa19df8 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26ae97ab16e4ebb888153b3751d28e1 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c97b7d7e3ab450fb3ac8e7c34622e9b : FAILURE in 8m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b84c7fa21e7467e86e14741fa6a8f88 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":1},{"id":"eeafec3c9da50aa089d5f963c47717800eb46cf4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 20:26:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aab2197d60dc2ae2675d51eca956fb65b8602427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 21:39:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7144f0b125e54a87b3092161a6913da6 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/579cc755d7fa42cdb6120a374923a438 : SUCCESS in 1h 08m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/746ef46e778641b19cfc2ea2865a80db : SUCCESS in 39m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a940be57f0d4fcdae93952d7acc93bd : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b6aaceba17f457caec24b5487b2119d : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63b191abf681407e8203a380ca97d789 : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ead9edee0a04b55afcce0b93d727841 : SUCCESS in 12m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad1f5e7299b845829a4f2dbb733196db : SUCCESS in 11m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe3ce6a8284409aac10e0f4bf5fcae9 : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f60066f275b487fb6829795a2a2747c : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/550f400f8eef4e919af4a673b899182b : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da1e75de26034ed9bc8aa0521f143d9a : SUCCESS in 15m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0ee9489a859b4d6f964439aca19a5212 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/05bb4a65ba99497b9c228b9f77d015b3 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2bf45f53bafe4b878769616eaaac8ea0 : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f047ef94571b4519adf1e44b7e653b79 : FAILURE in 22m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/81ecdd1f2ffd46fa9d819bb90560c9d7 : FAILURE in 38m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9403f3ebe1dd420b851a5387e3d76b55 : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d74ef311eca7485e98995ba192c3d47d : SUCCESS in 38m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c90c8da7404666ad5edee3218ec710 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1573eccb039c4a90bb017d5f702c043f : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd9df5450d0c4a208154de28333bad44 : FAILURE in 7m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06e5ca33057a439c88e01594e6c78c33 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":1},{"id":"91be5b76e2cc0b869f12a8f87efb9367a3967b1d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 22:18:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59b4b697c70acbfc360bb1fcc19faae691dc912b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 23:21:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/453b1f1b7ec44b3ca0b3b5cbc3c85844 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6f34d551736d4f0cb916a33943871cd4 : SUCCESS in 56m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3616fd6cb01a4393af2c35228f9d3b20 : SUCCESS in 49m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/540c7d92b56d4a1983dff45bba85d2a9 : FAILURE in 3m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5df6af5303c74431b223ad745249d0e2 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a38c6cee2fa455abe69bd22e96827bc : SUCCESS in 14m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe26c2cb7fd4663bce29a9128ec5064 : SUCCESS in 17m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b5721d03163436bac9f204bbb7e52d0 : SUCCESS in 23m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4bf58e038184426eae606acb628aeed7 : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8ab186c8842462ca016d8166b0e95b1 : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/80f72e3edd86477db94630004883cfc4 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef253394e3c34874abf554789a152f26 : SUCCESS in 28m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/22cc360dbfbc45f8a75f07dc22ce280b : FAILURE in 22m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eb81fdcb6d304099a7cbea84cb188cae : SUCCESS in 21m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/90725904491c4accab21fa3f2e9fd758 : SUCCESS in 30m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e87fb12d298486d9f7e31a949407863 : POST_FAILURE in 4m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/275aa0135eea4860b296d5f88b8b7b1f : FAILURE in 46m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0b6c73a9ad8461cbeb0d4a8fc8bdc6f : SUCCESS in 33m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b66637d738a0470b980a79fa523babf1 : SUCCESS in 39m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c00f10a07fa40be8e479a06de90d620 : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292eb09016dc478f92cc17aeb663ccc2 : SUCCESS in 7m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eebd36458b03406198e61f891ddd0838 : FAILURE in 9m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6b16003537c479aa9955215fcd303e7 : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1f76fe5f46571dce29c98427a3df04e96da5ac7e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 00:04:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c74712298e316b0bb4f61102d190144cd22e53f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 01:10:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05347179410a40bcafd815e0cee00c7b : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/3c13cf98d0604d85bf21badc58ee5e58 : SUCCESS in 55m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bf4a710b0224441998efd8086013cbe : SUCCESS in 43m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/675eab1e802c48f6b2f40a10a55d2ca2 : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5d90a482d234bcb90146d2504faa3f0 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6a564b7d3914289b8e9ee281352f1df : SUCCESS in 13m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f8e8db6e066406f9e4162346e1324d1 : SUCCESS in 14m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68420adc05aa4d15a06495ae38e4ea94 : SUCCESS in 12m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/829d0b34215e4458bb9b0ba2e2991a05 : SUCCESS in 20m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8062046181e740b297c2c9dd7cad9b88 : SUCCESS in 23m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10c24007bbd24eb389cb74eae11d8829 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b597279f9f240eaa8d0fb15eccea1d4 : SUCCESS in 17m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/24ed9158cb2b4f4d9948907604387166 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/926ea287f83241a4bfbb177bd2be78e4 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ad27ae22779481dbd10eec5ba7c49ff : SUCCESS in 18m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84302db165b24e0685c34905c66995f0 : POST_FAILURE in 5m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/660042864c824c8882f75d686972e4b6 : SUCCESS in 43m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcd56cc5e04347059004bf74ac4534b4 : SUCCESS in 44m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cc7c69bfbe145d7ba8728f2c90d31d4 : SUCCESS in 49m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/926bace0a2504ce7989a6400068f26c8 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927f233269cd4577b6601ece0d0c28ef : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e63b7f4e2036435cb2a26830177ac02f : FAILURE in 7m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9fca97a21ef4671aeb475d78a93f992 : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"39bbec6c3cc1897fa26fba12a6e2f70c622e9471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 01:10:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a74916dde824195e0976b132ac86278c7889ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 02:44:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2008e9c0b848e5b1afb17d2730710f : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/60b64ce2c49347948acff88b2944fc0c : SUCCESS in 1h 04m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5a2bb76202645dc820aa23ebd00c3ed : SUCCESS in 40m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ad6b0dfd3d4744888611d878574dc63f : SUCCESS in 8m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f0b68bd9203a43859fde09d46ef65fc0 : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa751892d3de4d988752b8fb00ed0a2c : SUCCESS in 11m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23fec4366c95487382caa1966e2a92dc : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0602eeadeff24e179d6917bfdc4ab189 : SUCCESS in 13m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/55e79574ae514bddbb9da79a99e3f50c : SUCCESS in 12m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a1a1e65b89b46708aae609f5aa8b167 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32cfd0f9062d44b2937846a2cd1099a2 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86d77e042fe64412a6ed89a25285dcd0 : SUCCESS in 18m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42ae1ebc943645cda88887925c1def69 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3f8138d6c1614051959508f32d1e05b3 : FAILURE in 18m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/216e4eac395f4d3faaf45ade35fcb90c : SUCCESS in 17m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c41d6b52e19b4dec8a265f6088f44ef5 : SUCCESS in 44m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a40dc4a612114015ac69aa4c85c572d0 : SUCCESS in 42m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc39b018835446858ef1cd16e34cba2d : SUCCESS in 41m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a37e0c3e182423ea44aba86c864dec6 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dfb4f61c4a041a4bda69c9ed9d9a41c : SUCCESS in 7m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/914bef31f1b549318f22d9e1e040057d : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5a8b4f4d36720a18bf42ace92b6786932ddf0e38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 05:03:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7d2e78596c28a203ba428f0ec95cc143ed17362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 06:52:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea060c1eb1042478dc9ed4215fecd1b : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/727014a3e8724fffb36f629dbbc271f6 : SUCCESS in 1h 46m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e80c6c9d5af4a5a9b8098a8d2d2c050 : SUCCESS in 42m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41d27dc07ff24a449b0465881198adb3 : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1373014e3a204dd4a698db3efcbd6eec : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/972c6d11e1a9412f84843ca810597bd7 : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0b0f24a7a845bfa6c9523e2669bb52 : SUCCESS in 12m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf25300b7d54bdcaf8d128e4da3c621 : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d6d9e380c6ee42c492502621265456a2 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2a8d56a202a94099965078c2e5f257a1 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6800c12851b7479aa42dab035ab82d7c : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50ba5138650342c7b95d73c36940d3c6 : SUCCESS in 17m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/06dadd90c81441a4a511b643141e0410 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a18befb067ec463da48a709d879643d4 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1ea5a06db98342d8b6ea615cc58076a5 : SUCCESS in 18m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea972c54c2d840e48a3bfbd6da0a02c1 : POST_FAILURE in 4m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/36542372a063442a9304f1cda3816914 : FAILURE in 47m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6687dab7d3634ec182353bce92150ff7 : SUCCESS in 40m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/498dfbc5b07c47c984513d69c2da890f : SUCCESS in 38m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/446e09189af14f6facf6cb51a2ac2087 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e83b0466ed4296be22e3fdbda583ec : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/666b3a11b5ee4e46912fc7616fc987e3 : RETRY_LIMIT in 5m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9816279483fa4296950342dc69cbd9e5 : FAILURE in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fa6a686cf7ab1ef2f57e84c098c55e6d587b835e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 16:15:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f60ab4cd74d5109e19af73b86f9398ba353b0eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6659eb22a6454b6486dab89197a25fee : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/2fca756bc73c4f13a7fbfe24bbaede60 : SUCCESS in 1h 04m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fe24e98275946139c24f6e7f838370a : SUCCESS in 52m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34f5ad1578d34231bd3a264a3656f041 : SUCCESS in 11m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e289bdc74cd34f509725517027cf6b84 : SUCCESS in 12m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35bd401e29b34ac39c1ce6d71b6d84e3 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/559b38eb7a7940c19022807cd3208666 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/621edcfa073b454480258e77f2516354 : SUCCESS in 14m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1c7402a465214f9792098313b5689f31 : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2029360833b145f7be0ddde0ac4e85ea : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6273c3414f3143999203bbe1f28a9407 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/18f0137a1aa04fdf98db18f44ae49256 : SUCCESS in 20m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7366d7fec3504c9c810c068fa5d13b3e : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5ab0b99ebb044c8ebf4fe45fd450351d : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/479ef18efc6a4a868b383f5a395a4b4b : SUCCESS in 20m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d32bbc29f3547dd90865f392eed4135 : POST_FAILURE in 6m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/794293a51b0b4f9bb19653fe54234441 : SUCCESS in 53m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6c0ef90582b40b2b93603e470dcfbd6 : SUCCESS in 43m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/895013d377f14d67be8168517e6e1828 : SUCCESS in 43m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08e534cb892544a29b41fe2f9d5e2093 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770314275d114891820c4eb3b0b2b60d : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18b4e2f9fff34682b20ea09dee3dfcd6 : FAILURE in 9m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e6feae49c764d77830fba78ba9dfa13 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9ff900185b11ecb4efaac1b2b9d68ceb25af598d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:03:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"995b8da2b79cc3c06962290733d4b63687ebc689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 01:30:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb54ed4f72304512811a4d28c7f8057b : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/66aa00f3992a432e804cc45ce1af7159 : ERROR Failed to update project None in 2h 45m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd5c3b3860e9435eafae4dc71a67234c : ERROR Failed to update project None in 1h 49m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/81add9844e6e449aa21485785f8f5939 : SUCCESS in 8m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/97f8e5896ce84542a21fe4180122d5ff : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0e6f8c6078e4a71896cfc412ec9204b : SUCCESS in 12m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd80094cc9245ba9ed0bbe80ed5eb50 : SUCCESS in 10m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1eb07a306eee4109ad2717e7e5a70939 : SUCCESS in 13m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6b33f22fcc4b52a137d50c8c2e0853 : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bf675eca16c4bb6aa900f2954c986ca : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d2fbfcbdf224f449e09edf1c798cf3b : SUCCESS in 15m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6b2ecaf64d84263b8c516fd9d4669b6 : SUCCESS in 17m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bee6e69e7368419194e5acb027de05d5 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6aa459eaf8c2486d8846bade72f374d1 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/75315363c1bd4dd196b60e080b2e0f27 : SUCCESS in 19m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cbeb8b9a6d5d46929046023b923f82ee : SUCCESS in 48m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d27bd5ef2f0f4b22973a28146e8d2338 : SUCCESS in 41m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0606a03eef0c41b1bc7ce5e96b6b334c : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a84a016b2e9421684f59d40f19ee9b9 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa7f341a1984d0ab29d41fb98db97e7 : SUCCESS in 6m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dff1c92d2794951be177cbbd75883ca : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"96f38902b35126b98337177af9c19d9b5a5d9168","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 06:37:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6632b2acdec85ab52ed7862105eb78f6429c30f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:46:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611bb8d21d0b455f9e1bbd7b55294c1b : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/6c7bbed27cea4d7d8d286c9e3cfe7b4d : SUCCESS in 57m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1346984e30e9483d9d01088c4f09a26b : SUCCESS in 49m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/833adf620fdd47af8a6cd771082685fb : SUCCESS in 18m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7fc18712e2f4c9f87947be7e5f4aaca : SUCCESS in 18m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c77ca467a28f4fb5a3f4b3a1e8caa1ea : SUCCESS in 13m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9094b3b918d149638776bb8fda9bb775 : SUCCESS in 14m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bee65b5df3904b2885a412db9b84c67e : SUCCESS in 12m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42140d572a144cc08f290dd73ba407a0 : SUCCESS in 21m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f9c99da48624ca89a5b7b85c87f4331 : SUCCESS in 22m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bb40d8f8da2417896845961760b9986 : SUCCESS in 22m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee7fc78776404674b69dc73113409de4 : SUCCESS in 18m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2b9761fce59f430f805ceb650032afc8 : SUCCESS in 25m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f36209d5f7e445b1bfeb847e9c069ef9 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/acc4499030d94359b0ea29778bb65396 : SUCCESS in 19m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/177debd8a3a946618ce8ec8129930cdd : POST_FAILURE in 5m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/983f59418e304fb5b7a281e8c1a90a4d : SUCCESS in 51m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fea6ebfff26a4e6890b3a3f659a9fde2 : SUCCESS in 59m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afcb7222fde1474da24c19eb7fe4c4c7 : SUCCESS in 40m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1efbcb8b7ed49a3b6ff690396c05f30 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165ce7c1ec524b578ff19549fa969d50 : SUCCESS in 12m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c6b53a9401a4a3a845bd2aee0dedd93 : FAILURE in 10m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaea7a9d8eb944b0985d892c248d18a0 : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":1},{"id":"403bd57672759ca82494074232d6e0da342686b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:46:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5d1fe21897ba1b5ee3e458bce28fb022330ac41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 09:58:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c174309865b462eaa293b519c6b559d : SUCCESS in 14m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/1d2d5262340d403b928237a8a4186f10 : SUCCESS in 51m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b3ff301d66245e9bed700ef1ae709cb : SUCCESS in 1h 10m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a0e07f100e2745eb9aae8486a8d709be : SUCCESS in 9m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/96ea96a3132344f1b4a53765d4fd8213 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/133298a39d1547fe85b95bbc84893f85 : SUCCESS in 13m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/719fcb06803c4570b98a0311abaa4b28 : SUCCESS in 13m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f3a7b3629d147f0abf4ca2815389eb0 : SUCCESS in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/526e55c14c134c39aaa74b3c474dfae7 : SUCCESS in 13m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b3833d3a42f47aabcbb56e10fb0263c : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5c94d61be7d4ae1a939732027dfdc6d : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87a91f730433417f88bc003a99619853 : SUCCESS in 19m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f1565253eb2e4f3f8a8db85fa0b372ae : TIMED_OUT in 1h 01m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8bcd32bfa11045f0a2cbe93ffa08f767 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/65bf76b67c694b03ad28697defb868df : SUCCESS in 17m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9df7f24c48f64ef0843762087cd59591 : SUCCESS in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1160e8f3bea4fe98592aa49ffbab2d9 : SUCCESS in 45m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/400870668a42426bade71beae7c64956 : SUCCESS in 35m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3535d0106bdc44bb9317e64a9566e183 : TIMED_OUT in 41m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9536adfde35c4597b2f111cdb03f2d49 : SUCCESS in 6m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b700ceb10b94f55a13b91cd0465a3dc : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b00013f9e5b96c267ab65388f0cdfa6b03dd26e0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 16:09:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7324328d0ec168675ce74914ac9ac12c6e27c448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 17:24:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e9a049a3f74e568f4b56319583c915 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/903f4e580c984600b223483f3a4f120e : SUCCESS in 1h 04m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d3a986ad1f945e098c4dfe60b64ad7f : SUCCESS in 1h 05m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29a8559df12142aaad75eb29601ea039 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/201380b27cad460ab067a4372610124f : SUCCESS in 8m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c9cba4ee1594bac814870f490a18b0c : SUCCESS in 12m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28d7cff9a5974f6aa5c94242c3dcaa2e : SUCCESS in 14m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65381d3e7a3a40fcb2b7f2e907bb8b92 : SUCCESS in 13m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad87dbfaf874af89782fc4b62847766 : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/184f1f3d64af4a598bae934b06dfb028 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6889d8cc9092436692ca908c4945ecaf : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2097ad67e59049308c3d30ef74b50113 : SUCCESS in 18m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4cb4f8a624524d1b89da70cca1e9a195 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e485b74d56ed4d67bb92ee2ccea3b7ac : SUCCESS in 19m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/db4b60b878e440be8f3e890ffa466973 : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5974683c4cf44d42b7ebdf16bef9e509 : POST_FAILURE in 4m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99145f13f19244c78cb644da17616dfc : FAILURE in 44m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bed3bcc2afb4eea931ef4134d5dbc46 : SUCCESS in 50m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed061a758550433fa625ec18dfba29ec : SUCCESS in 39m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbbe6ab4d9f5489a991328d83f35f6cd : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5eedd3eb574d99a0add0f90819ba88 : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11f7c4eba9204fe99e4ff079d583bd2a : FAILURE in 8m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f29ce2e793574fa48bd4bd00f94631b9 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"aacd9cc18dcae84961e5bffb00a4ac78a71c39a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 18:55:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ebf10ecf252dd0e0d8159e6296df697f00247c86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:34:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3487699510e49648758211c1f5226df : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/9a8055462bb54956be0482d9f064212c : SUCCESS in 49m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/34fcf66e76c34384923493d28f09ac99 : FAILURE in 1h 36m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a79226b8197f439ebdf5a9557d10c14e : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89e69a5a249943e9a5037a0a8a460397 : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88f119792dcf444e8688a5bc54cc4bb1 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ce89e14c7274347bf50ff03a43f1e07 : SUCCESS in 12m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6bbf63e574d243b6bbd05613f3383a74 : TIMED_OUT in 1h 01m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd24be17da64ae78cb268b27fb2f914 : SUCCESS in 24m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0093782def7c49ec93d30c21aa3bf2ee : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/388fcc72f076436d971ca210597ad005 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/522851ace626477fa6a60dbcfbc088e6 : SUCCESS in 17m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20c1539f9a74451ba79c9a50b196b42e : FAILURE in 35m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ab092b9cb58f42349b0e624f21ffdc38 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9eb37dc7be9b4f90b8b256772c4d8013 : SUCCESS in 29m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49f80386f0f540c3811c840cf6472d96 : POST_FAILURE in 5m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/baf242e46c4e42258910ad05f7d6e709 : SUCCESS in 48m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cff342453fb84396b181290cafcba003 : SUCCESS in 39m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/178d361bf46e44529e4b80d7ad2f5206 : SUCCESS in 40m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48171d36772648fab5d9e7e3b4147f48 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f6ddb54a7c4d4b96bedd939a971b9b : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06aa5500640f4438853b7cea264e6a72 : FAILURE in 9m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eef58f995a24c56ad07dec36288256d : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3a3e6a70fb97322a6bd11ef2a3fdbfbb13283b82","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 20:58:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14e27c965403d6199c00df1c0083fd97b3d55b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 22:07:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b95071e2e044ddbd9118e37a6c45ae : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/6ec2c52494264964a3e01bba3396f71d : SUCCESS in 54m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba5409e3d6394856bfde94b8af3effce : SUCCESS in 46m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9643a302f38e4649841fa6e3b6de7966 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bc33434cb6346b4838fcbc5cfec21a8 : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0767f6981d5249ab868b20b7eb6118c9 : SUCCESS in 12m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb9b0b4c1aba4bca9a3ffed7326dc7e5 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddeb5d96beef43a9927ba082afdcece6 : SUCCESS in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d4224acc913a4f768535d1c42b111b19 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cce010c6271b4cc2b09bd59141d89344 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a540c5d2a994a86a81a6137e4d8f939 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae9187cec92044c4b99d12ecf13f34d1 : SUCCESS in 18m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7dbf89f909274095b0a72cee8a1fa731 : SUCCESS in 22m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5d34e33b18864e1696908e73a15fcfd0 : FAILURE in 19m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c2844d1244c74548847d4ead6a656df1 : SUCCESS in 18m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87cd5f53e91a4b59848d5e05d77c176c : POST_FAILURE in 4m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/511b2ff7863b454cb3aaa49d3d748d9e : SUCCESS in 46m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f35189caa6a49109de2326635e7a2ac : SUCCESS in 42m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/635a5b01879f4a43adfb906efb595f8a : SUCCESS in 41m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1cd9f348be24e7cb9b229b81229acbb : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37affbf7eb6b40c1b9b7f802597768a2 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7bfe0f1f296431c9b1ce279de2ba9ef : FAILURE in 8m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07d7a7efc5cb4b4e9f554067b28145a0 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2bf357fa09b26bf5473a40097803aeaf30bdbba0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-11 00:05:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3de624508c2c7c71d464c126d6fb9ede45df3c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 01:17:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b18d7eff52d432dbea511386670745b : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/213d8bde740a40a7b8abf91a8e11383b : SUCCESS in 1h 03m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a19e96b8c03c43dc951035cb70fd5a55 : SUCCESS in 46m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b04af284b10c48039309b78f84f76094 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e20c6e7c0704ecda957b9dbf6c492e5 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44535f14927c4e62bce6efb2df87ad89 : SUCCESS in 11m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cdf3c8b0ea44faf997ccc965f8c76c5 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa5ceadeb4eb47f88baa02b05d7028c3 : SUCCESS in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/852d0d809c1848d0ac18642a222c227c : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ef23d1604d8c4cf19185cb3054701ce7 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8f28899b6154227bd89e039f62736fb : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6aed850686343f5831e09e80fca2eb4 : FAILURE in 16m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6eb4924803854b1d9a0e822a98722962 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5b7de033a914431886fb1a4643135c6c : SUCCESS in 16m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1b46af50926042558881020e0b3a9070 : SUCCESS in 16m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e604691740947869bd4d2cc49503431 : POST_FAILURE in 4m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/764502238b974f1aa2ae15c0d330ea70 : FAILURE in 44m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf75c11acdfd4aa28b68bb5795e54dec : SUCCESS in 37m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b63f5945e005468f86c0fce02ec302c3 : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff9c4ec215da40c2a5c9aebeb0fb0841 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860faa953f37423fbbef21db291c23be : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8ee690ea7d2435db80a9a6e0ba10842 : FAILURE in 7m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88e8ae90a9a0408880fe20526cedfd20 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":1},{"id":"dee8a5037a2d75d4b7dd5f611fdaa47776b51134","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-11 04:18:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4185db7419bc93dae5a1eaaadbb8a96e8b93773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 05:38:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73938eca374049af905f482d48064deb : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/865a264498a1427cb7f4c11746eeaa14 : SUCCESS in 1h 13m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9087bd0373c2474fa478784ad2d60f6f : SUCCESS in 50m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0141beddb19b4caf91ba8c1a776f99cf : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f56ef09320e47f6adc6915e132a55ef : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26a923a9b42442db90c38d7a892921bd : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d413d82916c14042a4a331a0ddeb5ced : SUCCESS in 13m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/abd0b77a85254acfa9f59583f545964d : SUCCESS in 15m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/77d82d8d560344528e28571694e24b81 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73c9ab1bbd0444f19048c9b640a64b8d : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6fc29b33b330455095321471f05b29bb : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d84b224eece64eec9b5387a42a5d50e7 : SUCCESS in 17m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7c876872efbb4e978c6d1c63f3e39878 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d357d1a6d6ba4aeda4a2fa5611909bae : SUCCESS in 20m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7d2f9ae41a3d4669b26752b807cf8cf8 : SUCCESS in 18m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d6f618ebaac4077a593685b8b5e2cd3 : POST_FAILURE in 4m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ffd04e7f46d740da9796bbe92ba51e8c : SUCCESS in 46m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b511e4c8c45143f5b0f92b8513b1be47 : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51efd6b70066456099b148b45fb5d48d : SUCCESS in 38m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb6b2a70210343f9b8058e387d5f0fb6 : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b253173a6d5402ea537f11f8c93878f : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83b4648ce4ba4966a15a2fada0c9e558 : FAILURE in 6m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c44db7740a844cc3a87ce798c56d7d49 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5ce6284c7f755cbeb4c8daed9c18f3434b3961e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 05:39:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dacdeab705d65f5984fc47914f405adce764161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 07:01:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aefbede92214fc7a7fda13bf216b48f : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d3ba7fe665114345875a899fb76112c2 : SUCCESS in 1h 03m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bf667f4cc8747d995a54ea63ce4b884 : SUCCESS in 46m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/685ef5875f4e4d92a75f662582a447f9 : SUCCESS in 9m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3812d7da4f494e05915a51b060da80cf : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d70d3d9a9c64417b8fc470708d257f49 : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/655eb751b24a470089860be21f9a29c8 : SUCCESS in 13m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be41c3595f4744e2b161e3a7dff99251 : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03c8bbc49c424f2b92a419c1020e61b8 : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78eb67dab8024c8292370807920f2c96 : SUCCESS in 21m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/179527ecb35d47509e40c370d9476d88 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5609fed877eb484cb30b34d97861759f : SUCCESS in 17m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6e31755d8a0e4678b4666600e0b691c2 : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4e9503c7815b4829b05b822ffcbfdaa4 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7dd9c60e83f7440b95594983d2eec62e : SUCCESS in 17m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f27fc3cb0c9a490c85782ea67ae8b692 : FAILURE in 48m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d19c6ed18cca4c79a8aeebbc99cf5dfc : SUCCESS in 45m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04e1601f558543ebbd11961c8c618179 : SUCCESS in 44m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/342fe6d0d9ea4a6a96e1116274bf38f8 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743d1ef5079f4501b1f598fdf5d9e513 : SUCCESS in 5m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b292c0cff9e44d208ba1a895cecbd22c : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"58815507e69e23e38c214fb1c367966382bdd66b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-11 17:56:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"548d45ce22a550414f08960cd24e90b16ef39c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 20:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b745f65fa884afca8b9124ad52d586e : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a6131cba3f7a423d886361ee136eff4f : SUCCESS in 1h 01m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eac1355925004bf1adba56509ac09445 : SUCCESS in 1h 00m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2cb845613494ea3929cb02480001624 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c56fad6a501c43ae89c3497aa79475a3 : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7c781748699c47c5bb90c2ddc2462787 : SUCCESS in 22m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8cb75bbf9524f64a3dc324f14ee587e : SUCCESS in 27m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/000923ae1b884f588c7e94f8c4da51e9 : SUCCESS in 13m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42ca32effd30460c9a813c26d10e1695 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/94b6a3f02b1546c59930b4e0bb9367b2 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ac681b8123d402f83b0cea6a2bb314e : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d7b62a9cde646f0ae5ee6eda5ad34be : SUCCESS in 17m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8aa136251ebf4dc4b9d1b7c75aa50aa6 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd49ce927ab14aceb73ec340ab5a07d6 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/13f161919d834eb49652716b72580ee0 : SUCCESS in 19m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c784242bc8de414aa7ab19983af5a945 : POST_FAILURE in 4m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d3d2d4e70249419295b07ef59fbf00bf : SUCCESS in 41m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f44f78c9035d4b449c4826e5101b8d5c : SUCCESS in 43m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78aa68747d7b4b299a53e24703e6940f : SUCCESS in 48m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f52544f3418463f93762358bc870658 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60d5144cfd5141b88c65e269e2319872 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4afd9c5fd83e4d0dabcd08b5c169a156 : FAILURE in 8m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f61cc59a9144204b9dc3441bbceb112 : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0e746557c765b2feda7453684f3ee00be8a24c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 20:23:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69621a48f1142cb1a5a07acf7019bb50004c2362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 22:42:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70cf48ace316470a8c15b7a96e19968a : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/31c30007f8d1444b9a13b690d6eb2587 : SUCCESS in 1h 06m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b567d78c48a3445e8e71d7df76bf44d6 : SUCCESS in 53m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8a4db9d0ebab4f7785fdf60bcdd1992f : SUCCESS in 8m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a9a05e93beb7420d9c7688b9aa5142bd : SUCCESS in 10m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59fa1a2992914b40863a8b90f520f801 : SUCCESS in 13m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1dd56eaaa4047c2afeec564ba93ecf4 : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a45d523d18bb4765ace9f7bfdd38c0b9 : SUCCESS in 13m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50534cc8f0b44e30978a45c8160a683f : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/918cbc9c8202450892c1ca10ec6ba030 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ccf7ecc57d064989963ed6300000d520 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b5ea79c010a54d9d88fca2000562c70b : SUCCESS in 20m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45e60730bde4497b93a8c5d23366d6f3 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9d98641cfeaf46628c813960a16768b0 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f8a30280d3574e43aa7ccc51b1fff12a : SUCCESS in 19m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/21367a9b633b49c2a77d37538469f724 : SUCCESS in 45m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d07709b005044a086c7c05fa14c1830 : SUCCESS in 46m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ddeb9dec545b4757b62ac02ddd867924 : SUCCESS in 42m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ad6420fe7434cb3b8539e800a28ef0f : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f4de29f89c42b98b2f3d5769bba6f2 : SUCCESS in 6m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f0e82454fc413f8a5f280f925cd23f : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3b9e4a0a42f1562d3dd3edaeaa9517523539f8e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 22:43:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c85043ebaf62e686e9233926dd3549206abe867","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 22:44:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b5d25288c94330a775d74abeed25fd : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cc4c1db693e145b1a5d7684f1ab8846c : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3d4208d5d6d342b49fded476836f972f : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c2c9bec7eed9f1c8a1ce3135cd89385df4d6520","revisions":{"8c2c9bec7eed9f1c8a1ce3135cd89385df4d6520":{"kind":"REWORK","_number":1,"created":"2020-11-18 00:44:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/763106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/763106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/763106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/763106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/763106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/763106/1"}}},"commit":{"parents":[{"commit":"6d988a451836b02cbab80f8b34832c8cce0ed729","subject":"Merge \"Dep\u0027s should be restricted by upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d988a451836b02cbab80f8b34832c8cce0ed729"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-18 00:43:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-18 00:43:53.000000000","tz":-480},"subject":"s3api: Clone ACLs when creating +segments container","message":"s3api: Clone ACLs when creating +segments container\n\nChange-Id: I3d23d5ff4b3f76db15e84ed37b1cb8503eb58dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c2c9bec7eed9f1c8a1ce3135cd89385df4d6520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c2c9bec7eed9f1c8a1ce3135cd89385df4d6520"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
