)]}'
{"id":"openstack%2Fswift~763251","triplet_id":"openstack%2Fswift~master~I0ae13905443aa8c8a1941c3212d01c9a5ae674e6","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0ae13905443aa8c8a1941c3212d01c9a5ae674e6","subject":"Stop explicitly handling StopIteration in proxy-logging","status":"MERGED","created":"2020-11-18 19:46:24.000000000","updated":"2020-12-10 21:49:44.000000000","submitted":"2020-12-10 21:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763251","meta_rev_id":"05007ae52bf4ad26286a2e624b068922c6daa9b6","_number":763251,"virtual_id_number":763251,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 21:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-11-19 04:11:00.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-10 00:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 04:11:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-11-19 04:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12b2a67ce49ffc5b234d040f6e2058555aa6441b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-18 19:46:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8259edd4f2210a139cd54a3f50b8de71d8231095","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-11-19 04:11:00.000000000","message":"Patch Set 1: Code-Review+2\n\nPersuasive","accounts_in_message":[],"_revision_number":1},{"id":"232a420a9a1e0a620a7917403e4b41689c2119b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 04:29:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db457e5f99e40b3a718b5aee9fd3fc9 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/9bb2090552014231b5e82759f53a28d9 : SUCCESS in 1h 05m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc000047506544d0ad278112e7c95451 : SUCCESS in 55m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/160e54ad0b4c45ee8bc7d40c0b5db810 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/51f63f472b24444ba6713a32301fa530 : SUCCESS in 13m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90f6f04828574df0afe67e9c484eab08 : SUCCESS in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7362ecdd47c64ebcb81b9b205698ad5d : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/edc873edfb6845c5b9ea4849c38d98f6 : SUCCESS in 13m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/27d760dcbacd481994781b0066d2c5af : SUCCESS in 14m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ca794dd6fa84df39cecb3216678a5bb : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f44b924292d5411da27bebacff21714d : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/716c7fe5113249f3b2c27f4da71b18d0 : SUCCESS in 17m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cfd325e696b042aeb07bcadbe0f2d372 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/48f315b484fd44f193e98102e81e1236 : SUCCESS in 24m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ae744e2cc9084afe9125dfa8ae576c29 : SUCCESS in 26m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3962b8e75e3c4bb5a891229ee643b5f4 : SUCCESS in 24m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c0c54f4a47c4618a67e57d6f968ac0c : SUCCESS in 42m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1f2b06457e344d6b8c82f4969eaaa86 : SUCCESS in 57m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b86638b96cb647a8a86c09cf4591f206 : SUCCESS in 47m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c359e3650324343aa2f4ae02abce13c : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ccb26924c1c4b5bbfe8f92bc6e10b5f : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/483140debf344880af367ce9ba29e237 : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/365e208191b3426d908a6dfa7ccf85c3 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"47d9b44170069770b8617d24ef391800fcde9a11","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 00:14:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cfe265448638810187448ef9591a898de48e46b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 00:15:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16412e67a53f1d72c54fbd20247b8427b42057a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 03:18:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53bfd506add44664b2489b3699ce2914 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/55e2943583a94ed9984dd0986b2f7a20 : ERROR Failed to update project None in 31m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8f819cd32944094ac75ee99b45bd4dd : ERROR Failed to update project None in 31m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d103ea5aaefa4efdabcdf23d451d7d4b : SUCCESS in 8m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/76dbdba3178642eea5618b1c1194f2e3 : SUCCESS in 8m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c8340d9d2eb04b28a23a38182291c78c : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/825fcd2e1d9643d9942ada5ed6cfcc5c : SUCCESS in 10m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/efc08b5b1489446abf9f7f2dd957dca9 : SUCCESS in 11m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ecce0da3d2423ca2b2ec85aa4f77c9 : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6ca8c36b8b3470ca1eeaa0df367f40a : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e058e6bcc9f94a479fe3b77686467fea : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0195dbd1f7504565b56911125b99ad7e : SUCCESS in 17m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bd7b230118004f18823274d287720ec7 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4d2fb60323cb490fbe442f8edffec7dc : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/06be5c06322e48eaa77c21a1884157f2 : SUCCESS in 18m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/47927ef3128c4d649956998bb483f201 : SUCCESS in 48m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08a96ef9a4584e6f9575b301a1369d5f : SUCCESS in 42m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc1ca64db8264ca58311a79225a694b2 : SUCCESS in 41m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae1df4b7d98a4760959f430d8690b4ba : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ea1c9864664458913e1c5e0493b794 : SUCCESS in 6m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02884343551246f9ac6d7db9012775e5 : ERROR Failed to update project None in 54m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dedcc9d13751cf664d8cf331fcf857aaaad1f5f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 06:37:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb04c9ce6f0011d6245567bc9f69437033a64bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:46:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4fbb66256b4d02bf4c583768580bd8 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/9f7a2d89233e463e947834b4806ef22f : SUCCESS in 1h 09m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6482e73afe6243c5bb151e3fd2251896 : SUCCESS in 47m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31e8027558ce4eaca24cc7dc1dd689d3 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/052bdc8a2b594c3f962104d6feec331d : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abb7491aabe1488da2b0f8f3394dcc5e : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7d5803d0c7dd499c82eaa41b068f2905 : SUCCESS in 16m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3992894aba2a41ab85f485c31c028ed4 : SUCCESS in 14m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/973a08f78e954c3a8686585048fe9a33 : SUCCESS in 12m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2012433621784467b70d3f09b199242e : FAILURE in 18m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c8a4d4aa9444ff692a14dc8a7ad0df0 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/646fbc6c360a4c11ab974ee36cd0577f : SUCCESS in 17m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4f6d64c8d202465da44464bde75d1208 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2004c1ba8f784f6ba34cdfa78d5c0cba : SUCCESS in 19m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/de039f8716084b7f8655850990a2f337 : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71e475ddbe094875bd28785f70470d4d : POST_FAILURE in 6m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/23cddae624764fd9b883e598a252be08 : SUCCESS in 53m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4a3b1d0279d461889320bcac1b9e765 : SUCCESS in 43m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8cf7e9422da4ceda6e4bed6de500c21 : SUCCESS in 38m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0123383136b4088880c0c77e4c531bf : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636b5da9c7474f3586bccd6e9188e9e0 : SUCCESS in 7m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c96537e68e44d57a7f4627a82674016 : FAILURE in 10m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/417723dddacf4c09909ac9037d80499d : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7bad5cc46ee25a9f05844db792f60c0b1c780451","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 16:08:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7671f5111b7d843e87351717f61c467e38b626ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 17:38:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fbaaa84f2f4e5db91739fd6f72dd90 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/597838d8be7247d8903f0db584894d1e : SUCCESS in 1h 18m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35b26da583b5486bbb82d48dd1ec972d : SUCCESS in 1h 11m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c76939d1e220499b82b280530f8090bd : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bf8448b296242ae9ff8a984adfe4ef8 : SUCCESS in 9m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/139c1a91970a4904a84b28b539e95b35 : SUCCESS in 11m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcb64a814c624effbb2da9ea1f37aade : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0985efc0f93f41f2bf3f26dbf929566e : SUCCESS in 12m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac51ce2b0d0548bc861e48b083fd771a : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bbe61c378d14ad3beaf087ef139b2f6 : FAILURE in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59f467192eec4494968e8115816e595c : FAILURE in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78409be239f64fd0b4da1f5bcfd2b038 : SUCCESS in 18m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20fdbdff7b3945888133bf9aa2ff317d : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bb49949410ed4b34933ddbb586e97278 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3df0971696cf4ad6bccd1af5e9ef4feb : SUCCESS in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e97eb4cee7641d9a56903c0d27c8a6c : POST_FAILURE in 5m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c56c339c92fb45f480d27a20d042b743 : SUCCESS in 44m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ef9809394ba45b785ec7c42dc7f4eb2 : SUCCESS in 50m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c67dc5e8385e4bf7b489d820757367ef : SUCCESS in 48m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97411405c610455086b6cff14183a519 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f1fb7d7ce643408687bd1934982c12 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12256e24b3e941c8975aebc1b0e8e6cd : FAILURE in 8m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e491d8d809ce41178245a555dacfa65f : ERROR Failed to update project None in 4s","accounts_in_message":[],"_revision_number":1},{"id":"1f3e526c38af298cabc8d06ef46be7c4a760c48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 17:39:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fbaaa84f2f4e5db91739fd6f72dd90 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/597838d8be7247d8903f0db584894d1e : SUCCESS in 1h 18m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35b26da583b5486bbb82d48dd1ec972d : SUCCESS in 1h 11m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c76939d1e220499b82b280530f8090bd : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bf8448b296242ae9ff8a984adfe4ef8 : SUCCESS in 9m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/139c1a91970a4904a84b28b539e95b35 : SUCCESS in 11m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcb64a814c624effbb2da9ea1f37aade : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0985efc0f93f41f2bf3f26dbf929566e : SUCCESS in 12m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac51ce2b0d0548bc861e48b083fd771a : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bbe61c378d14ad3beaf087ef139b2f6 : FAILURE in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59f467192eec4494968e8115816e595c : FAILURE in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78409be239f64fd0b4da1f5bcfd2b038 : SUCCESS in 18m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20fdbdff7b3945888133bf9aa2ff317d : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bb49949410ed4b34933ddbb586e97278 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3df0971696cf4ad6bccd1af5e9ef4feb : SUCCESS in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e97eb4cee7641d9a56903c0d27c8a6c : POST_FAILURE in 5m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c56c339c92fb45f480d27a20d042b743 : SUCCESS in 44m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ef9809394ba45b785ec7c42dc7f4eb2 : SUCCESS in 50m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c67dc5e8385e4bf7b489d820757367ef : SUCCESS in 48m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97411405c610455086b6cff14183a519 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f1fb7d7ce643408687bd1934982c12 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12256e24b3e941c8975aebc1b0e8e6cd : FAILURE in 8m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e491d8d809ce41178245a555dacfa65f : ERROR Failed to update project None in 4s","accounts_in_message":[],"_revision_number":1},{"id":"30a740efc22c0cdaf41c50ea827822e2672c42b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 18:55:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3583ea755e8176a576ccb2f77b5a7157d3dd3022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:04:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d76fd6ce6843a58896547a3ae199b3 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/746bc02c87554620aeefbf45f90e62ff : SUCCESS in 1h 06m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e748798df3e54a6fb5f22b1a06995196 : SUCCESS in 52m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/99e21d54a8b84587825e1607ba152345 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f13ead2f082a425691b34ff00d39c765 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f48ad1e055ee4476bc337a9b9473d779 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/782635cc42f44e6b82cc2b3f592d23e2 : SUCCESS in 13m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b671adbac5c47e6825acec64f8f738b : SUCCESS in 12m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2db9483e24df45ff8d4e2c7de6fd6392 : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3534778a9fc485eaf26ad3a3818109d : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3262f7dd10e9486592035d076071e2fe : SUCCESS in 19m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47aacbbc70554b7ba87b26c1e5d597bc : SUCCESS in 15m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9eaf0cf9c91f4df1bc03f5eae6a5eca6 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/79dd077149034aef9e6ce3d0200af269 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3681ac1e9e9d4b9fb1ada59f9bdd9f87 : SUCCESS in 17m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/663c2aa024db42e3bb7e7bd28eeb2db9 : POST_FAILURE in 4m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e8897ead0104ce5a358b41e62762930 : SUCCESS in 47m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7939796500fd41b5bf705fa433ac8f02 : SUCCESS in 39m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d04be2597544416a5dbc40a9622a88b : SUCCESS in 47m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d20689f891ec46de8b6d51a7b4261e8c : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a40699d295b46cc942091776ca27d7f : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/877ad0d3ac804cc0969de4a8b294834c : FAILURE in 7m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72b2fb2bbeb44dd28a59904b70cb3377 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0b6d23b5763f1d53553fe8c4482b3e405978780a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:04:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b59aa5eea5ce44fc5c6fd763d24333f7d602408d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:47:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3095bfc019314c9faa82c4a961471c87 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/06e8744c758b4ceeb527f9cb5b35051e : SUCCESS in 1h 42m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/764766a0ea164691b45eacc8683a81c3 : SUCCESS in 39m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/de4b5b0ce3ca44fdb59225de38d11fe2 : SUCCESS in 12m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3a5d1c642bb9404d9469e4591eb50368 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f2205440bb94e37a95c6f0002f2f991 : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b051fe5b57e44ff847e4217b2561dd1 : SUCCESS in 13m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6b04490d6c654bb7a1b7a02e106cb1b3 : SUCCESS in 15m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53681a4ae6e54f5b8193b59cf7eb6f3a : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4468d1bee7c3482da735a9afbdb817dd : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f10343c0b7441e6883f34c2a5734778 : SUCCESS in 23m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/938976cc66e0460e81a6d2604dee9854 : SUCCESS in 18m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b49a6b843d1c4ffa9777bde9b39112c9 : SUCCESS in 30m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1784a7aa6681410db50f9285dd21c39a : SUCCESS in 16m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c19ef90f65224f12b6eaceecaed9cca4 : SUCCESS in 18m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02358c024ae84c8bbd31441be0a9ddb7 : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79245b5f71414952b377c5cdab0e45b0 : SUCCESS in 48m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13d7769f13a44f988f56f5d7f5eef1da : SUCCESS in 51m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9926887c43814087bf9e50e59bbc0d2b : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348a5613c0044e08b627af4d90086c1d : SUCCESS in 6m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/272fb8d477f94b00aa31fef1e44127c7 : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c4d5a21aa344a381f4cc2d03fb4cdc61e740a372","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:47:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"05007ae52bf4ad26286a2e624b068922c6daa9b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:49:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fdf1e3efb643d2a795855c51aa6c97 : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9f647157ae9a40b29b1cf65dcf26deb7 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e6520f354ee14edeba3b544ee494a8be : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c532891dcd9f2158d2b7ca9e2cb9aa21b2ac268","revisions":{"9c532891dcd9f2158d2b7ca9e2cb9aa21b2ac268":{"kind":"REWORK","_number":1,"created":"2020-11-18 19:46:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/763251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/763251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/763251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/763251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/763251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/763251/1"}}},"commit":{"parents":[{"commit":"6d988a451836b02cbab80f8b34832c8cce0ed729","subject":"Merge \"Dep\u0027s should be restricted by upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d988a451836b02cbab80f8b34832c8cce0ed729"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-18 19:45:59.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-18 19:46:03.000000000","tz":-480},"subject":"Stop explicitly handling StopIteration in proxy-logging","message":"Stop explicitly handling StopIteration in proxy-logging\n\nThis used to be necessary when we were explicitly calling next() on the\napp iter, but we don\u0027t need it any more.\n\nChange-Id: I0ae13905443aa8c8a1941c3212d01c9a5ae674e6\nRelated-Change: I27feabe923a6520e983637a9c68a19ec7174a0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c532891dcd9f2158d2b7ca9e2cb9aa21b2ac268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c532891dcd9f2158d2b7ca9e2cb9aa21b2ac268"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
