)]}'
{"id":"openstack%2Fswift~763405","triplet_id":"openstack%2Fswift~master~I37c7410fbbfad4aa3ac95600f5a8b2317dc8a5af","project":"openstack/swift","branch":"master","topic":"finally-byte-enforce","hashtags":[],"change_id":"I37c7410fbbfad4aa3ac95600f5a8b2317dc8a5af","subject":"Enforce response length harder","status":"NEW","created":"2020-11-19 16:20:03.000000000","updated":"2020-11-20 01:35:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a9b767709b1e458c5a121d6cdc5b27639086f08b","_number":763405,"virtual_id_number":763405,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-20 01:35:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-20 00:51:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 00:51:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-11-20 01:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a161c6234781c42228ceb92f84051802ec03929","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-19 16:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d35f6c2b67f70a711e17535c7cdd3e413d7d5421","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-19 17:18:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb5dfadd916b5938f270629b326805e772935c44","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-19 18:57:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d57165debf28802dc55110c737912432eb481c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 20:46:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a5a9eec52f422eb4062b9b1f41cf66 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/dd8a2f69c20343e396a35ee4bc0e744c : SUCCESS in 1h 01m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aa999bcebe2466c985cdcb44ac03ee6 : SUCCESS in 53m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73a62969b28a4660a96e68f95e80bec6 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/befc13835e674d598dae72a16a77bdf5 : SUCCESS in 11m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/162c1ef5901b4d618acde36b185d329f : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4cb142956c481688e8236f34657f40 : SUCCESS in 17m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37469d67576f42da9c9df8ce8a372590 : SUCCESS in 15m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ffe8820744eb4a5581ce99188fe419e1 : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/871bf6f5a259424ca86b5e83f1fd1c9f : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e83299000c0b44aa8a743fcd03f76346 : SUCCESS in 27m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/604dcf6a17fd4bf3b775b4bde528c02b : SUCCESS in 27m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cf6a11bb18b944339ffaaf598aa541fc : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c471759e51ea4bd4b317cbfa21307e43 : SUCCESS in 35m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/487911485a1d4e18abe8df2fed73f9db : SUCCESS in 23m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ddaf6657def4e22a76aa927889688a8 : SUCCESS in 30m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdc1913b323841c8b87061023461f7d1 : FAILURE in 9m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fdcc6ca933e942e596021a0929c99834 : SUCCESS in 36m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f825769978564c329a02c01d7600b4c2 : SUCCESS in 51m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/925e51e5e36d4a9a891d80bcf0038283 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13721d2fb5574f358a41a8de4385cb43 : SUCCESS in 9m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eca8fba5b3754af69ea2534aecb7696f : SUCCESS in 22m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/478cb68b7ac148b2917c3c6b8e3e14a1 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":1},{"id":"910edc7655e0312b851d77b5a05a79698cf1dbec","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-19 21:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66fd2fb666c5bfa26aa1721bf3f9ee66467250f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-20 00:51:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nI think I\u0027m on board with this...","accounts_in_message":[],"_revision_number":2},{"id":"39ef188e7644ed5861c8f28ce30644a0e8db9839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:35:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3214057d8eac43ddafb6ec7d65e059d0 : SUCCESS in 21m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b44a0b9b4ef647cfa3f437270adc68a9 : SUCCESS in 1h 02m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/179e4f8a24fe4caaa76abd55c90946d3 : SUCCESS in 45m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac18405c985645029c4827f8c27e3e0e : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/49b03814219a4b7896261fbc749026ae : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9f5af85d2cbd4479b80204cc61f3c368 : SUCCESS in 13m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84613620788c427e9fb29c163987eb50 : SUCCESS in 13m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e470da070e94fe6accd843928aaa95e : SUCCESS in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ecd8e10fb2ee4b218ed652f3961c44ac : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b442ccbbba54314b3f4d5374538a9d3 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee0c2cbb8e79409bb7ad8f8a053ca73b : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c9c616a2ef9a4e5ba5d5c2ef1529735c : FAILURE in 18m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/63f42e81be974133a2edbe64d1e88054 : SUCCESS in 22m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f023c4aecb1a47168e90f5e3c51a281a : SUCCESS in 21m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5fa3de448c1e4d10b3536869534ea481 : SUCCESS in 18m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2ae1f15f68f428e988ba6a155db38f1 : SUCCESS in 31m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/312931fe7c394ec995c159bd0804e1f9 : FAILURE in 17m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d81c7003f9c64c969990fb347f607610 : SUCCESS in 43m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a27cdd43a63040a2855eddd1f7a598b2 : SUCCESS in 39m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55b6e407199d40038e1785b04d70bf1d : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e85dac0b234f9ab5bd9e96264164d7 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a03e0a5b412496a9f2b108226846475 : SUCCESS in 22m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a034358523ae41b9a6017ba0307b5f23 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a60c4cf0df19bb64bd6805e9d3556e25bce80ff0","revisions":{"e9b8b674d335575f37ad28f3bc4446ad1afc6911":{"kind":"REWORK","_number":1,"created":"2020-11-19 16:20:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/763405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/763405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/763405/1"}}},"commit":{"parents":[{"commit":"0dc918e71a75d7bb094707cf8bcbcbdd033b174b","subject":"Merge \"py2: Suppress more CryptographyDeprecationWarnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc918e71a75d7bb094707cf8bcbcbdd033b174b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-19 16:19:56.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-19 16:19:56.000000000","tz":-360},"subject":"Enforce response length harder","message":"Enforce response length harder\n\nChange-Id: I37c7410fbbfad4aa3ac95600f5a8b2317dc8a5af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b8b674d335575f37ad28f3bc4446ad1afc6911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b8b674d335575f37ad28f3bc4446ad1afc6911"}]},"branch":"refs/heads/master"},"a60c4cf0df19bb64bd6805e9d3556e25bce80ff0":{"kind":"REWORK","_number":2,"created":"2020-11-19 21:10:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/763405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/763405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/763405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/763405/2"}}},"commit":{"parents":[{"commit":"0dc918e71a75d7bb094707cf8bcbcbdd033b174b","subject":"Merge \"py2: Suppress more CryptographyDeprecationWarnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc918e71a75d7bb094707cf8bcbcbdd033b174b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-19 16:19:56.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-19 21:10:11.000000000","tz":-360},"subject":"Enforce response length harder","message":"Enforce response length harder\n\nChange-Id: I37c7410fbbfad4aa3ac95600f5a8b2317dc8a5af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a60c4cf0df19bb64bd6805e9d3556e25bce80ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a60c4cf0df19bb64bd6805e9d3556e25bce80ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
