)]}'
{"id":"openstack%2Fswift~763812","triplet_id":"openstack%2Fswift~master~I1e337cedbd36cef36391e4cd808ed4dbf2cab8a2","project":"openstack/swift","branch":"master","topic":"bug/1889951","hashtags":[],"change_id":"I1e337cedbd36cef36391e4cd808ed4dbf2cab8a2","subject":"Unit test to cover ssync receiver no updates start","status":"ABANDONED","created":"2020-11-23 17:43:11.000000000","updated":"2020-11-25 09:48:51.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"eb1417db7adb9f1bcbe5cdaae555216829e6576b","_number":763812,"virtual_id_number":763812,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-23 21:03:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-11-24 17:31:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-11-24 17:31:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 21:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 17:31:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"fc17fdc99fc1a4d70aca107381abe18a361b807f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-23 17:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5229f41fa2fdd6847783882c5abf9365093b04d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 21:03:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b6a0260da94b0da4d8c3c7ebdc114c : SUCCESS in 8m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd1b309a1ad444ab81a381dc0e5411f9 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b81d9aa61fd45f2bce3a3120087a96d : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d8c4491f58a4d3380fcd5454fa8b014 : FAILURE in 4m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e33259a8c1848e5b10cf9412968bcf1 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b6500df27e24a398de8b018e45cfe02 : SUCCESS in 13m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c9ac47b6cbaa4c8ca67a35305a93da47 : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91fa0e915bf645d39b6456926afc89cc : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c832f12709543e7ab5ad1e9f92bf377 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c7e9195049d4892a9d38cf2a9e2251b : SUCCESS in 18m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a5fff49f3c2e465d9e97d43ca2158f47 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dccf7fa225084a2aa3db047cb9f06e7b : SUCCESS in 17m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5e89caf32f9d4bf7a15c87dba8c89b7d : SUCCESS in 17m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9b2c7203e1f48fc9c540e7a01f20b65 : SUCCESS in 43m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d4a89cd366f41e194a5d24068dc0ad3 : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc6207e7ef39455e970677a9952f57b1 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607ab77937ed4791af79425a2f1f79d5 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03b0555b82c644ebb77b2955835ef3d1 : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a488b897ebabe1c7c1358b35d2c58039ff4351c0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-24 17:31:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nthis is great!  seems to be the only test that covers the new branch:\n\n\tdiff --git a/swift/obj/ssync_receiver.py b/swift/obj/ssync_receiver.py\n\tindex a3f175d20..54ef4046e 100644\n\t--- a/swift/obj/ssync_receiver.py\n\t+++ b/swift/obj/ssync_receiver.py\n\t@@ -423,9 +423,6 @@ class Receiver(object):\n\t\t of place partition, for example.\n\t\t \"\"\"\n\t\t line \u003d self._readline(\u0027updates start\u0027)\n\t-        if not line:\n\t-            # Guess they hung up waiting for us to process the missing check\n\t-            raise SsyncClientDisconnected\n\t\t if line.strip() !\u003d b\u0027:UPDATES: START\u0027:\n\t\t     raise Exception(\u0027Looking for :UPDATES: START got %r\u0027 % line[:1024])\n\t\t successes \u003d 0","accounts_in_message":[],"_revision_number":1},{"id":"57be34d1965fa146eb931de74fefd0172b2fba40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 22:12:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b360d7accc94463d88106b161f399c9d : SUCCESS in 7m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a56908464df1411a91a277c12348c9a7 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/504a3287884242a790f120246318d527 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9d240e0ba5648fc94e4eae8f442a36c : SUCCESS in 15m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cdcd53256f744cea06c7b5fd8af094b : SUCCESS in 14m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eab4f423fcad44088377222933137d6b : SUCCESS in 13m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/89656839bc3f49c9af1ebca0b925c3db : SUCCESS in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e96cd6d403e1431993bef27397fbab49 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c0a3c53a95944c53b1b3e01094184c2b : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c57aa0af0cc46ad860ad50cdb4f39da : SUCCESS in 16m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/565df647f4bc4df2883d8724ab993478 : SUCCESS in 26m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/817257cbed7a4461a7a47ecbdc9baab9 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/188a3d330fdf41b59d49020808b8cbc6 : TIMED_OUT in 1h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44bf87f0de494805980dc392b9f139d0 : TIMED_OUT in 1h 02m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b85ef06aa454eada9b30b847d34f165 : SUCCESS in 57m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a57ca9e2b2f840d9bb7ea1d2b46ecf95 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e04fc1fbedad4f7aaf0265857db8ccda : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/765ca3a732474692b22dd82913e24f23 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"eb1417db7adb9f1bcbe5cdaae555216829e6576b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-25 09:48:51.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/744270","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"314a262e2c17cc0371bac0490bb3a51e9fb7c6c4","revisions":{"314a262e2c17cc0371bac0490bb3a51e9fb7c6c4":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:43:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/763812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/763812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/763812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/763812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/763812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/763812/1"}}},"commit":{"parents":[{"commit":"8ee72d75add1b6704de770a4adecb38d9b17e280","subject":"ssync: Tolerate more hang-ups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ee72d75add1b6704de770a4adecb38d9b17e280"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-23 15:26:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-23 17:41:48.000000000","tz":0},"subject":"Unit test to cover ssync receiver no updates start","message":"Unit test to cover ssync receiver no updates start\n\nChange-Id: I1e337cedbd36cef36391e4cd808ed4dbf2cab8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/314a262e2c17cc0371bac0490bb3a51e9fb7c6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/314a262e2c17cc0371bac0490bb3a51e9fb7c6c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
