)]}'
{"id":"openstack%2Fswift~764163","triplet_id":"openstack%2Fswift~master~Ib913aebc4459530f42ac24de77621113906baba2","project":"openstack/swift","branch":"master","topic":"p-shard-audit-shrinking","hashtags":[],"change_id":"Ib913aebc4459530f42ac24de77621113906baba2","subject":"Only merge shard ranges from root when SHRINKING or SHRUNK","status":"ABANDONED","created":"2020-11-25 13:10:02.000000000","updated":"2021-01-07 11:51:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e98f342959c5e4f1e785da6baf4afc851c43fc8","_number":764163,"virtual_id_number":764163,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-02 00:32:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-12-11 20:38:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 19:06:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 20:38:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c9726d8675bf3fba095de49051eae515590d2e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-25 13:10:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ead27384875e6c84b0b33e0164de8aa3e4203a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 19:06:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3453b619f17e4a7d9ea3543e919507fa : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/84cddde70fba46dabc68262618eb60b4 : SUCCESS in 53m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9f9962a00e04fffa365978d51a6034c : SUCCESS in 1h 07m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45eaddfa5d354aba80abe089dc830ecb : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd4a2f8527ec4a0f86ac0909546b5630 : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51f3a79caa424bb4b1ee3c5e7ce3404d : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d274b57db9594162b139309733a30c9d : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea2420ab469f410884e2bf2ff60c461b : SUCCESS in 19m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e776fc1e3a84cbf879337a70ae80583 : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74381f69dd174473826f194c61fd772b : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddfcb6a68e384836a391460b0040205b : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67011e1d344b485ca3f116c810fef0e0 : SUCCESS in 18m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/efa04e4f77794bfd921833e5e9e470a3 : SUCCESS in 33m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e8d3af24eff44756b5ded17f561e208b : SUCCESS in 19m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e8e6f518c87e44d18ec6703410eaf1f4 : SUCCESS in 18m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/084c2f79276e4c1299d5466e72a0f9e6 : SUCCESS in 24m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4aeab01f524d4f6ba3b3bcbe086a63d8 : SUCCESS in 48m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8ac792a3e68420a94945fc27633bb06 : FAILURE in 19m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4908d04322e4fed94c0378075c5016e : FAILURE in 16m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4321678527d945cd96699b7a64897d04 : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811531ed362745a5b87f8198365e012e : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/394caced00f5431a8eb57f82f33fd5cf : SUCCESS in 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cb0d2c5aec14e45a9b2f6bfb4a4b7f7 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fc0af1904415b7f0521f165d0e9fef35ac37d137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-27 21:21:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"850b1ac6ea8ca38c428db53c5979a527795fb098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 22:31:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cdbc0cdf6f4465d817bdd3406cb27b6 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/88984bb9c98b49e5820ad3f144e2271b : SUCCESS in 1h 00m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb2e6419895a4e1b80f5634729fa6df2 : SUCCESS in 50m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1bc89c4e9a304b9cb3950e10a09c1210 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ab86aa577fb49af8cbbaeee254ea2cf : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4361be2db4de47d0a976aa55023086c2 : FAILURE in 18m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17b8f08b91ca408c9ff51bfecf393eed : FAILURE in 18m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9305424a11054a56bad40cb30b681856 : FAILURE in 11m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/def0e08945e541bc9b40de42dbcde75d : FAILURE in 13m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bf941e51639414fa0e12afec2f8868e : SUCCESS in 40m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7caffe039cb44da49173b617198f11bd : SUCCESS in 15m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d403906bf7524a649898d9b77190a759 : SUCCESS in 20m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/801bd4a4b917419c87c44d3f779b8371 : SUCCESS in 26m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6e86aca6db1146fdac6b92c88dbf9d85 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/082123264119439d8b67c1199d6e0f64 : SUCCESS in 30m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94026b0f212845b0b1550c367b968ea2 : FAILURE in 21m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0cba0365638c49168d085676a2f0c815 : FAILURE in 43m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14aa701a1372450ca23bc45741a334ce : SUCCESS in 39m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/911ced810e564e24be59d9284b47778b : SUCCESS in 38m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f7f2cc7bdfe4e4c8a1378df296ed112 : FAILURE in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eab0c2a59724caab838ea042b3c6c93 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4455e6f2f9741a4bb952f91154cdff0 : SUCCESS in 27m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bd5c0403dd54c94a44e3a1e2edd8ac2 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":2},{"id":"196e46aa10bbf7907e63e2e807bba991e94837c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-11-30 16:38:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c62b0f7fceeeefc3843b0a1bbdbd05a67797209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:56:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819be6e0d98e4ca98d7cd83397f37153 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c9e3856e06824706889b970184c1b4eb : FAILURE in 12m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a8403873d3e4c11adf9d5bcd85ee5d4 : SUCCESS in 50m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4db97c2c341418d8b849d7b1f249b97 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c07b56fb7a1a41e09828d075d9c08280 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e63555b14fe7491392dd9e20167c9293 : FAILURE in 13m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c91e77fe0df94d7bb1f2fc927e27cdcd : FAILURE in 11m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4cb7ff7e5284f52b175fde3e9386a14 : FAILURE in 11m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6765d7a02ec644feb4ceba8b73cb2cf8 : FAILURE in 16m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12a8d1bce73344e88e6ff9d2b9760759 : SUCCESS in 26m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5abc3a5b8a5144d69d419a959e67f056 : SUCCESS in 28m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c2d6e801bf34d25836e309ac57ce71a : SUCCESS in 16m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d6ca71198d384bd8acaa78bb9ee65504 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a405adccac3243209af24985b49cd1ad : SUCCESS in 23m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0adbcba9418a4bcd9a1ca2b00eb2b7b2 : FAILURE in 27m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5088f941e0c44b0f80acecf04e0a5a14 : SUCCESS in 25m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35bdac06d09844bda2f31da63c47750f : FAILURE in 44m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/062dbe5d832549dc994808c3cb37e4ea : SUCCESS in 43m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/586ba9cd13e2442c96fcec2b4f09e573 : SUCCESS in 39m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd8438081d594e0bbc59982a4b39ebcb : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87186883b1e94f43be3be30fdac08956 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/864e772269ae4ff2b9aeab5c43480589 : SUCCESS in 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb2951bfb1ae46c5b6705cbb13493b48 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":3},{"id":"118ad0614f373c309ee38629b2760cd3d55a8cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-01 15:14:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee055750254168a9cd0232ef94c9bb2a86c53b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:32:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed3bdb61f33474cb98679e5169bc900 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/3282406cb8cc436ebabed2c02a13cb40 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f7551c063b045b09c107c9bdd07b488 : SUCCESS in 49m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/beb0d7d42aba41d2b79574ee121ef8ee : SUCCESS in 7m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ab8866a4e9a489486544d6a0822f656 : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2473b80c7fb14d6fa3053099589c9391 : SUCCESS in 13m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62ef73057cf240269e3838d872e80fba : SUCCESS in 12m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3862d2976858490faa618901d2626ad3 : SUCCESS in 13m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb7bad5c6e54198bcb7e7621bf2f4d3 : SUCCESS in 12m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7abd4136a234421c9f9aa9c330c72ee8 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eff8382e8c764908be449b444fc82b81 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/126a18dd33774e0eb15affca857676ca : SUCCESS in 17m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/00c0984b2df142b0b1cdcf5e64081b8e : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7ea18848d488417e859ebb8587fcbdf4 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7c1db6b4debd4093b2cb37353e28b19b : SUCCESS in 16m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82dd042e8a73418eaa8fa7f9177a8f29 : SUCCESS in 21m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab959ab8cf2b4db8a59af562ba6c1ff9 : SUCCESS in 46m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bff3935e53554ff29dcad8ce7d4fca0e : SUCCESS in 43m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/036fc556d8df47f5802a170290efc026 : FAILURE in 21m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c82f824e2604146badaa5c296540717 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f06e50d65f446fdb814ef885cd658db : FAILURE in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c519291090744001bda9acba1d34f3da : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0952858c7a451793581ce9fcb1507b : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":4},{"id":"1418445f959b7d9d48010bfdb8add96e2902ba75","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-12-11 20:38:57.000000000","message":"Patch Set 4: Code-Review+2\n\nI tihnk this is the final state we want to review in - the two changes preceeding in the chain make the most sense (and pass all tests) when we get to here\n\nShould we rebase and squash the \"shards fetch from roots\" with these state limits?","accounts_in_message":[],"_revision_number":4},{"id":"e951cce4612e5cc794c777157e305e87e384efb8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-14 18:45:37.000000000","message":"Patch Set 4:\n\nsquashed this into https://review.opendev.org/c/openstack/swift/+/741721/12","accounts_in_message":[],"_revision_number":4},{"id":"8e98f342959c5e4f1e785da6baf4afc851c43fc8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 11:51:20.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2c3f97cc3c70c58afe5d61ad897a5cd5727c8d4c","revisions":{"9b09562d0a0dea365bf9f81f1b5d004ec2d9081f":{"kind":"REWORK","_number":1,"created":"2020-11-25 13:10:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/764163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/764163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/764163/1"}}},"commit":{"parents":[{"commit":"9d428034576525cc22717ae6dcce81095f587b05","subject":"Populate shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d428034576525cc22717ae6dcce81095f587b05"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-25 12:28:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-25 13:03:52.000000000","tz":0},"subject":"Only merge shard ranges from root when shrinking","message":"Only merge shard ranges from root when shrinking\n\nDuring shard container audit, overlapping shard ranges discovered from\nroot are merged into the shard\u0027s db. This patch restricts that\nbehavior to when the shard is in shrinking state, since this is the\nonly state in which a shard container needs to discover shards from\nthe root.\n\nIn general, a shard is responsible for processing its own sub-shard\nranges (if any) and reporting them to root. Replicas of a shard\ncontainer synchronise their sub-shard ranges via replication, and do\nnot rely on the root to propagate sub-shard ranges between shard\nreplicas. The exception to this is when a third party (or\nauto-sharding) wishes to instigate shrinking by modifying the shard\nand other acceptor shards in the root container, and then have the\nshrinking shard discover its acceptor shards fom the root.\n\nIn other circumstances, the current behavior of shards always merging\noverlapping shard ranges discovered from the root risks shard\ninheriting undesirable shards from the root. For example, if the root\nhas become polluted by split-brain shard range management, a sharding\nshard may have its sub-shards polluted by an undesired shard from the\nroot.\n\nChange-Id: Ib913aebc4459530f42ac24de77621113906baba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b09562d0a0dea365bf9f81f1b5d004ec2d9081f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b09562d0a0dea365bf9f81f1b5d004ec2d9081f"}]},"branch":"refs/heads/master"},"e3976a33bafa93f6b238b3da70c756678d94a9ef":{"kind":"REWORK","_number":2,"created":"2020-11-27 21:21:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/764163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/764163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/764163/2"}}},"commit":{"parents":[{"commit":"9d428034576525cc22717ae6dcce81095f587b05","subject":"Populate shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d428034576525cc22717ae6dcce81095f587b05"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-25 12:28:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-27 21:16:20.000000000","tz":0},"subject":"Only merge shard ranges from root when shrinking","message":"Only merge shard ranges from root when shrinking\n\nDuring shard container audit, overlapping shard ranges discovered from\nroot are merged into the shard\u0027s db. This patch restricts that\nbehavior to when the shard is in shrinking or sharded state.\n\nThe only time during which a shard container needs to discover shards\nfrom the root is while shrinking.  Ideally we\u0027d only allow merging of\nshard ranges from root in shrinking state, but we need to also allow\nit in the \u0027sharded\u0027 state because one replica of a shrinking shard may\nhave its own shard range already moved to \u0027sharded\u0027 by another\nreplica. This is non-ideal because a shard may also be in sharded\nstate while sharding, not shrinking.\n\nIn general, a shard is responsible for processing its own sub-shard\nranges (if any) and reporting them to root. Replicas of a shard\ncontainer synchronise their sub-shard ranges via replication, and do\nnot rely on the root to propagate sub-shard ranges between shard\nreplicas. The exception to this is when a third party (or\nauto-sharding) wishes to instigate shrinking by modifying the shard\nand other acceptor shards in the root container, and then have the\nshrinking shard discover its acceptor shards fom the root.\n\nIn other circumstances, the current behavior of shards always merging\noverlapping shard ranges discovered from the root risks shard\ninheriting undesirable shards from the root. For example, if the root\nhas become polluted by split-brain shard range management, a sharding\nshard may have its sub-shards polluted by an undesired shard from the\nroot.\n\nChange-Id: Ib913aebc4459530f42ac24de77621113906baba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3976a33bafa93f6b238b3da70c756678d94a9ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3976a33bafa93f6b238b3da70c756678d94a9ef"}]},"branch":"refs/heads/master"},"9064ddc1a6ac74cd3704c19dad7128a982633e44":{"kind":"REWORK","_number":3,"created":"2020-11-30 16:38:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/764163/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/764163/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/764163/3"}}},"commit":{"parents":[{"commit":"9d428034576525cc22717ae6dcce81095f587b05","subject":"Populate shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d428034576525cc22717ae6dcce81095f587b05"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-25 12:28:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-30 12:53:57.000000000","tz":0},"subject":"Only merge shard ranges from root when shrinking","message":"Only merge shard ranges from root when shrinking\n\nDuring shard container audit, overlapping shard ranges discovered from\nroot are merged into the shard\u0027s db. This patch restricts that\nbehavior to when the shard is in shrinking or sharded state.\n\nThe only time during which a shard container needs to discover shards\nfrom the root is while shrinking.  Ideally we\u0027d only allow merging of\nshard ranges from root in shrinking state, but we need to also allow\nit in the \u0027sharded\u0027 state because one replica of a shrinking shard may\nhave its own shard range already moved to \u0027sharded\u0027 by another\nreplica. This is non-ideal because a shard may also be in sharded\nstate while sharding, not shrinking.\n\nIn general, a shard is responsible for processing its own sub-shard\nranges (if any) and reporting them to root. Replicas of a shard\ncontainer synchronise their sub-shard ranges via replication, and do\nnot rely on the root to propagate sub-shard ranges between shard\nreplicas. The exception to this is when a third party (or\nauto-sharding) wishes to instigate shrinking by modifying the shard\nand other acceptor shards in the root container, and then have the\nshrinking shard discover its acceptor shards fom the root.\n\nIn other circumstances, the current behavior of shards always merging\noverlapping shard ranges discovered from the root risks shard\ninheriting undesirable shards from the root. For example, if the root\nhas become polluted by split-brain shard range management, a sharding\nshard may have its sub-shards polluted by an undesired shard from the\nroot.\n\nThis patch also adds a \u0027--auto-shard\u003d[true|false]\u0027 option to\nswift-container-sharder so that probe tests can selectively disable\nauto-sharding.\n\nChange-Id: Ib913aebc4459530f42ac24de77621113906baba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9064ddc1a6ac74cd3704c19dad7128a982633e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9064ddc1a6ac74cd3704c19dad7128a982633e44"}]},"branch":"refs/heads/master"},"2c3f97cc3c70c58afe5d61ad897a5cd5727c8d4c":{"kind":"REWORK","_number":4,"created":"2020-12-01 15:14:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/764163/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/764163/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/764163/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/764163/4"}}},"commit":{"parents":[{"commit":"24b0d1f49863cc9e3b0d37fea14c480028d1621c","subject":"Add probe test coverage for shard overlap resolution","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24b0d1f49863cc9e3b0d37fea14c480028d1621c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-11-25 12:28:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2020-12-01 15:11:52.000000000","tz":0},"subject":"Only merge shard ranges from root when SHRINKING or SHRUNK","message":"Only merge shard ranges from root when SHRINKING or SHRUNK\n\nDuring shard container audit, overlapping shard ranges discovered from\nroot are merged into the shard\u0027s db. This patch restricts that\nbehavior to when the shard is in shrinking state or a new shrunk state.\n\nIn general, a shard is responsible for processing its own sub-shard\nranges (if any) and reporting them to root. Replicas of a shard\ncontainer synchronise their sub-shard ranges via replication, and do\nnot rely on the root to propagate sub-shard ranges between shard\nreplicas. The exception to this is when a third party (or\nauto-sharding) wishes to instigate shrinking by modifying the shard\nand other acceptor shards in the root container, and then have the\nshrinking shard discover its acceptor shards fom the root.\n\nIn other circumstances, the current behavior of shards always merging\noverlapping shard ranges discovered from the root is undesirable\nbecause it risks shards inheriting other unrelated shard ranges. For\nexample, if the root has become polluted by split-brain shard range\nmanagement, a sharding shard may have its sub-shards polluted by an\nundesired shard from the root.\n\nDuring the shrinking process a shard range\u0027s own shard range state may\nbe either shrinking or, prior to this patch, sharded. The sharded\nstate could occur when one replica of a shrinking shard completed\nshrinking and moved the own shard range state to sharded before other\nreplica(s) had completed shrinking. This makes it impossible to\ndistinguish a shrinking shard (with sharded state), which we do want\nto inherit shard ranges, from a sharding shard (with sharded state),\nwhich we do not want to inherit shard ranges.\n\nThis patch therefore introduces a new shard range state, \u0027SHRUNK\u0027, and\napplies this state to shard ranges that have completed shrinking.\nShards are now restricted to inherit shard ranges from the root only\nwhen their own shard range state is either SHRINKING or SHRUNK.\n\nChange-Id: Ib913aebc4459530f42ac24de77621113906baba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c3f97cc3c70c58afe5d61ad897a5cd5727c8d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c3f97cc3c70c58afe5d61ad897a5cd5727c8d4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
