)]}'
{"id":"openstack%2Fswift~765420","triplet_id":"openstack%2Fswift~master~I14bc184d6414b73b43ccc65f3358f193480b7eaf","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I14bc184d6414b73b43ccc65f3358f193480b7eaf","subject":"proxy: Require container_info to read objects","status":"ABANDONED","created":"2020-12-03 23:23:48.000000000","updated":"2026-07-16 22:46:36.000000000","total_comment_count":10,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"5bdac629428a1e398e413fa0b9b233769f1ea1d9","_number":765420,"virtual_id_number":765420,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-04 06:08:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 06:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 15:40:38.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"130b71e0f2c7d29ad8abd75d068db07e051febde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-03 23:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae917144b804634612af5d94e50ca3cbbf114560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:08:32.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee3fd9acd3a4db5a00cb5c5ada318aa : SUCCESS in 18m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/7ddc1b4226ea41258cd63386f7ec2073 : SUCCESS in 1h 06m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a9f92726d994cc9951e1a1f02395ab5 : FAILURE in 1h 26m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/383e1978695c4589a5cf139a223c02fb : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f22552280ca9401cadbd7e4fd7408220 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9497d00eb53e4a97b5f985390bc651a7 : SUCCESS in 15m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6191a583c2fc4ac3abedb2f38aa6f638 : SUCCESS in 15m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/554c490c4b4a44e1a6560d865a13b124 : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/895cdafd02b142a996183018780251cf : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39cde31fa6db43b282af10689235704d : FAILURE in 24m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1aa1be24fede4039937d779ec3bea2fd : FAILURE in 19m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/513094bbb92940d0bcbbcfb9f598c8d7 : TIMED_OUT in 1h 01m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5d234e91f5bb4a589311380a287d70f2 : FAILURE in 23m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/19e53076315d40d89da14d06e091f8ca : FAILURE in 22m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2823b3577cc74c1caec5f7b273f8254c : FAILURE in 27m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91dda55826d5403aa1c5d689fe941d61 : SUCCESS in 32m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c05530ab3c0049e48bb954048a7d7c75 : FAILURE in 48m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9877f3c3e3184366889be5abaea1c9e0 : FAILURE in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/667ef9507a694ed1bc58a624db78d95f : TIMED_OUT in 1h 02m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7f1d64a658c4131bda136794b95acb1 : FAILURE in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e151b24614ce4916ac5b7f1a6d7301e2 : FAILURE in 8m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de676df23b394dcdbfceeb3ab448a4d5 : RETRY_LIMIT in 10m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8ba27cf6f14eb38f4d3876e6fab57a : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0ad0c82b3c17aff1bcc867ea02c122722cd74235","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-12-04 15:40:38.000000000","message":"Patch Set 1:\n\n(7 comments)\n\nIt\u0027s not obvious to me this is significantly better, the difference between 1 and N policies is a big turn off.\n\nI\u0027d be much more interested in approaching this \"problem\" where the proxies grow some robustness in the face of skeptical container responses - to the extent proxies might even be willing to try *multiple* policies in the hope of servicing a request.  That\u0027s obviously a more signficant change; but it\u0027s more interesting to me.\n\nIt could be I\u0027m just not looking at what\u0027s being offered here correctly.    I\u0027m not even sure I understand the \"main point\" of this change - something like: return a \"better\" 404 ... earlier ... if the container can\u0027t be found?","accounts_in_message":[],"_revision_number":1},{"id":"41f607db23a7991c190a831a14a171c9864f569b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-04 20:04:25.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e the difference between 1 and N policies is a big turn off\n\nI\u0027m just worried that not having the special case for single-policy clusters will bite someone. I suppose I could drop it and just 503... but it seems dumb when there\u0027s only one set of places that might have data.\n\n\u003e proxies might even be willing to try *multiple* policies in the hope of servicing a request\n\nBut *how many* should we check, and in what order? If we\u0027ve got a cluster with dozens (or hundreds 🤮) of policies, it seems like we\u0027d be better off increasing the timeout to get container info so we know what the hell we\u0027re doing rather than doing some massive request-amplification.\n\n\u003e I\u0027m not even sure I understand the \"main point\" of this change - something like: return a \"better\" 404 ... earlier ... if the container can\u0027t be found?\n\nThat was just an opportunistic drive-by -- it\u0027s the 503 I want.","accounts_in_message":[],"_revision_number":1},{"id":"6ae2f707098967d7d1bfef8006fc6b9c1402de96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-04 20:54:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n😕 I still think this is a reasonable idea, but it isn\u0027t actually going to help with the problem I had at hand when I wrote it.","accounts_in_message":[],"_revision_number":1},{"id":"5bdac629428a1e398e413fa0b9b233769f1ea1d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-16 22:46:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c148a9b4984a585752f2b0033ed4e47540a76dc5","revisions":{"c148a9b4984a585752f2b0033ed4e47540a76dc5":{"kind":"REWORK","_number":1,"created":"2020-12-03 23:23:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/20/765420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/765420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/765420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/765420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/765420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/765420/1"}}},"commit":{"parents":[{"commit":"e22cad666a7bbb16a40a85a162a829da0f23c187","subject":"Merge \"xlo: 5xx while validating first segment is a server error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e22cad666a7bbb16a40a85a162a829da0f23c187"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-03 23:09:39.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-03 23:18:24.000000000","tz":-480},"subject":"proxy: Require container_info to read objects","message":"proxy: Require container_info to read objects\n\nAt least, when there\u0027s more than one policy. Otherwise, our current\nassumption of Policy-0 means we may erroneously send back a 404 where a\n503 would\u0027ve been more appropriate.\n\nWhile we\u0027re at it, defend the object-servers a little more when the\ncontainer doesn\u0027t exist; we should be fairly confident that the object\ndoesn\u0027t either.\n\nChange-Id: I14bc184d6414b73b43ccc65f3358f193480b7eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c148a9b4984a585752f2b0033ed4e47540a76dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c148a9b4984a585752f2b0033ed4e47540a76dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
