)]}'
{"id":"openstack%2Fswift~766104","triplet_id":"openstack%2Fswift~master~Ie05521f6cd146234dc5615c96ad19681b43e9110","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie05521f6cd146234dc5615c96ad19681b43e9110","subject":"Give probe tests a second chance to pass","status":"MERGED","created":"2020-12-08 23:24:38.000000000","updated":"2020-12-11 13:07:10.000000000","submitted":"2020-12-11 12:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766104","meta_rev_id":"2c43710895434498794f21463477b5590b0055d4","_number":766104,"virtual_id_number":766104,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 12:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2020-12-11 10:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-09 05:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2020-12-11 10:13:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-09 05:00:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 00:34:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 10:13:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"0c3e5ea2ed9701c585a2d54f439d0ee479baa1ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 23:24:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4e235669268ebf568888b83b01f7cd2fe80c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 00:34:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971f3d772b7e4aa3aa0088290c0f634b : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/23d6b1d24ceb4c18a3d0dcbf2b9d80a9 : SUCCESS in 1h 09m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1be39887ab0c4bad907a6118724b3dd1 : SUCCESS in 47m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0bb8548eaf53491aaf708e6801c1e41a : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/93d615ee374c4107a1c6752b7adcef64 : SUCCESS in 9m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12696fa026e2472eb9fe199e561072eb : SUCCESS in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50e44639feb646898f4ad2879d74eb5e : SUCCESS in 13m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e1572aff666421099ce94a47789be4c : SUCCESS in 13m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57e1b993f4ef4b1da96e58476cc5c041 : SUCCESS in 14m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e9620c82ec8d4a508d07206e7060fe7e : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b121da657fc7455eb475c53648a36494 : FAILURE in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c874feb939f4e65b4a8e96de5687bb5 : SUCCESS in 17m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a38e5cf2ef314d498e3046a060118375 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7684f254d18e43c98a3aef1cba39a345 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0ae8ec1fd8e644ee964a228c892a993f : SUCCESS in 20m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fbd044a53d4b44f1b32f376ec12d621a : POST_FAILURE in 5m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0860746d2c684a05bdc55de04037be47 : FAILURE in 42m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ff278eaa36c451291f55e5f858a35d9 : SUCCESS in 42m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92fe1c7537174171a968068d2b0300dc : SUCCESS in 42m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a1634c8a59b4fd38970b8c43e0bc09c : TIMED_OUT in 40m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a2b091e5b44f72a532e36505e3340f : SUCCESS in 12m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46c068620c684dc1984eaa5b568e2600 : FAILURE in 13m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/261a6b5c867f4a53b183c43307a1f444 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5c4e6825dc4dc2b7ce64a0bbdb60b366169278d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 00:43:25.000000000","message":"Patch Set 1:\n\nrecheck\n\nWell, it re-ran the failed tests. Still had 5/7 of the re-ran tests fail, though :-(\n\nhttps://zuul.opendev.org/t/openstack/build/0860746d2c684a05bdc55de04037be47/log/job-output.txt#1488-1495","accounts_in_message":[],"_revision_number":1},{"id":"888399ebd8faa7764521202358933cfba69d7fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 01:56:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cdac447e23464bbdb57690d932e6ac : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/48c65d6771104cb0b425ab995029c39c : SUCCESS in 1h 08m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7d11be312a141cc8d57b6809601eb7a : SUCCESS in 54m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4962aaece80f466a922bfea1102db200 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/66d360bcaca84f058d6cfe5048d81384 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e0910098b53436583197c8c84f06df8 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86c34b7430344492bf3462fc8a0a87d4 : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac0fbfb1612545a3942530c4cef84edd : SUCCESS in 13m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/de48561114c1496c8c6288934257bc59 : SUCCESS in 11m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92880a638d2c4dc482af1402e19b3a1e : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24ba584765274bb9aed5feb2e698a568 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7447f999fcd487abe9a9d9b2cc15e3f : SUCCESS in 16m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8ab5d3ba9faf4112a8b7d54c2bbdd06e : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9c753b41cf8a466fad025d2d09c720c9 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0b4baf1632542a39f6423b47439dea9 : SUCCESS in 17m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c395b8e712cd4e80b68044c6f2cea8f4 : POST_FAILURE in 5m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/689b7121374643c0bc2dcd829694653f : SUCCESS in 43m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d87645583254b7cb32b71156766e111 : SUCCESS in 41m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37ea3d837f3a48239e8de6e5edf77a8d : SUCCESS in 40m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d802035e0e84f108ac518b3ece8840b : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5524c84fd2784cedb58d11bf88912c74 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4e015f98e094a238a23f46ef880ce21 : FAILURE in 7m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f87fc4eab2674fddbf10b50bf5471713 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"831bf1b4ea1bb5a3949b1389c1753b883d4b40cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 05:00:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"320ccbc48f227637e037210f47443264c6ff3941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 05:01:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91e5aa193512d686cf795dd9d5c4de0c7e7b60af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 16:14:33.000000000","message":"Patch Set 1:\n\nrecheck\n\nDon\u0027t see it in any zuul queues...","accounts_in_message":[],"_revision_number":1},{"id":"510c3b24450d5ac93bb21f04990fd75548aa2a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86798efafa4b49d68637f482e842e258 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/63286c1e097c445ab3dd43d4792d4555 : SUCCESS in 1h 21m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/999107204d82444f9a2f93d26ea90ded : SUCCESS in 51m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87d6b582b70b4b97b2b0f73227913666 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0b4599f5e4044d3b0f64cbf3b76dda1 : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18cdbb90a6594a709a95f7d211eb86af : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e92f3d255a33481d80a7b705182b6ebd : SUCCESS in 14m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e10f3276b6164b949e19e96192f1c9cc : TIMED_OUT in 1h 00m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f883b926e8af4981bc46afe812fc7af3 : SUCCESS in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12cff8468ab94fc6a9ac342b1a2e5414 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e15b0a7b5e8d4210b86651657348b9fd : SUCCESS in 18m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/409a4d59010e4e7a95aba190228fc739 : SUCCESS in 18m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2e5edff3d3af42ed924be149e7ccff01 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c347180c2da94f1fb75d50e564cb8967 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/57bf2c21f8f94f55aa6c00290c9aaa20 : SUCCESS in 19m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d328c0c2e9c496cac1725f6d599de1e : POST_FAILURE in 6m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b4359eb8a05343bca7cf998053899eb4 : SUCCESS in 56m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/283ca3c830b74797a00b139c04d8283a : SUCCESS in 58m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ed62274c8fb4ec2be6eb8107d08a194 : SUCCESS in 59m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c4c5f7b2b4d40ceb974d262319ad056 : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8c1d49c0b54d0caaa033e881898f10 : SUCCESS in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a95d19a6540c4c098703d0ab53993447 : FAILURE in 9m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02bad68ecbbb4f8aaf6f5a076b4eca6c : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"131480d1978a12e787ab925f790fe76b154174ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 20:44:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a82f98882a2797107f4dda2b357caa1f868f4fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 23:43:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556554484cc24f05a96b7c94f157ba3b : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/cd33464b94a24c98a0425bf7e2a356f4 : SUCCESS in 1h 10m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b72d341f95846ccb723be06482e60b6 : ERROR Failed to update project None in 1h 28m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e416c65e8a564c0ca1bd90507b29ebce : SUCCESS in 12m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5461e088b4d41c0850a1cce04cafaa2 : SUCCESS in 14m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1c3688be96942c1b76d5da695a668d0 : SUCCESS in 13m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9250fbd0cfab44319303c95581e4d8bc : TIMED_OUT in 1h 01m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad2efb887ccd4d0b8702667b03c7e107 : SUCCESS in 17m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/741b3437cd6c4367ab2594e6876d975e : SUCCESS in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c99705a9c6c4dfd8c01acc4194e1ad0 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/165b653805c04afaa1e925346b11145c : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c64d6ae1eeaf406caf25c8bd492ad2cf : SUCCESS in 24m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c2bc9e98a5e141d79ec23eeb2387a317 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/99add63a014d4dcaa963de9efb5fbb37 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e81e11d1296d4f96af9d2805381844c2 : SUCCESS in 19m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b36ac8d1190a4308ada8bcc257a94efb : POST_FAILURE in 6m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5991e413c0ca4c97923907c263f01acc : SUCCESS in 51m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2338097fd50848fdb3e3a183270d6fd7 : SUCCESS in 48m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b330aa2da0314194bafba10f63d646ef : SUCCESS in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb970eed494147f9931e668733db14bf : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330a686d9d5842c083b0a3e521e92c65 : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1eaa8ef081247c18707b5e192836668 : FAILURE in 8m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2da4a726d849b7a76bd1bf945bbfc5 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"23af3c2ae0ea8471931ff78159032b49b87e02a3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 00:10:27.000000000","message":"Patch Set 1:\n\nrecheck\n\nWhat the...\n\n ERROR Failed to update project None in 1h 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c8821cfaefa98005a54ed12573e6f156b77a8479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 04:42:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1de2b02f84b4065b8b843adeb1d6864 : SUCCESS in 7m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2da379bafd9b4b538b308d137ac57d11 : ERROR Failed to update project None in 1h 02m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bdb738a3c424426f85268db5a275a20d : ERROR Failed to update project None in 1h 02m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a7ac2fc754048b396971185426baed2 : FAILURE in 2m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e9bffd4ce694e5e8675da7861bccf70 : FAILURE in 2m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e28ab7addde64499a3793a6c7299eb1b : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a18608c414674e528ab19d05dc16a7f2 : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8999d52550154a5ea1a60615789487d2 : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/141126df4e14447b981890dc7301c5a4 : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb91b86e20644b49853dea54fe102169 : SUCCESS in 21m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a903cacca3824de68ac1e6e4ade2b26c : SUCCESS in 22m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c942498ee5f4c9aa299f71315c81cf6 : SUCCESS in 16m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ca6bfb05b89f401cbb4029740248e466 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/234b017d20454491915b91bee23f688f : SUCCESS in 23m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/901e2a269514488090d0637dbabe1feb : SUCCESS in 24m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ac716435b6049469c963c89b51bb810 : POST_FAILURE in 4m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95ba1a672fde43c3a4457da335018b28 : SUCCESS in 46m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/003b47a9ee9d485582e52e2f5119f728 : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e6e97587f754aaebbde37b9ae0235d9 : SUCCESS in 38m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b52b65c8a4449fda456fc8ceb7649ac : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b55cd4a2d064c9d93ef1d3903eaae69 : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cecf4fa1099b46a69c731f5012709a05 : FAILURE in 7m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c88f6a7562dd450e8c3ea8be1138b4ef : ERROR Failed to update project None in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d96f935b20c9bd324febac5569d493842fb1f52a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 06:37:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64b4695cd09136cb90475d1c2427a88cc2051313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:46:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387219f9151b4bb5810b489f3f9283ef : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/9be56ce251614f81a3f50bfc7c5135bf : SUCCESS in 1h 05m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5408eb724164fbda9e4b23b85a62f63 : SUCCESS in 49m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad23d320aca74d8a8911486c0a1df914 : SUCCESS in 21m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6ece173a5f341d08135ca23815a0fe7 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e263e592af7d425690dbe44d6162069e : SUCCESS in 20m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4c097e2e4234774953786f45a99da8c : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/847f04632b8b4be7a6c87c1f64758068 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1985186af7be4336abaf3ed9475eb806 : SUCCESS in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d671a8aa1e0b43afa9e255bad55cdbb6 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ebdf8b19fa04922bd4186438e43582f : SUCCESS in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/313f36f2de6f4df4b9683fae4048c34c : SUCCESS in 17m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/35abdead266e49c782f7301621423b3d : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1c4295a751354a308b63f4a0fd8740a7 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/60ad7ceb4cbe438e8756c6f9af60969f : SUCCESS in 18m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb373575609846fcac970f94752e104b : POST_FAILURE in 4m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/feaee4fe7d3047a29187cdb5b04f7b96 : SUCCESS in 50m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0595e43a7d6b4016bec3b71f04cc528b : SUCCESS in 42m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32999a77605e4c41ba18c6f9c7100ac7 : SUCCESS in 46m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47506dcde266408eadc727d82c8a488e : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff5125a3dc64a068bd6d156f5039a90 : SUCCESS in 7m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd4703a44e574af6ac2d06cfbeabb697 : FAILURE in 10m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/070de02b3e964ec7927b468a19eda58c : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c648daa0cabf4a3e284f5166b5dfd71a3c3d5c86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:47:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ee2aa3e8c7f8b213941b612f065751f64565ec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:47:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a3b4816df1410680deea8746afd36a : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/e578716d8c71470781c754dba29ffed2 : SUCCESS in 1h 09m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ed08bb9cc854c0594eb22067f47df09 : SUCCESS in 51m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1ed62f9b92c740c5a6890ee318778fbf : SUCCESS in 10m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c4cbec740e424b2b9323b5d097dc9cf6 : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee1f741ef6df407da06e5a21d81e970a : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa608c3a88d49c085448e2d8f86cc93 : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/502e940d90474a6692ee913dbf128ea0 : SUCCESS in 17m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7adeb94ef5464d8089df73233c05a2a5 : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3a56b3808d14fd1adc0c705b97f7064 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b362e2108acc4f718bc0547fdea120ca : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05c1bb3bdf7e4c32a6df153de47c1a49 : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/309f882b9c0d4861813098e9979c997e : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3fd186c72b864a9281fb1d96cafa8215 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0cdd0b6bad894b10bd40943343186c00 : SUCCESS in 18m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39d4fdd44f774d36ab33691a132d3735 : FAILURE in 50m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08753e2a2d214e2892080f6431b1467f : SUCCESS in 47m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3ad41697aad4b96aed84d40c7896c85 : SUCCESS in 39m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6980659f0c2748cfbca3e1bf239673f7 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34312123faab4789a504400336b0f480 : SUCCESS in 5m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23784ac5bc6f4e85bedbd760038638a4 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5f280d8614b54895d99a8f929b4fcad18f80c162","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 16:10:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ab26f9e0eb63df43ffba7e1184208159742081f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 17:26:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1cc1bfaf7145a0b0ec991ece9aba13 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4c31c45bb6f44d7e90618e64b16d2fe3 : SUCCESS in 1h 06m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2bd06e9b184406182e8bafd1c86cadf : SUCCESS in 52m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6def7137b2fc43aa8ffa3854a6c586d1 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b2cf3ef64ae4985b019a303b00278e8 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3cb214dcbc95498badcfbb81d85adaab : SUCCESS in 13m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0d01a86f2a64939b9133c82f2e376e4 : SUCCESS in 13m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f023107553774976af16d2af59bceffb : SUCCESS in 14m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0f1875bdb0604a0c97b07977271c4356 : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0dcec9706574aa8b5c998dd7ca9d1ce : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ece4d3314bd46419241348d674dfe7e : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc8ffe48e8d94ba197906061e242050f : SUCCESS in 18m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab43b39b91d84543b4af06e16d4646a4 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2148c81f8fe643169e956ceffed787a0 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9d032da87ace4a2caf95a660a0d14863 : SUCCESS in 19m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/43dc19ad654241fda30364d2d88afbef : POST_FAILURE in 6m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78b062a4bcf74e17b3a2e9c26f925542 : SUCCESS in 49m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f9ad07a97094f5da36618e7875b09c9 : SUCCESS in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99d097c4e9264b6ba50de6d62b93772a : SUCCESS in 45m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd244ce1a5f64c51a147b89fd5108e72 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/357eb88bb4764994b94f8a8c27055662 : SUCCESS in 7m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3db9af8752604be29d617dcb8866743f : FAILURE in 9m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee5b380fdf674a789546c23a3b152358 : ERROR Failed to update project None in 5s","accounts_in_message":[],"_revision_number":1},{"id":"5de92c267586249db72d727d26e2a788bba82522","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 18:55:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9953e4f64dc5cef04f4f4d455d0c2d6b1ea9e59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:04:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e343a6b3e211420fb9761a344cb1f399 : SUCCESS in 7m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/45714b0467674cf293c2a99a6114e8a9 : SUCCESS in 1h 01m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cd9ec9fad1e49c0a7d7662c3cc129a8 : SUCCESS in 40m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b130649135894d59b81f15822bf2f0e9 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c1fdd5b7741f47eaac7757a02b6796dc : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80908154c1a14ebd9fd90f10b66ab2d9 : SUCCESS in 11m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a00a5eb929c43c09bbd66b63410d4b7 : SUCCESS in 11m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2475da0a038e4e47b93ca3c2b8a5d407 : SUCCESS in 12m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd2ad8a5683746ef806d347426902337 : SUCCESS in 11m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/704b7f1f3c164a2eadd61ef68d9da299 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2bd86c074c8240e8ae03b24b5356c6ac : SUCCESS in 15m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0009b94670f04e8aa06b28aa189a36cb : SUCCESS in 15m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8d2f8834d08840dd857abb6c1767cb08 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4af227fe93894bc19d91b77cb9b3a598 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a617fc6a71454d8f89945c239fcd5324 : SUCCESS in 16m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64d61a96355449b58490e30ff9cb8029 : POST_FAILURE in 5m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b5ba2134ac5c4c11996601da6235d8da : SUCCESS in 45m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ef61b64e9e84ecfa177d0408be48825 : SUCCESS in 39m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8478ac463253402cb474845dc44132ea : SUCCESS in 41m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1450c40c24f34614a795ed6bcd2ed027 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b671c9bf77441c83df57a96af8d34a : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/639226daca0e485c91b494251d302de4 : FAILURE in 7m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00c5251dde454a0993de3db119bffab6 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e93fde4fc265237eacfaebe34b62572e8b5a2ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:04:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d5cdf375e3019c922905e4165b8a9fbe6279bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:48:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee0af0ce3e74e6c908b27c3417c0ee4 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/16bc9312d47045ddb8b58da527035dcd : SUCCESS in 1h 11m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0f88bde10b142aab4353248ddeb0d24 : SUCCESS in 53m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/273473e9cf124b0baecd9c7998711cf2 : SUCCESS in 8m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/35d22ffac4be49dbab809bc1ed009dab : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96f957121aff42cb9f65dc08233320ae : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc6da3619f149a79ad34ca86f07c615 : SUCCESS in 13m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5dece6cb7b2147bd9dce91dac4aa44ca : SUCCESS in 12m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5e2c400d9d4248a9fe80472276de61 : SUCCESS in 11m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bd1cb029fc54a1a816514730a59ade6 : FAILURE in 15m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8ce83c8ab41434c85bbb5da7fff0018 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33e3d2abe5a64e2c84556ab45be2ca17 : FAILURE in 15m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff4e48080fff4c65a8dcb7a2cd533a52 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cbff6d3cea3e4c5b9b7e7b3bc5087611 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2e53d73a6e734fc2bcbf07851580ac76 : SUCCESS in 18m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94e6063ac4ff4ee4aa7c501411d8a92f : SUCCESS in 57m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0b1f270977f4880b1d7bf04911069b6 : SUCCESS in 38m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85d333bdf53a4956b4db57a37b81d5b5 : SUCCESS in 45m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a1012aa348e43db945d4638872da33e : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6630997b2d2a45c4a501664f0b04c91f : SUCCESS in 6m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f929f7d527ef41c0a11b46cbd2503558 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3f565a365cc9b6688a3602b3f34497c7a858ba24","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-11 00:04:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53b8e9d5d3ac33f0558b2c5cb3cd6d86bbaa24c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 01:04:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b4c33f5108436bb5d902704c1dc8b8 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e18e9cce845b44af87694b8da1d55c90 : SUCCESS in 56m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dccbb67fdab4654aea7d7981f0cb628 : SUCCESS in 45m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93f921db95814319b00483361efab0f7 : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7b4c84e3ad34ba5a2d9c5517371e0bf : SUCCESS in 10m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d3a003874a64992801724b281377bef : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f816d8db29d84ec3b923f6916f1461bd : SUCCESS in 12m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fca31b46d81b4d58aa5e56b14aa44901 : SUCCESS in 19m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43b280af111b43dd99ed589aab7776c8 : SUCCESS in 12m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5cf63b5705c64c8ea13cdf7edf2f9054 : FAILURE in 27m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3213c920cf5143e59d135f35376d6e73 : FAILURE in 16m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c85f5da99d764733abf018f6e59979b9 : SUCCESS in 18m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4a71d293e3434b36a36f0205012ed44a : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d7cadd5236b47e9aeb387956db49406 : FAILURE in 28m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/153fd46d4d9349039f80399a1503b2db : SUCCESS in 17m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0689bff23014c0b8f3391bcf9c8efbb : POST_FAILURE in 5m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/75653740ea814647ac0838b25ad13f62 : SUCCESS in 47m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8360b3a7c92b493195ed63b9d27ede75 : SUCCESS in 55m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a853f4fc206248c19443af8e4cfbb141 : SUCCESS in 40m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2af7cfe59fc946139484941494dc8148 : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eefee6f376cc4307828c84db069ce596 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/272b0dba0be1414aa68de862e1a5bed1 : FAILURE in 6m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e0103b7813146d9a61d88bf233f2b35 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"43b99b9ade57a9a973d251aa4072313655e2438d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-11 04:18:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e698dc4e7342052fb6ae74eb118b39f0c6480d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 05:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01342e5d0e0246459d524a28e8b9dd67 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/5dbcc67f10374f59a0eda825a7e5c7b1 : SUCCESS in 1h 04m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c67b793f4c6f4054906167ed13cb4c9e : SUCCESS in 57m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbceda30cb424cfeaaf44ffa669d64cf : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9213fc5bce154305b0fab6ace9f68a4a : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18d214322e1c4e07bb32628c0fe9e8aa : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c83cefe3a734886b69e898cf6290360 : SUCCESS in 12m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c623e8db47b4b38b13a178b040047aa : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87518d5631e8415d92442a69e4cb704f : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4329452da9a4660a77519f0af8557a8 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba54704c34b14ebfa9d790f55e1c66b6 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe9297d708c149b4962f15c961d23719 : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7be1952013e54bcf9d8a189b6fba1bd4 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eecc02b2da6d44988e8811e6614aa288 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/31b9c8d5388c46b68319f85d6fc36776 : SUCCESS in 19m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2ee305d668f4712a3719bb0a94dc379 : POST_FAILURE in 4m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35f7e4fd21754e1caff3f3241befcdb0 : SUCCESS in 53m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d733aeacd9a42658b7ee2b5c7a5f213 : SUCCESS in 46m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/031bfa35c73f426ca0539a35a94716c8 : SUCCESS in 44m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05be440201c246b5a00aabdb4b3c696b : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e9980d24d34b30833903320a6268ac : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7e43c21f7b84693957bd497bced1b13 : FAILURE in 7m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/651af63d17f7407e98f86256944b8780 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"09eec0a694b21e0a9fca050265a61db7f29af0ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 05:26:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dd3be86ad9831b2cb721efa4e3ac0c635f36a71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 06:58:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb7ff7af3d24a368ea02a9b67c81c6f : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/01fc20a8152f413a8d637c3f3a38c797 : SUCCESS in 51m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c57ac5016b4a4ee88e76950160bb8307 : SUCCESS in 45m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9279ac58437946618fecff49d398d981 : SUCCESS in 10m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5b91037b5dbe454494031d27d22a72d3 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60cf14d2e1bf463595e90d7523d72fc0 : SUCCESS in 51m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0859971b6a47de8f79978bec5f4867 : SUCCESS in 12m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/540ffed59dc64e2c9f4704f0958f1dde : SUCCESS in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aca2ce54d7d34160ae16edcc4693e65b : SUCCESS in 50m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a8b8bcc0bd54ddd923d538a0ca0d115 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf5906e5ad8f48b09a0a5ea84cbc2bb8 : FAILURE in 17m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e031ed4d25f4cb2aef86e9efbaf3307 : SUCCESS in 1h 08m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f3de79aefaef47dd805968a3f06bb580 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c6df6035d7cd42b1986baaff9f9a3dae : SUCCESS in 1h 08m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/40804339903a4d2a8f6f6068caf202f1 : SUCCESS in 30m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9abd617261554c6f9365d06bdd4796f2 : SUCCESS in 1h 28m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fd0b27000c444eca70612a7a67391a1 : SUCCESS in 45m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78d4f8bcc4494318b00efc6319d3d249 : SUCCESS in 43m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f86ac8e158e4c7190abe3cd6af6eda5 : TIMED_OUT in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62765c13f0e470abc9143f5bb1d302e : SUCCESS in 6m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01248eace2894159ad73312fd31749b6 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4c9ea559a35bb6138bcbb1637ef7f7ed7ec9364c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-11 10:13:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4a3f64287dbae44647e021da0635fb68626a273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:31:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233db20cc2074b26bb71ea65e7024910 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ac21b72f8d594e7188d158d02ee26336 : SUCCESS in 1h 05m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5bc7adaa09643ecb8ff48d3ded0b26a : SUCCESS in 51m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f3dba86aed94c7688935c7540f352c3 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/50d56b97be844961a2fcfa532bde3c82 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a50db4cdce1c4998ad5422837b97dd5b : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fdbd34618528416899e11192eb3c997a : SUCCESS in 15m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96d79574f65b49ad89738e016bf5e94c : SUCCESS in 11m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecc71702ce44927b7e898e287e61abd : SUCCESS in 13m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec95be0177324204bf17e4cf239370d0 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f9dd86d97914fe4a806737eef285363 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/adc191417d2f408598adfde99352e806 : SUCCESS in 17m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3792ec3684e44035bcbc77fb7d65436c : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7916f8e1a75e41cc9009a3ef9c2248cf : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c377370eaadd4edbacca38d656da2e1c : SUCCESS in 17m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fb7164633f54dbe92552350e380bd05 : POST_FAILURE in 4m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7acef7bf3c445c8b749e7239ebc3c61 : SUCCESS in 46m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bedf0028c944412b82507846549c026 : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb6cdd26a39f4e46833cc960bfe2bc45 : SUCCESS in 37m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1c878e0b9a84506a847d0d2176bdddc : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40358f0a81447f8877c82eea8f64bb7 : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67a166a5aa0d4683a7ffc1671272e5d4 : FAILURE in 8m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f46b092d803d49818922e7469b2416c4 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5bec9d44bfa0d86267911953dd16a786b5462db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:32:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97b9135cfd65f0d468b043ca9865f687f4e8cee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:57:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6af8aa4f894b7e9ddf9af950c18160 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/959845177ece47d0836c3f0fb5b8c1e5 : SUCCESS in 1h 06m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/905cc07dc90a4a459320644ffd813652 : SUCCESS in 42m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5171890ea67d4bcfa5d37e7ab00a3812 : SUCCESS in 8m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/297f6f4bb9b141b49b33836f70f9517e : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7779579a7e0b4ebb81a0204726ab691c : SUCCESS in 29m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2735fcd3871d4642835bf21f9b7c1af5 : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6226ae8ad8ce438d838eb3dcbac51580 : SUCCESS in 29m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/501062a5986e403bb3371c78374c6ca0 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3626697e69e34263898efedc0f1457b4 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14b7318ec461423e86795d9ec0c424ab : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e02238a42ff40489c74e613add2bca1 : SUCCESS in 16m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/74224374cb8b43308f94c272480ce697 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15b2cc1bb2c64cc289025332ca295a29 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/caa0c28685fd4dbe8ef4c94083edb6da : SUCCESS in 19m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e4d1630ba3b04ea6807669cb3c4ff47f : SUCCESS in 49m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b53f1e1ee35437eb6db7fdbd691b214 : SUCCESS in 42m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8eb7d15e0afa4b618ffd737924cb5c2f : SUCCESS in 49m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caba75a7e2644288937c224a1915c3d3 : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02ed2787b7e43168431f9f192a7202e : SUCCESS in 5m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d3f95ed15ce42a48f2d2a2f2920f93d : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1f9c10c53e0daed5bdb4c695bf0f47c344591f23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c43710895434498794f21463477b5590b0055d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 13:07:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65e1eda7e2047ec84ca88a65a50e94a : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/572b74234ed44134ac78f8ce38d27d92 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1033d85bdadb43908857c079f28f96fb : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4c0a7d3b3106f8b491e78ea21fca36c99ad04d9","revisions":{"d4c0a7d3b3106f8b491e78ea21fca36c99ad04d9":{"kind":"REWORK","_number":1,"created":"2020-12-08 23:24:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/766104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/766104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/766104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/766104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/766104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/766104/1"}}},"commit":{"parents":[{"commit":"fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d","subject":"Merge \"Give unit tests a second chance to pass\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 23:21:36.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 23:24:20.000000000","tz":-480},"subject":"Give probe tests a second chance to pass","message":"Give probe tests a second chance to pass\n\n...and bump up their timeout, since that seems more likely to happen if\nwe have to retry.\n\nChange-Id: Ie05521f6cd146234dc5615c96ad19681b43e9110\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c0a7d3b3106f8b491e78ea21fca36c99ad04d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c0a7d3b3106f8b491e78ea21fca36c99ad04d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
