)]}'
{"id":"openstack%2Fswift~766105","triplet_id":"openstack%2Fswift~master~I0a77025237549b192b6b695b6f18512dff28af6c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0a77025237549b192b6b695b6f18512dff28af6c","subject":"Increase gate timeouts from 1hr to 1.5hr","status":"MERGED","created":"2020-12-08 23:31:40.000000000","updated":"2020-12-10 20:18:08.000000000","submitted":"2020-12-10 20:10:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766105","meta_rev_id":"d8e13829cb1009ad77cc4912bede71c2e32bad83","_number":766105,"virtual_id_number":766105,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 20:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-09 05:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-10 18:57:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 00:35:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"239174cf3eec19b1cf39e22721546c400dcf0e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-08 23:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822808193877b5df9008daace00a5a989833c9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 00:35:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0613dba1b8c4dac819d022559de8cf3 : SUCCESS in 14m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/2b409905cf01464b8809c772bedfe2d4 : SUCCESS in 1h 00m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2adbedd43b804afb9f219daf61743ff8 : SUCCESS in 46m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1d6dd085b0f49b2ac6d061301ac1bbe : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/85415b14061d4a928780c919aa6c2f47 : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca82929e65734a5391df0712686258bb : SUCCESS in 12m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a93f9f6c933f489cbf5f70d6356e5ea3 : SUCCESS in 13m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d24014bc72e14187bb3c780ee459fba3 : SUCCESS in 13m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/adea219e82d44218a792e9c7a2c223a9 : SUCCESS in 12m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ed3c1eddb9247aba52bb8d31cb4125d : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab7b098c3e2b40db9cdc95ffa969cde6 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/249230b9ee0e40c3be4d313b5d07175e : SUCCESS in 16m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9b9f4a42e5934ba5be18c0f756653071 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c313aec7fb40454a963b0b1985b4d0f3 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3cb6edd2bc5b453b841298ece2df18b0 : SUCCESS in 21m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b4696d209b84146a35f3828986db48c : POST_FAILURE in 6m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/504de97bad1f4059945780b3ff47cc83 : SUCCESS in 55m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/56145c20509b43a48cf70598e6c9bf0d : SUCCESS in 42m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e6ef1b1cebf4546ae395a12efb98235 : SUCCESS in 34m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb412e7d34c0485aad8c85aeded1b113 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3465b407503340da8d7157e2ef78dfae : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/caa4db05690a4d1a964640f9c6d12597 : FAILURE in 7m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a18a2555c2547fb8626967a0c583337 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a380d2fc2d2fa3a05f50dc0a61273970502fdd93","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 05:05:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI should maybe also bump up the lower-constraints job\u0027s timeout; I think that\u0027s currently inherited from elsewhere.","accounts_in_message":[],"_revision_number":1},{"id":"c980831a71ccab266dee86c9105f3e5b97053f42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 05:05:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a16964875530fae4c68f686362feb8d12458dc14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 06:43:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b022701b539d4645a850181b0559923c : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ec8cf4fab7c249b3804016053e2e4a1f : SUCCESS in 1h 01m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dcfc4bd267e44235bcf4c5c3a4f5aa91 : SUCCESS in 48m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/400b7fe62e2243548133bf22546506fb : SUCCESS in 11m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/caef9f00e8394a1489a0bf8194a4e9ac : SUCCESS in 11m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/725cbac86f36417cb9cc686db565ab90 : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8a69e12efdd4c2a87a8ee94ff34c7ee : SUCCESS in 12m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc9457b003bc4429868cfec3029e9b39 : SUCCESS in 15m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9915d958ef8f4512a23e73d6103d4ebc : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3f947ad49bf4412ada55c45f9e943f6 : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71123ca677be4f10b60c9539a2138d9d : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da6c27d0369c4c5497a4eccaff54763f : SUCCESS in 22m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3db389b0c7664ee9944683023f5b09f8 : FAILURE in 17m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/06f1b73bb4c745edb342bf2765beb0d4 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6616536d2f964715840cfd80bc16c904 : SUCCESS in 17m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3a35e6b3a644a4896cc5d5ddacc08e0 : FAILURE in 52m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e288870bcf79405bbb2cee0789e05de8 : SUCCESS in 41m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f66f7a83a414273997f5b1f9f7c3db9 : SUCCESS in 44m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f17a24cb3ed740bb8e9a85800ab58682 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d8fa7297ea4c8cabc4c7f7b96cf96e : SUCCESS in 6m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b2f528df49640d2bd3cddbd2bba44c8 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":1},{"id":"390a74dce5c989ed9f3ff81dfd01dd40c8a68c5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 16:15:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a312520039c9bc87521f00366fbc04710388866d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:33:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54b462957fc41abb53bea136483c3b7 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/34b22debf2014ef0b909254459012ee0 : ERROR Failed to update project None in 1h 44m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/392a082d70994e12aad30a1b735c0547 : SUCCESS in 43m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06316fd902814152b0c2b28e4b391a13 : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a6cbc99814dd40d9947098cfa4f472af : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb6f0f0714e54072adbf7acf87b36f1f : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/624fb77dae1b46a4a6f460b718a2f571 : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eda9478564d7440098c98a641fd424e1 : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/749ecce672dc4c8886b5e51f5ad99a78 : SUCCESS in 12m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/994d2c4b52804fcfbaac04ccc7a39a66 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97702e39da1a46cfb214a351a0bad61e : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76ff1927cd724509ae86d7355b6e586d : SUCCESS in 18m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4ac92c7a50e5477689d9c21cd8d013cc : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/62a40c239fd04c129086fe362f5fd62d : SUCCESS in 18m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/21e4fed2e05c4e02bf50b46c71c1d46d : SUCCESS in 21m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f7a559aafcf45fe911d6f3f671ff541 : POST_FAILURE in 5m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d3e4a3ba2de449609dcd5a47112f6fbc : FAILURE in 49m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e456a060a6df44d6a031e6576e0b6032 : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7210655cac0c46f385c7f91cd6a255d9 : SUCCESS in 42m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2cefc8aca2d42178a4734f1e4761553 : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636c2b18fb584e36ac48d142ac903cc6 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f90d756e620d48b2b3adaa51ff083884 : FAILURE in 9m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90e0de24c7ae464fb002ce7527499163 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":1},{"id":"af0cfae2d8a2246a85d20a55de6023e831749726","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 20:44:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bff8e89f06c7864ae641b06257e23d5cf4a203a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 23:01:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4025a75bdde846fabdbb409b8d44ffd2 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/9c4d9b56befd4f6e9e36f7947a653b94 : ERROR Failed to update project None in 14m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/10a2ab4dafcc408bbbe53b67483dd9d4 : ERROR Failed to update project None in 13m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6384521c9b234c3bb626558d0082c85e : SUCCESS in 13m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/65b6537e836840adaae60c447c0854af : SUCCESS in 13m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/21cd2c25cb304e65b8bffbdd527686d0 : SUCCESS in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c00b50e28d1f4af6b258c7b8c307a731 : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05905ef2ef9343ee836a1ed318418f2b : SUCCESS in 14m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/72c4b6016e6745b99e5d60ff33b2ecdb : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb223e8f632b4ea495d2b6a43acdaef9 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a01f00af8aea43fd880414a27b753ed4 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61c6fe76b98246a6b1d379ad614cd563 : SUCCESS in 17m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5417a4f9488e40628bca2cad3ff88469 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e85474ff2090408a856f0857573b9f4e : SUCCESS in 19m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/566923355f27448c9260d46f2e5e31b8 : SUCCESS in 18m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d34284a1f5184d6f977fcd75f665e485 : POST_FAILURE in 5m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d242c15a72af4e07812ae60c0041427c : FAILURE in 46m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62eec11c6bfa4fb3a765b6e69ff61117 : SUCCESS in 41m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0305bd803154461aa8e123ed28a44afc : SUCCESS in 44m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a5e51afce0549d2ab9c4fec814624df : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6be5eb229e46659226e5813a96668f : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db0c88d7fb6b48749d44b34b5a2edeae : FAILURE in 8m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb591ad300c94a04ab100f34f0482753 : FAILURE in 46m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7148b69ad2421566576bfe7a4f2184eeb5d942a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 00:10:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91225a300b7a589f2ad11c9917bef1776a777d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 05:36:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0a24fd09044279bb93c4b5776f7820 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/6586cce9cf9d41de9e5f6ec61bb7502d : ERROR Failed to update project None in 2h 03m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/612490db6147479796698fb6366f38f2 : ERROR Failed to update project None in 1h 02m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db68bce6e6e840f7b31c459899be4e1c : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/64b89b78e8f1461298447d435439c0f7 : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d197d58bab7d45d39c7b07427e07651e : SUCCESS in 13m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ecb3ad3c640454aa25980d4a9c94b71 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd177b40df44441aabf7a47d52166d0d : SUCCESS in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a3cd41925e43aa88fba972bfbc7a74 : SUCCESS in 12m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3da676a81624b778550d4b3d8365011 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14abb49d06114d25904b3723cb889b83 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40f3cd1acd4940e49802c47657bc5664 : SUCCESS in 15m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c3afafc822b54db2a5289521a013a9c9 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/26d95631cd304728a172048d2fc0cec7 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7a703c44232f4adba6c2e6017a214c34 : SUCCESS in 16m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d969fed213844d095b8d53f79ca7471 : POST_FAILURE in 5m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96f0bad450b24f5984139fe7bf754775 : SUCCESS in 43m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab20d42dbfeb49baa46c324b698f507c : SUCCESS in 41m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/426f6ab1a6dc44fcbae77e41a213ae9d : SUCCESS in 37m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/757c3d722e80455bac4df5ce8b277c20 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eadb788a9384b0b9b19b41fefae5e07 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fba0c8dfb5be4487ac6a5b908badca7f : FAILURE in 8m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/613cf2a1a89a42d78c2b1bcd517b4a42 : ERROR Failed to update project None in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9c67624be30671828074e742267283adbc572ee9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 06:36:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed8dddc0dfa6d17dd0ae584b0c2965adcdaae95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:34:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf96d48f1a347b8a1d26bd94379b151 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/37ebf710a0584f288302e8e1505b12ee : SUCCESS in 1h 09m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55213705e5b547d5915759e288d0f4ef : SUCCESS in 55m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aaa7aeae89ea420b929cee7a54861872 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82bd22c17eb642578e3949aaff525f03 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ae2187346a448eb8e0451d206f757c0 : SUCCESS in 13m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58962edbaba44e27a793a9879c5b63d0 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8926199dd464d7b877606ba4730e3a8 : SUCCESS in 13m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5faac82a90c147f081f5acc1a186a7f5 : SUCCESS in 10m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/26aa20fccf0f428a900dbb0cd81e6dbc : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7fb7b3a3db64d63b19d29c0b8b1803a : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c45ff542e8044ebbf2afcf8c8d15f33 : SUCCESS in 15m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/987bbbe392c14af1a3a63de7c6a7ecce : SUCCESS in 28m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c780eafefe6948d89f486d3b80b47e5a : SUCCESS in 30m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7f4a3f1a958c43ec93c0533bad5d92d5 : SUCCESS in 38m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87ac86cb9c6445a2886e6a9c2ec118fe : POST_FAILURE in 6m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d981d5aa4b59421c875cafe64f4e9ee3 : FAILURE in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff07ba01f94c43babd8191c42ed8f456 : SUCCESS in 47m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ca0320ae8ac41e28d3e06cc42c0e923 : SUCCESS in 42m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ba7b2c30fe4be5a792c377d3e63d06 : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37388faec7a245919cadd489968ed9c5 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e50e62bbd48433581408feae4abcc68 : FAILURE in 10m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa2a7c893e5a40008fb973e9b873d2d3 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1b6b47d132f6c1be08c39e75b37d0f8ee498979e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 08:36:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"daa90f6378a8816716d0605f3843bae189d9f1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 12:40:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6e42ed54a14e47a2772848803e3278 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/24faa6818f084415b8755b18ca1470ac : SUCCESS in 1h 05m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea4a9e593061491d81c1e4796996470e : SUCCESS in 47m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8768e977ccfb4be3a4a48bb1d43116fe : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2542068bfd44454f9db96cc34e00a066 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e55554c703ce4218afac3c00c717a0c4 : SUCCESS in 13m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e65df7ec0d5f4bf0885b23ebd8986ba8 : SUCCESS in 13m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8006350ab54045cba820a5759a284fde : SUCCESS in 13m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50b805f708bb4837906b9a5e9ac06d08 : SUCCESS in 12m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64ed065cfe5b46d9ae6cca2dba93a6a2 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50623862038641588f8eea48737fa7d9 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8c85109e59c4871ad72e3a357fdcb22 : SUCCESS in 18m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/291212536caa49b289b914cbceaedef9 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f916c0243b1d48df95b820cc0509cb4b : SUCCESS in 17m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/29fa06d44a2342a0b8b66f916c109353 : FAILURE in 21m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5414e9ef21e5464e90b1509650ca455d : POST_FAILURE in 4m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b82e62175f184352b4cd41cca89c60fa : SUCCESS in 1h 09m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8fca1bc131234cb588ad60ed00fde289 : SUCCESS in 1h 26m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dfe4d2e069f4d8d901579853a491d6f : SUCCESS in 46m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75bc966a69c74e1bbcd8c5e526509929 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10100d3218434312a0ce8050aa885800 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb90d3a64dac48a0ae0c2b1ce7c3e4b7 : FAILURE in 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb03a1a5a1b04c28b869438dc45cb607 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5c032273ff5b458da113e7bb2a8cee42b036a667","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 16:13:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"17ba780caf1f1f2bfc1a53b13a910034551d9a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 17:45:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da7deab81064c4b98d64a9c318d0684 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c54062497e0140b187ac31f316691001 : SUCCESS in 1h 03m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6da842358324fafbe1789afff611f7a : SUCCESS in 45m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e4dd2195d2c4208a0cc50d63176d252 : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c657707dd0243808cb5f2377fb062d5 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e17836a597f48d89ffaa8d97f92bdf6 : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/be457b677c7c4ec8bf379504d8df3433 : SUCCESS in 15m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f276d4af5a14b92948ac342017049a2 : SUCCESS in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ed6553a5e140789fd707ed0d5bd54f : SUCCESS in 13m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d8de0ccd1da49d4b50535da6c23f70c : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5b4340c5b0d420b9ba2f1c52cfd582d : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8452596e922a43a297116134fa394d07 : SUCCESS in 16m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f1ae89447d67419f96ecce463e797f7d : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d0fe175b7ae5444aa6cb19f6208adda2 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0cbbdd1133d4527a80c3ae632f05c5d : SUCCESS in 21m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9afbab565150449b818e11b44e7594d8 : POST_FAILURE in 5m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2bee7eb59a34b4fb643e6e2b692b4ef : SUCCESS in 50m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef814e89f6e944628f87caca73cadea7 : SUCCESS in 47m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a86f7c855fc946899456149696ff8b61 : SUCCESS in 39m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437eabb555774711b27c55fafe456f92 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec0203c136841f0b072875a0c8434cd : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1764b7447f094542a8318f2fdc24b68b : FAILURE in 10m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06f616855d9a4360b48acc071737896b : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b2e226555edc9fc3bb691b276a7af756700a16a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 18:56:55.000000000","message":"Patch Set 1: -Workflow\n\nToggling workflow to see if it gets this back in the gate queue.","accounts_in_message":[],"_revision_number":1},{"id":"edb6a4563c02d46f021383d4381f4148d4e38454","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-10 18:57:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3a895b558cded0756ffa4e97b53c0352e9eb9c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"148320c6202f4c3703fac1157797eb525591a597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:10:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e71f42086844d33a13d2888cfd1e489 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/249ca10a3e13488f9a90b65330551d2e : SUCCESS in 1h 06m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5ec2cb496fa4f858718d5fd0689a1f9 : SUCCESS in 49m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bf1b79eaa3154989ae83b0dd23bc4530 : SUCCESS in 9m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c750d2c589de4358b31eefabbde5ee71 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/021ba61571cd461aa8fcca7d5d137807 : SUCCESS in 14m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b71fd7fc9bf6445c8424b843097f21bc : SUCCESS in 12m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/309f9f43cd6f4b4383982e4005daaf47 : SUCCESS in 22m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bae997d1953346d38b0039adaf051786 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/68aa1f48e4d24d958960c2f4084a575c : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16e6a3ad73284ad69344007eac3db218 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf9d1bf8d18644d8a38b5bfd8313ddf5 : SUCCESS in 16m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1adfc579e4224d0da242cd330b6790cd : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d690cb799ca4850bcd4ce4192a474e5 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c27ceb0b43bf481282a8b578b45f09ba : SUCCESS in 17m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/514a2a93efa14ed9827671cffc1f9582 : SUCCESS in 50m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b968e27be1a646f1a64b80a8a76a7a8a : SUCCESS in 44m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7cb73ee054954cae9d85d8d0d5e44c62 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f53a624b74cd4dce98cbecfa7cf44c53 : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a47cce66a74f19953e26bf2646d806 : SUCCESS in 5m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35c8b026ebbd468ca6c6fb607110e7af : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"672351f16914110fc1bda4fa5bad9edb88ea7d6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:10:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d8e13829cb1009ad77cc4912bede71c2e32bad83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:18:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888f4dbb607143dc8f5ccce817521190 : SUCCESS in 1m 15s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1588cb13a134488fbe27e193a3131476 : FAILURE in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/24f2ffad1dc34f2a8ac9478ce86b3b7d : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b87bc7d18f5d88d41bb147a30e845bff5fc74b91","revisions":{"b87bc7d18f5d88d41bb147a30e845bff5fc74b91":{"kind":"REWORK","_number":1,"created":"2020-12-08 23:31:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/766105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/766105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/766105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/766105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/766105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/766105/1"}}},"commit":{"parents":[{"commit":"fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d","subject":"Merge \"Give unit tests a second chance to pass\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe1f1e8271d3c6cdc773b447a55c6a5cdb3efd1d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 23:31:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 23:31:20.000000000","tz":-480},"subject":"Increase gate timeouts from 1hr to 1.5hr","message":"Increase gate timeouts from 1hr to 1.5hr\n\nChange-Id: I0a77025237549b192b6b695b6f18512dff28af6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b87bc7d18f5d88d41bb147a30e845bff5fc74b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b87bc7d18f5d88d41bb147a30e845bff5fc74b91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
