)]}'
{"id":"openstack%2Fswift~766214","triplet_id":"openstack%2Fswift~stable%2Ftrain~I294d5350e2f418614e54ff7bdd47dff16bfdcdbc","project":"openstack/swift","branch":"stable/train","topic":"bug/1907438","hashtags":[],"change_id":"I294d5350e2f418614e54ff7bdd47dff16bfdcdbc","subject":"[stable-only] Cap bandit to 1.6.2","status":"MERGED","created":"2020-12-09 12:39:59.000000000","updated":"2021-01-08 07:53:57.000000000","submitted":"2021-01-08 07:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766214-bug/1907438","meta_rev_id":"0becb38d395d438bd0fb31e30ee7af71537f082b","_number":766214,"virtual_id_number":766214,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 07:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-24 03:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-24 03:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 19:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 17:47:25.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2020-12-22 19:36:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2020-12-23 07:45:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2020-12-24 03:44:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-01-06 14:08:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-01-07 07:31:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"5cb2fa0d24851381757ce667ca13c20c74b0bc90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 12:39:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1cd2932c164d507153379450e72452fa50d2bf2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 13:24:57.000000000","message":"Topic set to bug/1907438","accounts_in_message":[],"_revision_number":1},{"id":"51bf1642b5030cb99b70f445b6a1a887d4129039","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 13:55:08.000000000","message":"Abandoned\n\nhttps://github.com/PyCQA/bandit/issues/663","accounts_in_message":[],"_revision_number":1},{"id":"a1edfafeb9f3a067f951515a40077997ea88f166","tag":"autogenerated:gerrit:restore","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 15:41:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"0ff5ff3ff6038ea624026ee03abfc9520fd582e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 19:39:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a626dec087854fb0ba3e0283464a299d : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/31bd4b57e908406ebf61323b6d2edb0a : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/377b2ced83e941d28ad4eda4b1fae22a : SUCCESS in 5m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/8415ec025f8344bd8734ec427d66c8a5 : FAILURE in 15m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1cb1cfb5a42d452bae22350740544d7a : SUCCESS in 59m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19d1aaa956794a62952293d2211ac3df : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c77abbeaddfa4775aa4f17bb901a0b02 : SUCCESS in 11m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/565b08dcea164964873c9ae7a2969353 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f899f30996e54db8a7c1bc94ca132d2c : SUCCESS in 12m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b5c800183a64730ac7da29b1f2b5532 : SUCCESS in 10m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf74119919964abd9e86a95c3f6d8614 : SUCCESS in 11m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f1f50a5d537f4382bb1dbedd23afeeca : SUCCESS in 12m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bd262269d1a45de901ea7be01f045c4 : SUCCESS in 11m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9890f879b3ae49b798aae2c281bcc4ea : SUCCESS in 13m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f5c0e3d95d4445229a7bf6a83baeb8d4 : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d49f4103a43341f8896466393e46189a : SUCCESS in 11m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3438a960918c4064acfdd30dd9bc7db2 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2becc2c4af52445b8c63ce53206ae2b4 : SUCCESS in 12m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/17192a52c8844828ac7381eb372e3163 : SUCCESS in 13m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/228420b5673e43e09f69d64ad2813c5b : POST_FAILURE in 5m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/383191fcb30647339fdee9b6a4e9d802 : SUCCESS in 35m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce2f444e53c54c288122d373a94a661e : SUCCESS in 32m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0c28831f90f24d11b2e60f3e44ce497f : SUCCESS in 30m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ae1b145125d4e7ab978d0da530f6519 : SUCCESS in 31m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df7726171d994c33bf2bd820f85f8f3a : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccca28f2b92e4466adc5c1e6f3a5714d : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53f9841079784926b9f9dc88a77414d9 : FAILURE in 9m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bb40b77bcc8474e9511de4b52e01fb2 : SUCCESS in 1h 00m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b6ffec88086480c8e716286a70fdd9f : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8920d7519e7f4dc5b08eee173999b82dbd7f6a68","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-11 12:21:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"4e1654b17c1231958cfa821a784724294372638f","tag":"autogenerated:gerrit:restore","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-22 12:25:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c9f3dd1ee755b22c993624cb7b9a6cc1c56dce34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc52303d24a0495fb332dd037f1a915f : SUCCESS in 5m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/49f91d42217a46aa8fcfa0a6d8e67d17 : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/657e739262c24b668ec26d61e0c30532 : SUCCESS in 5m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c7b648036e1f42fdb8ff63d80c00f425 : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fadfe012ccbe4cf8990573d187ee9e69 : SUCCESS in 40m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94905f4287d642a2864a88b379bb944e : SUCCESS in 12m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc57bb749a0e47d380d69cc59f20a207 : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bb990251b78476f868c197cb6c7aaa7 : SUCCESS in 11m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e6303a0b85a418288755144de602972 : SUCCESS in 14m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97bd811b43b244e39b4b9fa67b106426 : SUCCESS in 9m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19f78f5aa0994217a58f2f03d979f7a1 : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/366d3041bd0f4fb3a89d376d5a4f6fd2 : SUCCESS in 10m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d16578085e8440e299ffa936a2976314 : SUCCESS in 12m 25s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a4d3720ca0a940bb891cfe68d6f03b78 : SUCCESS in 14m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3b233ef781aa448888b7853bffa08054 : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/95b72e96370c4ab2b8e75c5dd3dac278 : SUCCESS in 11m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3b6a276cf2044df08fb01129e566e31e : SUCCESS in 10m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0e611a8d968a4c35bcf859aa3d91e3cd : SUCCESS in 11m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2958af12f0ff413ea0eb9546c7104c5c : SUCCESS in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32ef30846c964812a07c017c580a6959 : POST_FAILURE in 5m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7a28ef95873e44b4b2c8e4848722c55c : SUCCESS in 33m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/064bc32236aa41e6b6f87d150719faf4 : RETRY_LIMIT in 11m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/64c8998d51bd44d080e985bcd32dfec6 : SUCCESS in 29m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/213a707f602740d3b0f828394975f68c : RETRY_LIMIT in 11m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc69d88b381c4f54a43d87e6ca2fb8ab : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0d64ffd1234b789846c7d89d65fcbd : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7f5ec0c4df8431a92f8439ed4d11ac9 : FAILURE in 9m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa66df40ead3478690318968eac5ab8c : SUCCESS in 53m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/18787648ace24532b275f34ec4f39ef8 : FAILURE in 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0e22ad001f82a58b923de915053ec06223eeb455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-22 13:52:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"060545d49b4ef0d72c07e70fcfe45c0a202f7cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:05:43.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"fbbe33626f87ae63a0f789383cfc45c24cfec928","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2020-12-22 17:47:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"431cafeef1f8e50a96682ac8198e9fda68f35149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-22 18:51:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f48fa20a1ebe8f9a23fd72690478c95cedd7641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 19:34:09.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"26a524acd653b24a4f15ea43c453a2fe8a5e3e4c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-22 19:36:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c2524fab537b948ff9265bcb52e5c74174214601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 20:53:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac6c1ae764ef43d488608f7a9f7ed572 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/406244db84f64841bce4cb9eea4820a0 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6becdb6e8fea4d059edb0e5424e6f5cf : SUCCESS in 5m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2344e4efa3c74e62a2d99c5ef8e12168 : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/499bc88780b940f588771285206addf6 : SUCCESS in 46m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c6bdb695cd74ce4a0b9d50097f4d3ca : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f2c8be62ad04292a17c4b6f2a85eca5 : SUCCESS in 11m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c11e059333549f3b95287f75180d52b : SUCCESS in 12m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a4b1611c57064480819218086c75834b : SUCCESS in 12m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/786d0f73191b4374a5d6cd2278ec6314 : SUCCESS in 10m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6817766f87f04600a0dbb1fe5b5bdc49 : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d0d3e6df4e104ef8b64f4f66e7892bca : SUCCESS in 9m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeba61e176e24170a782b89e777acb29 : SUCCESS in 10m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0c3b0d214e7f48e4b73506e1775f52f5 : SUCCESS in 12m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cf016c8c7ca441ffae653f0107417e68 : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cac7255e2b904398a770d4df4b61e609 : SUCCESS in 32m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e3613fe2a37f4128a0f615ac0e5381bf : SUCCESS in 11m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f968ca886df04d28a0147d487867e14e : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/df9e02f53461469899318edb423a154a : SUCCESS in 12m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20d04d80da8c495c8442835eaddc4e8f : POST_FAILURE in 4m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e3de3c172ef4a36b8afdc04b05b149b : SUCCESS in 32m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a53a9222290f4eeda49e0cee6170b915 : SUCCESS in 27m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2980e8b0abff4078bb81d6aa3aababcf : SUCCESS in 28m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3e29cf310fd4a2e9d81138ae79db66a : SUCCESS in 26m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c592a3f413e45609e4313fe8acb1c2f : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79949c674564b2e943b2bf39d9e7bba : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0d2a70081474954a96cd8d350061471 : FAILURE in 12m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82ca343e04d947af82a3cc97a712a459 : SUCCESS in 52m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/590493d333a0467b95d920b28c631c70 : FAILURE in 27m 08s","accounts_in_message":[],"_revision_number":3},{"id":"21fadf33ade7c2e311bee473bd6bf783b84b6610","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-22 22:08:44.000000000","message":"Patch Set 3:\n\ngrenade job will be fixed once we have stable/stein fix merge - https://review.opendev.org/c/openstack/devstack/+/768257","accounts_in_message":[],"_revision_number":3},{"id":"3a6be44333910e3fa567a8123dec37a533c4abc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-23 00:19:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a6d7f0621305c24dddf49b4e533c9befef1c30d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 01:25:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc3619118b743fab8816e674f9b2e3c : SUCCESS in 5m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b172996c239456492fa2b168dbfe5d1 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f6678724f284d439286bf5be252b73b : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/9d7a10df05ad4870af765c81937645a3 : SUCCESS in 56m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86bbbf4b00a7408d9848bb9d78e64ada : SUCCESS in 43m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aca60f7ced8b47638850c04ca6a7f418 : SUCCESS in 8m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6bb5737f623483eb9d6e57d13be4658 : SUCCESS in 11m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a00d5e1e7d9490a9e57a3a6b53dad6d : SUCCESS in 11m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3864a7fb9ef940819e71b39bc800d5e1 : SUCCESS in 11m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01ff5e97c2914e6a99986026d7175d13 : SUCCESS in 8m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b9d23273d95438bb39683d4ca060763 : SUCCESS in 8m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/517226421ee84118b47a84ccb238356f : SUCCESS in 8m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4d0b3218bb9a49bba0eff57609bf76b0 : SUCCESS in 9m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/aeaf96576b4944c0a981e8a0f58affda : SUCCESS in 12m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/248189d8fa73425496b0a93a425f0532 : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8599dbf592a94618aa8477447e10f9cc : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d4ad7bfab4a74f34a49660c2351bf126 : SUCCESS in 9m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/649944151ef44d8b9e79a0909faf2841 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8dd5e1c1412d4847a38da896ae37a0a8 : SUCCESS in 12m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3110484c33ac493893e450d6038d91f4 : POST_FAILURE in 3m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/17b398f51dbd406b8c96605f11e10141 : SUCCESS in 31m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94d11e56ab554209a1c23ed8d26d76ba : SUCCESS in 27m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4be2afff186e45c4ad0dac18c73713a2 : SUCCESS in 26m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83dc402eced84a35bcd90178e86d953a : SUCCESS in 25m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a23276519f634e6f9354655e9c1168ba : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead8d0dc7b5e418da3d4c64445369f6b : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87c1a4fa745148d6ab186ee9eeddfa69 : FAILURE in 6m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8bf63f5ec32496abf318f3f5721229e : SUCCESS in 59m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a9a8dd7d47f34bbd8f70eefee19d9030 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":4},{"id":"5c5397e5ddf4a2d6f52916e89428042d59fd309b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-23 01:32:21.000000000","message":"Patch Set 4:\n\ngrenade-py3 job passing now but some other failure hit.","accounts_in_message":[],"_revision_number":4},{"id":"1c7dbeb979e00adef879d59c72401e4d61e044f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-23 01:32:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"50f1db5860626574540304cf158b98360816bb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 02:46:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1902d5778cc8437aa7fe33c62033b1d3 : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03d5fb0bc0754855b8e4a9d450958567 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e7f0935cbbc4bc5826c6f78cca90372 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/66ee21a503874392bcc3e2cdcb1e5098 : SUCCESS in 49m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c936c3f179714283add271225e1c1822 : SUCCESS in 44m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9544e01b2e2946488d303f490654f086 : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5618ce7717444b0bb6def28582bf5ac : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/297c813cd47f4c69b3c70db486e89657 : SUCCESS in 11m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e733566d9226496cacd39b80f1287803 : SUCCESS in 9m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/913fea2b314a4b9c97dd321379c53902 : SUCCESS in 7m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f41091e4bda1402ababd762ece4832c6 : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c743abdacb9e4e628926cf005e19f576 : SUCCESS in 8m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b392a210da77478ea047f3809b668bf0 : SUCCESS in 9m 43s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e99f93efab10456e8c592f5a1c145aea : SUCCESS in 11m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/363e885f52c347da87507573666dde71 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8254b43b817845318565dd414ba1dbe8 : SUCCESS in 9m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/983eb41c3cdc4ec7830507b5b2dcdefc : SUCCESS in 9m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7bddf8a24a204cd692f9d5dcd4dca24e : SUCCESS in 9m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5f2f3ca1cff8423bae90bd4e7d8183ec : SUCCESS in 11m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e276c5c867324b14a245ae65c9d18e4a : POST_FAILURE in 4m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9be8e40356424965917749d61548dcfa : FAILURE in 32m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/195b761e7183480b808d0047f77c5735 : SUCCESS in 39m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/dda284144638484d8832c43ab8e201f5 : SUCCESS in 24m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83cd87176056428b810eca2c23e05f1c : SUCCESS in 24m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9530794c3b3a4db996eded97d1709886 : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd711d395114d098445a963682b7043 : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5df93d1240604048b60258456f12ea06 : FAILURE in 7m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60d65661b20e4f4baadd6a2271cea694 : SUCCESS in 52m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fcd816ad74614a42be242f543acc2032 : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":4},{"id":"52f8b52dc61d15b63938838f2c272e25d3430c6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-23 07:45:06.000000000","message":"Patch Set 4:\n\nRolling upgrade job needs to be made non-voting, since the previous tag won\u0027t have a version cap for bandit. Probe test failure is most likely just a flaky test (at least, flaky when run in a sufficiently resource-constrained environment).","accounts_in_message":[],"_revision_number":4},{"id":"21fe323091ceb762d4737e23ec82d6231bffb9d4","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-23 08:46:36.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Rolling upgrade job needs to be made non-voting, since the previous tag won\u0027t have a version cap for bandit. Probe test failure is most likely just a flaky test (at least, flaky when run in a sufficiently resource-constrained environment).\n\nACK yeah sorry I thought I had already pushed a change with this, obviously not. I\u0027ll sort this out now.","accounts_in_message":[],"_revision_number":4},{"id":"3b4ae84c03753ee628849323789f5079d8d51b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-23 08:51:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f04be4c66307ef44bca9062df74f58f54cd480b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:58:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d787086d734a9a90ffeed90db7204e : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ebe3be234c7a4e008623a6eed73e1e2d : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e750e5f2b44bfe855821a47235f09c : SUCCESS in 5m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/d9662b3ceeae407898c46feee1a378e2 : SUCCESS in 49m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1a6ae0d33d44a15a7df372c7920b68d : SUCCESS in 39m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22f2bc1f2a02447182890da1299d7868 : SUCCESS in 12m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7770e64ddea04a7f83da5406ccdf9968 : SUCCESS in 11m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea336914400c40089be6307eb776eef8 : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e680b53d6b448b2b4ab2776274b7d85 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4cabd04ebb240e48f87b2670bff8c39 : SUCCESS in 10m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0b8e2d2d37e4ccdbd30843a25e57800 : SUCCESS in 10m 29s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f1dbfa6ff06546cda5b9f5e0ede2ae23 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56881470b5b64b3194ceefe26ed4d4a4 : SUCCESS in 10m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/764a9db5a9114cf9a6886cf4685dfdb2 : SUCCESS in 13m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7958537421284bdd821ee933c2d20b16 : SUCCESS in 10m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67625ef059e4431ea7d056e6a7ee4a1b : SUCCESS in 11m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c4583aaa37af48ee8ef45b424a9181fa : SUCCESS in 9m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/94a41f9001164e3888704ad97956b9ed : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/305d7baf4b904e32b83d958060b96d65 : SUCCESS in 13m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d0d504d48ac4a19873b94afee0d36c0 : POST_FAILURE in 4m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3905f9387c1b4fa293db3cd4f3715c80 : SUCCESS in 30m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a524e702e3b42adbc7fedace8dad0fb : SUCCESS in 31m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/dc2def742a4d434d9869c9dff9569ae5 : SUCCESS in 28m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6ac623b06ea41e9a6259af74d402d71 : SUCCESS in 42m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e138e60dec347d380628ebf0a17a8ca : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a190e13113468ca5b7529f7390ac1b : SUCCESS in 7m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76574e8f8fae4afdbc784ed61bcc3e42 : FAILURE in 11m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/743c8c4b4eb14038b0316ecc2b60a55e : SUCCESS in 52m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a09f13479f44401ad23ab3154dbe0b7 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":5},{"id":"879d7ec1922c5a1e7f8e43d9205f8caf0b0e2384","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-24 03:44:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nNo need to apologize; thanks for fixing this up!","accounts_in_message":[],"_revision_number":5},{"id":"47460696194d4435f75011cdebb75cac13eaab34","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-07 07:31:56.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f12d7d14cef3d033f6ef69dc2c67f2d4866e22dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 07:34:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2f633d27c27352db50c0531f51b4078a9ee85e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 08:45:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ba17c2fb2646278a1ce3288891c5bb : SUCCESS in 7m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/22c93d19b4924676a9c740b063394d18 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f378a6b0255e4c3a959986c3915eba1b : SUCCESS in 5m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a53573fc9637405ebbf4fc62282ef8ae : SUCCESS in 55m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebee4f9c0a1147be9d95fb2cd83ceff1 : SUCCESS in 40m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d90ab9d6c2f54bfa80dc6c7c6b726e2f : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38f999bc1392477da6acd9814c789f2d : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/15f77c4f36b6427d87408dec7f9b2773 : SUCCESS in 11m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5b537ec0ed0498392b0b5711da7ea44 : SUCCESS in 11m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39cc88baa1484eb8b659bfdd29d26d3f : SUCCESS in 9m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/626b67dc70c64c4bac89e76fb9061be4 : SUCCESS in 9m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8536c8d1d8114d9bbf4ebc5736d57d45 : SUCCESS in 9m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93fa0f5eb5364e498999d7eb04a97fcd : SUCCESS in 10m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7a7886fded70446f9363e666583d0048 : SUCCESS in 12m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/372f80bbf4b24212b3696d1298d38192 : SUCCESS in 10m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e2ebe3253c80436cbace1f5ea74e4384 : SUCCESS in 10m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d4cd8fa64f034d2990ee0e5718608dee : SUCCESS in 10m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6943018635934d82a0547eec0c0ec062 : SUCCESS in 11m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a10e110397d948c98c1256a9d24576d8 : SUCCESS in 18m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/649e9fa2b23f4782bf831cbd34f32286 : POST_FAILURE in 4m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b4edeae992241589998ce2dc3421dad : SUCCESS in 29m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef10bcf047974bf8b30e466d3b45b41e : SUCCESS in 28m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c6e5a059beef4089a6112a97b62e8d2e : SUCCESS in 27m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb97e4d1f6cf4436ad554c7d81e09ba9 : SUCCESS in 29m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe68efa016074c5c9274447be60f4f38 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f514a55a4c394f7fa357b7e3bc58abb3 : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9555dc588720489ebd16f8664b33d2a7 : FAILURE in 7m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1540ef503e8a4274a270f6d2d280a376 : SUCCESS in 1h 02m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/69d0fb7d8fc24e769587e87097ee00f3 : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":5},{"id":"14c6643e67cb5d0841d7eda1d5fd96e6c82b0d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 09:30:21.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"18cd62dce4061c11de159d78df217d27163fe359","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-07 13:32:07.000000000","message":"Patch Set 5:\n\nrecheck Depends-On are merged","accounts_in_message":[],"_revision_number":5},{"id":"22b787f77f7688132bc7e38860bdc0b04d1956a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:20:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59a5c3bbca2482e8a9e1ea5881733a3 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f63b119692b34e629c1c8d89eefc5d07 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c17a9852b9546e6ad5abf11db65518a : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/8ab5c1b492da42319270ad45d85c9f1a : SUCCESS in 57m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f8ef4dffef14cb0890b5b777764966e : SUCCESS in 50m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5166e7134b134bcd91d3d1fc78a5447f : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83e4a340a02548e39192dc89299d0791 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b22483f3b704ab8bfb21171319fbf2c : SUCCESS in 20m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d47420425474e29acc206c1a4a297e9 : SUCCESS in 11m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c27fc3a8e4940c4945c8181c0cb39ac : SUCCESS in 10m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/68af2126dd09484eac6c9cddfbe42889 : SUCCESS in 9m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/fd01530f55bd49799c91cf51e2a148f3 : SUCCESS in 10m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12a81b0781934cb199512d0c05f3fd92 : SUCCESS in 10m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6825970921784e4597d14e40e4c7cb9f : SUCCESS in 13m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9a396f730def4a7986a72ffd0d2bd049 : SUCCESS in 10m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/98a38abb581642e08b0907471f8841f5 : SUCCESS in 10m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8b19e6ffc0a746beb2fa017741c6ed7e : SUCCESS in 20m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b518f5ce9214ccfbe4dcde22fef25da : SUCCESS in 11m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/46f9f2e871524fc3a4add3680e45ede1 : SUCCESS in 15m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94a8e1e9fadb43489b6e893a0c8ee5e4 : POST_FAILURE in 6m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c5697df5db4f4ffea4c738b6f0871193 : SUCCESS in 33m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c65664c6de16447d8914748c06d2780e : SUCCESS in 46m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6fafb275e9584469aaa3e445f737aee0 : SUCCESS in 30m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db5ff34422004e6aaf3d529f96dddf73 : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ac4868793ec4f71a7cc2540c58bbbc7 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811f3410cde242f5a4569de75daa68ec : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/871266a39cda4ec5a17d1288f4ccaa9c : FAILURE in 8m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac8ec8e27450452a93c5fa6ea321c94e : SUCCESS in 57m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/acbdca71431e4496886ba62872b6d252 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":5},{"id":"92d588c43ecf452711f74e73a02589496c617b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:21:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a522cb365548ac2c57bb007da4d92dd926b268a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:03:28.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2489a1acd44916be0098e23412b049 : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f49d483c6574d5483b81635162c93c1 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418f5bfa87714abb97ae0fe4cbe98580 : SUCCESS in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/6ebfd094e97e4604b37cb86a6a973e44 : SUCCESS in 1h 09m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47a8d1b60a574ae5b11fdc381ed957eb : SUCCESS in 39m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/226a1bab50f74203b9ff0be51dbcd7fe : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84d099bd82de4af78aefe8bb894fd523 : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a96401acc014a8e8c05c429047cbeee : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e77fb9e377da44c0b130c19ce240e27e : SUCCESS in 11m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42e445f36127483cab82d419c3cdb88a : SUCCESS in 11m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57df7a2b5d33493db62712ac6da0aada : SUCCESS in 10m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dae99ef459d94948b9da016e299a0524 : SUCCESS in 9m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ebf4afc5f224866aa2ea14aa739cfbb : SUCCESS in 10m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/64cfce6ca75e48628339c5fdbd6cd47c : SUCCESS in 17m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f4f5d3a290e0413d9cdd6f8d66c4fe7e : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a527dc98adda46f2a483a4e074a69b2a : SUCCESS in 10m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c50ed8dd451f4a1f859f77ab25e0f4c9 : SUCCESS in 10m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/02d2c80e671b42d2beacc5aab2f8c472 : SUCCESS in 13m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/feb0fed7a647482ba4d76cd007b3a10e : FAILURE in 26m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32eea8ce9bb64232afeeaf717bcc7405 : SUCCESS in 31m 24s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/34214f373ea24045983b765ee7e42d34 : SUCCESS in 28m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff280bd76baf4beab529ef2bdcc02eb8 : SUCCESS in 25m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47ba4bbf7c074521ba8b73060403fa43 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26df446f95d4fe5855133ff45340ab3 : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/881080f4c0e740ab9afff7b6f7e13249 : SUCCESS in 47m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/53f70563cde74c149822a05a26c7c509 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":5},{"id":"df7c01df15436e45d2fc2694394b99d592015aa6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 05:38:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"91de286cf3e8dbd80b677c77430f5942e8a20076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 06:44:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4a414b16c346fc84ccc59ea4ba5e1f : SUCCESS in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5b3702854dc43c6a1e3b029b87c23ad : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a45a660cfbdf42dc8e1fea9b2bc8cfd0 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ed2d67924a8549b08db9351d4d72dd24 : SUCCESS in 1h 01m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9df2386fa741419cb795db89b55c1835 : SUCCESS in 42m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d671f021ff54a1aab93192a78a537f8 : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b84a4f75dc14eaeb533cee64beced3f : SUCCESS in 12m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2958f01a4c7c44929ad12328a06e1e17 : SUCCESS in 11m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de9b4b80da2b49e392ceb6ed2918918d : SUCCESS in 11m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1981efdd8264d7ba13f4c4ccd2259c4 : SUCCESS in 9m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7cc5b6ed1a8c4660b1a3cf975b9de103 : SUCCESS in 10m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d250f8318c59499ab379fc92f42d1bf5 : SUCCESS in 9m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04c4877d82d046c5bb233093517711f2 : SUCCESS in 12m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/116fdee3531b40ff9b9c865198e7d7a1 : SUCCESS in 13m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/af797cb95b774bc5baedfbe7d47ce6b7 : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aac1679412fe4176a917f05d8ae7b8e9 : SUCCESS in 16m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a2b8bf4195b84abfa00f0bbc0c8822d1 : SUCCESS in 10m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4169156b3a584b2193509866e52afe03 : SUCCESS in 10m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/308b8ab39d2f4a778c42ddc21ebd6cab : SUCCESS in 12m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bafc3188c1e4f6f8a6b84640024e9ce : POST_FAILURE in 5m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9aad4ed2fc8540f2b063f956e6619b72 : SUCCESS in 32m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3c5f3e703064e87a0583b5fdba9a141 : SUCCESS in 29m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/21ff7643b40144f890c26764df2a3188 : SUCCESS in 27m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60a41e82740045d491aa8e7a67f4a0ed : SUCCESS in 30m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02e1c19518e7499cad72187a1badb85d : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6513ffe666141198d44d2c8f266521e : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4ac4c49612243b7a6b1a7fb5d1925b0 : FAILURE in 7m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c06cf9959864e9a9f30d1ed519dfd75 : SUCCESS in 58m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/faf7145e9318497784327ce0bf5d1b33 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":5},{"id":"ba7f7ab44d259bb8ef25a0c6c85fa43e48233b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 06:45:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"51ff39622cde7ba9092a6b13171e2ba9c8e0a091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:50:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56469efa975341e2bd5a9d2941120767 : SUCCESS in 9m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/77119f00c8d3432fad13f0a337f9f4c9 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cfbb94016ea404a8fa3c0456716d497 : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a600be33ae2b4b85b2a0e9b4ff3c6985 : SUCCESS in 46m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d31de426e4914f74a78352bbcf0c7965 : SUCCESS in 40m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ada0a56c83104663a1530e30f07870e2 : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dca2a8bc175f4277a8d6651de7a0f6d2 : SUCCESS in 12m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0feb65b620d43cbba79dd67c7071faf : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e82f0de3765c4ebdb358b0e1516fd75d : SUCCESS in 13m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7ee3b26ee064c13a6ad1365d5eae681 : SUCCESS in 11m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2162f5d4f22e46fe884134014f3c8f07 : SUCCESS in 14m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9d2316b8a888409b805893cb9bcfab1b : SUCCESS in 14m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da5c72d3921a42d08ec0f8b27dded0eb : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/074dc63abb2847838c31ce45a6323862 : SUCCESS in 13m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dd51176119474336b8aee3b91c29d480 : SUCCESS in 10m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2ad5718a687c45f0bd530ef8dd333eff : SUCCESS in 9m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/10619726d22c40348e2f735e0f820b67 : SUCCESS in 14m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2026eb83a4394bd18fc2920acf6c30ac : SUCCESS in 14m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7023c383ec84b21b431dfc745f514a2 : SUCCESS in 50m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c87c4ab531349d8858fac8407f4fd43 : SUCCESS in 27m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d4b37edf3fc64e4ca22f2766f1e05fb8 : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2997df816bb846f08f5d747f6d570e87 : SUCCESS in 26m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d2a01ec3c9e47be94e47f3d6a237c01 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b196fdf36752477ebd2686c1b952f6e6 : SUCCESS in 6m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/823dedd5fb42453d9781eba3e5f4bb59 : SUCCESS in 55m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ffcc54d56c674a5b8b7b76b7f8e2c0c5 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":5},{"id":"ee1d5bf95f3f560ab28c85140a893a00f84d4f9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:50:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0becb38d395d438bd0fb31e30ee7af71537f082b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:53:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e03a7f16fd34bf0a37ab52ab2d0722d : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/554f87a7b34f4e74990659511e7a1823 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5949cc89b2df4ab09abce824ccfce8db : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d63b7013a8e363200ad5c2c5c039b37aa7881878","revisions":{"1f2326cd7cf5f8e32a55091f976f917093a0bede":{"kind":"REWORK","_number":1,"created":"2020-12-09 12:39:59.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/14/766214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/766214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/766214/1"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:38:03.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:39:57.000000000","tz":0},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nCloses-Bug: #1907438\nChange-Id: I294d5350e2f418614e54ff7bdd47dff16bfdcdbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f2326cd7cf5f8e32a55091f976f917093a0bede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f2326cd7cf5f8e32a55091f976f917093a0bede"}]},"branch":"refs/heads/stable/train"},"eb50e7e642669c50240be48467d4dd112f339b45":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-22 13:52:38.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/14/766214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/766214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/766214/2"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:38:03.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-22 13:52:38.000000000","tz":0},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nDepends-On: https://review.opendev.org/c/openstack/swift/+/766489\nCloses-Bug: #1907438\nChange-Id: I294d5350e2f418614e54ff7bdd47dff16bfdcdbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb50e7e642669c50240be48467d4dd112f339b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb50e7e642669c50240be48467d4dd112f339b45"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"5462860dea35fbdb184e0e7fb56f13574019335e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-22 18:51:01.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/14/766214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/766214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/766214/3"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:38:03.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-22 18:51:01.000000000","tz":0},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/768256\nCloses-Bug: #1907438\nChange-Id: I294d5350e2f418614e54ff7bdd47dff16bfdcdbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5462860dea35fbdb184e0e7fb56f13574019335e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5462860dea35fbdb184e0e7fb56f13574019335e"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"182982d55b2ac241cb6b35e70bfbf11a7d80cd3c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-12-23 00:19:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/766214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/766214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/766214/4"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:38:03.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-23 00:19:53.000000000","tz":0},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/768256\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/768257\nCloses-Bug: #1907438\nChange-Id: I294d5350e2f418614e54ff7bdd47dff16bfdcdbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/182982d55b2ac241cb6b35e70bfbf11a7d80cd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/182982d55b2ac241cb6b35e70bfbf11a7d80cd3c"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"d63b7013a8e363200ad5c2c5c039b37aa7881878":{"kind":"REWORK","_number":5,"created":"2020-12-23 08:51:20.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/14/766214/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/766214/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/766214/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/766214/5"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 12:38:03.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-23 08:48:12.000000000","tz":0},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\nswift-multinode-rolling-upgrade is also made non-voting in order for\nthis change to land as we cannot deploy from stable/train without this\nchange.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/768256\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/768257\nCloses-Bug: #1907438\nChange-Id: I294d5350e2f418614e54ff7bdd47dff16bfdcdbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d63b7013a8e363200ad5c2c5c039b37aa7881878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d63b7013a8e363200ad5c2c5c039b37aa7881878"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
